)]}'
{"id":"openstack%2Fkolla~831520","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I5f446fc276e245c1e6974ceeb6ff3dccd84ee502","project":"openstack/kolla","branch":"stable/train","topic":"centos8s-train","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-02 16:45:31.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I5f446fc276e245c1e6974ceeb6ff3dccd84ee502","subject":"Remove missing collectd packages","status":"MERGED","created":"2022-03-02 12:35:35.000000000","updated":"2022-03-02 18:20:12.000000000","submitted":"2022-03-02 18:18:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"831520-centos8s-train","meta_rev_id":"acaa89db9994f2fef0a72e99ec3193faa4ebb3bd","_number":831520,"virtual_id_number":831520,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 18:18:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-02 14:38:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-02 16:45:31.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-03-02 16:45:31.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-02 12:35:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-03-02 14:29:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-02 14:38:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"3fde68a30ee13ba87a4eda53649ec10a243ee5e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-02 12:35:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"145a3c0b57f40b003b269e293c1658fb10ad5db8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-02 13:11:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"69207fddc33550b2beb6ec1f4a5729afed073a8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 14:29:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e343bd8369644552939b04ea5bab1806 : SUCCESS in 40m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0e4674bc38f84e8e9865cedd51ca1c21 : SUCCESS in 47m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d2240b8f216d4cb5a37dc7ba13558177 : SUCCESS in 1h 10m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f93a443a1e4c40948040bdd2450ed35e : SUCCESS in 1h 14m 32s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5e3282ab2fe14847929ec706d1e83be0 : SUCCESS in 49m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/941371a492964a6ebdeee4ba85ae3d5d : RETRY_LIMIT in 4m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fc92d99aca78443887f65e058e961098 : RETRY_LIMIT in 5m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6b75f73770404495a9c9c9d527948501 : SUCCESS in 1h 03m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fe5ee47aedf4cdda48548d2b811be05 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f52f54940e44d4ea96369d716cd1045 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/304276ce60a34622b30b15bbe940a302 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe211ed1edf94cd7bc7021c343e655f8 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1bade016d464da78d1de8a842c0e271 : SUCCESS in 7m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d8aba03c8944995a290d39346810cd4 : FAILURE in 27m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aca33e6329964084bd25a31991efaa18 : SUCCESS in 22m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/482adf60d425472cb93388eb5841b498 : SUCCESS in 56m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a9d8761b999647639a4a5806d75a3178 : SUCCESS in 45m 45s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"8b865448fed127814f53f7141e9c6231dbb834e4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-02 14:38:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ca469bd6d8883166139a2b56dbd23d6a19bf5d23","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-02 16:45:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"da1f27c4a923a2c6666407d66e133d0e9f5ca7ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 16:46:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"789329231c2173ad7051c302318c76b7aad80db2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 18:18:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/06d09b4fc1474d3f86d4916a5bd82c98 : SUCCESS in 49m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6b63b4eb9d124821b4df66e1c9caaf11 : SUCCESS in 1h 26m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a979efbdf2334fca90f074bc518ae56a : SUCCESS in 51m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce222848efef4f7b865b48b8f59b3b5b : SUCCESS in 7m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e90c9df68a534c8eb1e0cf82e6a20971 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5224c4c00f83447a986ba724f8b4bdad : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2398f1b2ecb1465faae6c95a938f9566 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18f298ad96344f28b88d7f8a5b1b0763 : SUCCESS in 6m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9cb52be3421648849788f1c02c45c1f6 : SUCCESS in 25m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dbc0f03a11844f24a04cf794e0359bbd : SUCCESS in 48m 50s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"cb126afcf8cd274940850c3cbd29b8e3ee593b27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 18:18:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"acaa89db9994f2fef0a72e99ec3193faa4ebb3bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 18:20:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ceda08b3da4662a0871ac0d799a0c6 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5ecb3327c65e37feb503676840bd188f4bc6fe6","revisions":{"f52c6536175994480f35d8c072498c01c15c1b0f":{"kind":"REWORK","_number":1,"created":"2022-03-02 12:35:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/831520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/831520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/831520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/831520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/831520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/831520/1"}}},"commit":{"parents":[{"commit":"6b970b4b484ee75b48b8f361cf65354000fb8a7b","subject":"Merge \"Mitigate two Log4j vulnerabilities in Apache Storm\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b970b4b484ee75b48b8f361cf65354000fb8a7b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 14:52:19.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-02 12:23:46.000000000","tz":0},"subject":"Remove missing collectd packages","message":"Remove missing collectd packages\n\nSince CentOS Linux 8 repository retirement from\nCentOS mirrors - we need to use CentOS Stream 8\nversion of Opstools repo.\n\nListed below packages are missing from the repo,\nand according to the maintainer dpdk_telemetry is\nnot going to show up at all (it\u0027s basically dead).\n\nThe rest of the packages might show up at a later\nstage - but it\u0027s not a priority for the maintainer.\n\n- collectd-dpdk_telemetry\n- collectd-libpod-stats\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n\nChange-Id: I5f446fc276e245c1e6974ceeb6ff3dccd84ee502\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f52c6536175994480f35d8c072498c01c15c1b0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f52c6536175994480f35d8c072498c01c15c1b0f"}]},"branch":"refs/heads/stable/train"},"d5ecb3327c65e37feb503676840bd188f4bc6fe6":{"kind":"REWORK","_number":2,"created":"2022-03-02 13:11:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/831520/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/831520/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/831520/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/831520/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/831520/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/831520/2"}}},"commit":{"parents":[{"commit":"6b970b4b484ee75b48b8f361cf65354000fb8a7b","subject":"Merge \"Mitigate two Log4j vulnerabilities in Apache Storm\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b970b4b484ee75b48b8f361cf65354000fb8a7b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-02-03 14:52:19.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-02 13:11:39.000000000","tz":0},"subject":"Remove missing collectd packages","message":"Remove missing collectd packages\n\nSince CentOS Linux 8 repository retirement from\nCentOS mirrors - we need to use CentOS Stream 8\nversion of Opstools repo.\n\nListed below packages are missing from the repo,\nand according to the maintainer dpdk_telemetry is\nnot going to show up at all (it\u0027s basically dead).\n\nThe rest of the packages might show up at a later\nstage - but it\u0027s not a priority for the maintainer.\n\n- collectd-dpdk_telemetry\n- collectd-libpod-stats\n- collectd-sensubility\n- python3-sqlalchemy-collectd\n\nChange-Id: I5f446fc276e245c1e6974ceeb6ff3dccd84ee502\n(cherry picked from commit 405fca90f2c8d3763ac56ce8b925611d136bba6c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5ecb3327c65e37feb503676840bd188f4bc6fe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5ecb3327c65e37feb503676840bd188f4bc6fe6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
