)]}'
{"id":"openstack%2Fkolla~831548","triplet_id":"openstack%2Fkolla~master~I56f8fa49d3158fa74bba9fa054f64e2f1d78412a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-03-07 10:22:28.000000000","reason":"Change was abandoned"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-07 10:22:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I56f8fa49d3158fa74bba9fa054f64e2f1d78412a","subject":"Debian: always enable osbpo repositories","status":"ABANDONED","created":"2022-03-02 15:39:36.000000000","updated":"2022-03-07 10:22:28.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b2196872ccffc2269f6eede66dbca4f98b6501fb","_number":831548,"virtual_id_number":831548,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-03-06 19:46:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2022-03-06 19:28:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-02 15:43:34.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-03-02 15:45:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-03-02 17:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-06 19:28:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"dbdd4b5ad0a5056752ee0595829fb98c0d304d7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-02 15:39:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ef1dac0193a2d4e263281c6a63d556a8b67ce45","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-02 15:43:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6461a64c4a5bdda01486706dbe1aa7fab9e31908","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-02 15:45:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88cca2e44be2f2e6f914c18db89593edd64a077d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 17:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1945fedc97db4024920792a7b7905250 : SUCCESS in 28m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7ed055e95389411b903842d8f8721d24 : SUCCESS in 43m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/22035a958aad49839478019d4e4317ae : SUCCESS in 55m 27s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c8ccc02b84e345b9a841fbbcb71b9009 : SUCCESS in 57m 18s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9ed7bf032e8d437daabac045442d77c6 : SUCCESS in 1h 28m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5835e79514a8459f815e978c7244d630 : SUCCESS in 30m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c739fc69d3e64f84804b9bed80adeb35 : SUCCESS in 22m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d15c5bd20ccd487893a91b62877d61af : SUCCESS in 55m 47s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ad4a252fa9af4bfe9da1c3f8c4502ce2 : SUCCESS in 1h 04m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d8eee8b8284435a09cd72395484915 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d367f29d8c8b4d39b5ef70b75037f9df : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47104172047242cb95b611c30299e784 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e937a97d07c448808c43b8547ee40e93 : SUCCESS in 7m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5fc9c8e5f2ca484fbd7276863c8ec18f : SUCCESS in 20m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1a5fb87cceeb4eed8bf5daaa971afce6 : SUCCESS in 34m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7cd7c62262624f67aa94d4edbafc2208 : SUCCESS in 46m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c2aab00c2fc4355bac96c7a7a9e7668 : SUCCESS in 54m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/087e87f67fab424a87169ac0de96a31d : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":1},{"id":"e1179166783588d9671a277e7ed943d1d5ff1cd6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 21:46:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c0b9a63c7c740ca92eb501a3f4cc333 : SUCCESS in 1h 41m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fdf4c11ac4b3463883e18fbda448b1a2 : TIMED_OUT in 6h 04m 01s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f1487e7d4d3342ad9e7a2722bed00227 : SUCCESS in 1h 43m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e0312975ede4dfa8a7b6f38642063a4 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6318cf9bb6fd4b789a06527b39698366 : SUCCESS in 8m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8482083919d74c268b585b968fa20036 : SUCCESS in 1h 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f17480f74ce26d523bf38e253f5dd09c33245b6f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-06 17:48:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c65af06fadbec8643981e0c1d5b269fd6e66770","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-06 18:27:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f836f2794ccc721c4646d4997e533d7da9986d6d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-03-06 19:28:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a3516f1e9fd990a0ee32cfbd88571715f6e9b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-06 19:46:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7ba273f24c5341579a38186025cbc427 : SUCCESS in 23m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/653ea1fb122f4658b54c9523b8a66d0b : SUCCESS in 40m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7f6372b644d34757aebf88c6ec21a443 : SUCCESS in 53m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/89f88e8a921942fa95ff53bd3f88b215 : SUCCESS in 1h 00m 39s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3355c518e1f44da1a3dea85ef6371ec2 : SUCCESS in 1h 14m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f868643729e549b69200b43502d1ae9f : SUCCESS in 32m 02s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7db1adc7c4ab46d68b65382c2815dc21 : SUCCESS in 30m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dc11c7e78d65402f97c788595212ebf5 : SUCCESS in 54m 30s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0a353af3bd674988ac3053da17e42bfd : SUCCESS in 1h 14m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d071be4162fa4d47a9411b1d44d9cb63 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f600f85e91e4f16acc9acb80196aeb7 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0a98339736e41e791c76b474f6c5e6f : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc3b783c24404765bd69e0467d77f97f : SUCCESS in 7m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/874355c4ca9d4a49bf45dd38adfedeac : SUCCESS in 20m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa3c0ad78c894045817ba51e690efebb : SUCCESS in 37m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/66f3c5eece8248969b8df3fefcbfba74 : SUCCESS in 55m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d50f2ed85af34ef7ae41ba09da3e58bb : SUCCESS in 1h 03m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b63b176c95734d2f8e11ea96299c3b4f : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":2},{"id":"534cd220e5a419f07d992ba467c2bdba7d4cc8ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-06 19:59:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/992396e8858d4784b8f0deaf4a568960 : SUCCESS in 1h 23m 22s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/790599044ab944b1a76c0ef9fbd21007 : SUCCESS in 51m 22s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d77ae0f1ba1744a1b430faf73913f605 : SUCCESS in 1h 29m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48ec5cd569ee4096ac2ff999f3c2d252 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c1f72572cda40358174ad3702816259 : SUCCESS in 7m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b6f5b74bd62f4bf1ac6fbbfc5886438e : SUCCESS in 52m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b2196872ccffc2269f6eede66dbca4f98b6501fb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-07 10:22:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e225da50eac038d738857fc7c5b385b985e18295","revisions":{"d8622b6d5dc12a3965b87d4a52b781a30dd330ab":{"kind":"REWORK","_number":1,"created":"2022-03-02 15:39:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/831548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/831548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/831548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/831548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/831548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/831548/1"}}},"commit":{"parents":[{"commit":"2d7112345807dc1beee7b5d6c006056e8ecf3916","subject":"Merge \"Revert \"neutron: fix build in stable branches\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d7112345807dc1beee7b5d6c006056e8ecf3916"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-02 15:32:44.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-02 15:32:44.000000000","tz":60},"subject":"Debian: always enable osbpo repositories","message":"Debian: always enable osbpo repositories\n\nThis way we can get rid of extrepo. We need that repos anyway.\n\ndocker/base/sources.{debian,ubuntu} is new format APT repository\ninformation.\n\nChange-Id: I56f8fa49d3158fa74bba9fa054f64e2f1d78412a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8622b6d5dc12a3965b87d4a52b781a30dd330ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8622b6d5dc12a3965b87d4a52b781a30dd330ab"}]},"branch":"refs/heads/master"},"e225da50eac038d738857fc7c5b385b985e18295":{"kind":"REWORK","_number":2,"created":"2022-03-06 18:27:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/48/831548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/831548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/831548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/831548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/831548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/831548/2"}}},"commit":{"parents":[{"commit":"2d7112345807dc1beee7b5d6c006056e8ecf3916","subject":"Merge \"Revert \"neutron: fix build in stable branches\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d7112345807dc1beee7b5d6c006056e8ecf3916"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-02 15:32:44.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-06 18:27:36.000000000","tz":0},"subject":"Debian: always enable osbpo repositories","message":"Debian: always enable osbpo repositories\n\nThis way we can get rid of extrepo. We need that repos anyway.\n\ndocker/base/sources.{debian,ubuntu} is new format APT repository\ninformation.\n\nChange-Id: I56f8fa49d3158fa74bba9fa054f64e2f1d78412a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e225da50eac038d738857fc7c5b385b985e18295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e225da50eac038d738857fc7c5b385b985e18295"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
