)]}'
{"id":"openstack%2Fkolla~832650","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I120be2b28a7214e6b03976719319e4208d8050ab","project":"openstack/kolla","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-09 14:06:27.000000000","reason":"Mark Goddard replied on the change"}},"hashtags":[],"change_id":"I120be2b28a7214e6b03976719319e4208d8050ab","subject":"CentOS Stream 8: drop CentOS RabbitMQ 3.8 repo","status":"MERGED","created":"2022-03-08 15:07:57.000000000","updated":"2022-03-09 15:31:27.000000000","submitted":"2022-03-09 15:30:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"832650","meta_rev_id":"abf5f5708391b25a87d66d6c8fad335ed6918f76","_number":832650,"virtual_id_number":832650,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-09 15:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-09 11:04:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-09 14:06:27.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-03-09 14:06:27.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-08 16:28:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-08 17:53:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"1c25468be603765e42b484763e32cca3143c06c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-08 15:07:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70c540adad5388f7073a4614ea944b3856f5df08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 16:28:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c3aa8f7cbdc4497a84a6aaf6a7b80c1b : SUCCESS in 42m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/55e14b59af8b4f04aad2e64de6039dfc : SUCCESS in 52m 09s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e4dd9d1f6ab347d8aa37661e512200cd : FAILURE in 10m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/03f8ca6d0ded46b7804be3f3a7607df7 : SUCCESS in 1h 14m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9620af17b0c441109f8ec29771288d60 : SUCCESS in 1h 13m 05s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/07524a7784ef4fd4b9ea2a38709304c8 : FAILURE in 14m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/672d89760bd24668ba1667552a8fe36e : SUCCESS in 58m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bf7739741ef148fe88ca890e1ea52038 : RETRY_LIMIT in 4m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/10545c48f3444bfcba4425ae1cf0ab36 : RETRY_LIMIT in 4m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c2853aa3b2694c4ba6429a23f3add09d : SUCCESS in 53m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/876b8d5227874e5389fdb3a33c598eaf : SUCCESS in 6m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/07219adf7d954383bc89d1c9ddd283fc : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8440a93ab63440b9f06c24291110604 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56b86a55ec4e4e379704ac617cf554c7 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/146583f8cbc243d7bec2477de358a431 : SUCCESS in 8m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ed6e6af8301f44a99ff8818606bba970 : FAILURE in 24m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6810fcef29744262a0a5379394f0dd68 : SUCCESS in 27m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c10803aa3e143ce8f7837cc7b596a3f : SUCCESS in 50m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b240514ca433411f9e52ff0a7adc82e1 : SUCCESS in 50m 40s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"e9e72e5f9bd36213ac5239303d3bb9705973bd09","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-08 17:53:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc8cd14145a696a9a8e280f3168202c8fc78dd82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-09 10:51:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"670186737621eaea1d9dec01f716ecb66815a3ab","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-09 11:04:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"76622ee45a1dee86ff5f9d7a7250804eb9585b7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 12:04:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/571e4917f4d14c2fa0a27ca050147f47 : SUCCESS in 47m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/05d5655bd05a4ee98b14fdd5973dc789 : SUCCESS in 56m 50s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2a8e1a575349494d94ae840d48077c45 : SUCCESS in 35m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3616a03e1aa443fbace56b0767d7a30a : SUCCESS in 57m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b7a237c4e14c4102b52571addc9b4e0a : SUCCESS in 55m 06s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0430bac8ad7a40f7964f24c282f61697 : SUCCESS in 59m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/df48544677b743d8a2518e66c34cfbfd : SUCCESS in 44m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dd95bd44931745dcabd4e75afa29dbc8 : RETRY_LIMIT in 4m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/68ce6efb236f43fa8f3a75b6553ed72d : RETRY_LIMIT in 4m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b59f3aeb68724c3da153e5d335b4e9d0 : SUCCESS in 46m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75f0984609af4fa89c7a4163c9c84325 : SUCCESS in 6m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b6867d5a5e840eba9cdf140b3393d62 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2717a1fc19b49748e02744c9a7c7c62 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/128b3b9c9d784feaa006263778688d97 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd57ab85146f42fb86cb2f1f642b51d4 : SUCCESS in 7m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e48ad7befe04ce6ac9fa27979581587 : FAILURE in 21m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/46dc1e23c4834a1d8dd0725f09562877 : SUCCESS in 24m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/54b23c71d0fb4d5ba2f6e6dda1c96506 : SUCCESS in 50m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/50cc7f1b894847a3bb6a48957975de87 : SUCCESS in 48m 25s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"01203a483f82653dde1143fcce32228e85b53083","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-09 14:06:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dba667eb0826ac0827f1070799ce0e5acaac0dee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 14:07:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e31209f307aec1f973b10d21f41123081eab29d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 15:30:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a92c9c7bac0c4fc9a10228f88d83d88a : SUCCESS in 1h 11m 58s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0de18f7006de4605b2f9a297ff1979d8 : SUCCESS in 50m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/de1c14aa1cf249b3b706d3297df6796b : SUCCESS in 1h 21m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b7b02a05070a4389a7b0d9bbca5dc4af : SUCCESS in 57m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8d084d8912ad4a9383d476debd1c22a4 : SUCCESS in 48m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0687bce7d6a943bdbbfb294aa8c7c616 : SUCCESS in 6m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46014937581a46e2a07140cd4deaa71a : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f826bda4939941a7aef275aea4cc8a24 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a3a4bed1352b448c8dbc9088ab45f303 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/273bfc1ce9da4cdcb9a4c62361d03962 : SUCCESS in 7m 10s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7461c12fc07a4d7498d45201c8eb647c : SUCCESS in 40m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/906739aeccce4823876364ed08966a66 : SUCCESS in 50m 39s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":2},{"id":"e64f46eb2f1158ada9c2854cbd4196143f15442e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 15:30:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"abf5f5708391b25a87d66d6c8fad335ed6918f76","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 15:31:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/302696145db24a1cb8f1b17fc5fe375a : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3db029c8bf001a4d3bcd5be1c1b35c736013f32e","revisions":{"5768669eb37ba75a5056b1832793d56babba5329":{"kind":"REWORK","_number":1,"created":"2022-03-08 15:07:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/832650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/832650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/832650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/832650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/832650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/832650/1"}}},"commit":{"parents":[{"commit":"76841d99ec35dccf7231ef67938aaf3629957a6e","subject":"Build CentOS Stream 8 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76841d99ec35dccf7231ef67938aaf3629957a6e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-08 15:06:07.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-08 15:06:13.000000000","tz":0},"subject":"CentOS Stream 8: drop CentOS RabbitMQ 3.8 repo","message":"CentOS Stream 8: drop CentOS RabbitMQ 3.8 repo\n\nWe need 3.7 in Train in order to handle the transition from CentOS 7 to\n8. The RabbitMQ image ends up with 3.7.28, but for certainty, this\nchange removes the CentOS RabbitMQ 3.8 repository that was added with\nthe switch to Delorean for CentOS Stream 8.\n\nChange-Id: I120be2b28a7214e6b03976719319e4208d8050ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5768669eb37ba75a5056b1832793d56babba5329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5768669eb37ba75a5056b1832793d56babba5329"}]},"branch":"refs/heads/stable/train"},"3db029c8bf001a4d3bcd5be1c1b35c736013f32e":{"kind":"REWORK","_number":2,"created":"2022-03-09 10:51:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/832650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/50/832650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/832650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/832650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/50/832650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/50/832650/2"}}},"commit":{"parents":[{"commit":"76841d99ec35dccf7231ef67938aaf3629957a6e","subject":"Build CentOS Stream 8 images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76841d99ec35dccf7231ef67938aaf3629957a6e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-08 15:06:07.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-09 10:51:06.000000000","tz":0},"subject":"CentOS Stream 8: drop CentOS RabbitMQ 3.8 repo","message":"CentOS Stream 8: drop CentOS RabbitMQ 3.8 repo\n\nWe need 3.7 in Train in order to handle the transition from CentOS 7 to\n8. The RabbitMQ image ends up with 3.7.28, but for certainty, this\nchange removes the CentOS RabbitMQ 3.8 repository that was added with\nthe switch to Delorean for CentOS Stream 8.\n\nChange-Id: I120be2b28a7214e6b03976719319e4208d8050ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3db029c8bf001a4d3bcd5be1c1b35c736013f32e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3db029c8bf001a4d3bcd5be1c1b35c736013f32e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
