)]}'
{"id":"openstack%2Fkolla~832661","triplet_id":"openstack%2Fkolla~master~I4658a06b077d096ab9f2cb5070e39c98b5a27254","project":"openstack/kolla","branch":"master","topic":"bug/1964140","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-10 11:49:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4658a06b077d096ab9f2cb5070e39c98b5a27254","subject":"Install glusterfs-client in Debuntu","status":"MERGED","created":"2022-03-08 15:30:54.000000000","updated":"2022-03-10 11:50:48.000000000","submitted":"2022-03-10 11:49:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832661-bug/1964140","meta_rev_id":"4888949dc2ca2145f346817fe62978fee0fcbee4","_number":832661,"virtual_id_number":832661,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-10 11:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-10 07:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-10 10:12:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-03-10 10:12:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-08 17:08:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 07:51:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-03-10 10:12:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"332da31d8898a119d3a297cd6f65eb9b51a8e560","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-08 15:30:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f93888cd287233d72629ba2f5cbf6ee9dc915a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 17:08:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3b0fe1a9a33942db888e63548a53de08 : SUCCESS in 34m 24s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/14635eabca2e49c59040d409fff03927 : SUCCESS in 48m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c05e964eda944bc8b996f7411795f112 : SUCCESS in 1h 18m 02s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/062681c44d3c4a37ba5396b7a45e4081 : SUCCESS in 58m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4cdfdf0ac707486d920c77955a2cca7f : SUCCESS in 1h 29m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3dbe76bfbdc74915ba852e6ed7f20134 : SUCCESS in 27m 36s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1cc650d3e58847009d357443b9ecc1c5 : SUCCESS in 22m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dc71d71acd884aad9e02682a9334f115 : SUCCESS in 55m 00s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/553a5f9f980644049bb17926fffe00c7 : SUCCESS in 1h 21m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34cf3f8ff37143ec8e8566ada57a4669 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5269d845565a4285a29da3b48f493b87 : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25a2583c2c804c6d9cfff514fdb777a9 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74f5344e145483fbf2e93dbccdabe71 : SUCCESS in 9m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69629867a5cb470a9bd34b6fac074941 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c6e195a6947e4723893e82875b5d3f70 : SUCCESS in 23m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/291be562daa04e269aa3d39aac69a18b : SUCCESS in 30m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8738200cc943495fa22f71a31984d028 : SUCCESS in 56m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/659dc2556f7d4b3688253b704ee3b206 : FAILURE in 52m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c24fb7968dc145349b24e0bbb8c1e36f : SUCCESS in 1h 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"dd9631b9258cc39f45cf2d7bbf257523b7533d30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-08 17:55:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0846a3781659459c85764304fce4f85f : SUCCESS in 1h 56m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/79bff5707b6e4672a2a8ceb41fa535c7 : SUCCESS in 1h 35m 39s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dbc26428d6d44d808cb3065607c755f2 : SUCCESS in 2h 21m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a900c73aea5d469eb23671297e9ec705 : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9a50d34ba4b4fa58612666eb3037c30 : SUCCESS in 9m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9775689a99b1418380b4422532751fc7 : SUCCESS in 1h 43m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"200551d9f1c569df1e70a119aacdbb97437015a1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-10 07:51:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"71b9552bb859a2093554fc4d06ca0476b402dab7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-10 10:12:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d15dbec0c5ba845587de98025ba05c121d690f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 10:13:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5a13c7dbaf649c4c301914ff899e89cdc50446cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 11:49:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d700854b9d71449bbade9fbbee887698 : SUCCESS in 48m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/21a88251e7a64ecdb0f408ec36b527fb : SUCCESS in 1h 15m 55s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3b80fce1a5fb437db2f00a090b762b8e : SUCCESS in 1h 31m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b255a931bca04e95a593f2027de9864b : SUCCESS in 37m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ec57d8ddb4b44e19ba2b1d62f3985665 : SUCCESS in 1h 12m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/358279b24f884b32938aa719a952243d : SUCCESS in 1h 24m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/821ae37e5427432ab263e38a47341350 : SUCCESS in 9m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0df3bd7d8454c468b80a4cadaa2cfef : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d5fabd6f33f4c12b12ea3866cea0d9a : SUCCESS in 9m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35f3c49ba63b445e9925c9cd416029a9 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34a9a5415b0e4b0093633e59383c52e5 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c87a4f19be6f42698347f8adf6e1e208 : SUCCESS in 36m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5acc096998d94f33b219346cb71c45f6 : SUCCESS in 50m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6b9e10592c3445488345ce4e2e81d144 : SUCCESS in 1h 07m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ba264dad5c293375f918e0b587ee7962abbe76e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 11:49:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4888949dc2ca2145f346817fe62978fee0fcbee4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 11:50:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adffc3173bb24166b0fb26dc407edff1 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4db80b86c9c24f719f01df03df4bb776 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da33e18ac83ff4696dcc599f46c12f007ba54856","revisions":{"da33e18ac83ff4696dcc599f46c12f007ba54856":{"kind":"REWORK","_number":1,"created":"2022-03-08 15:30:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/61/832661/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/832661/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/832661/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/832661/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/832661/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/832661/1"}}},"commit":{"parents":[{"commit":"c14a366582da0bec621b103418da1668fce5f591","subject":"Merge \"Revert \"CI: temporary disable voting of Debian jobs - gate\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c14a366582da0bec621b103418da1668fce5f591"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-08 15:23:26.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-08 15:23:26.000000000","tz":60},"subject":"Install glusterfs-client in Debuntu","message":"Install glusterfs-client in Debuntu\n\nCloses-Bug: #1964140\nChange-Id: I4658a06b077d096ab9f2cb5070e39c98b5a27254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da33e18ac83ff4696dcc599f46c12f007ba54856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da33e18ac83ff4696dcc599f46c12f007ba54856"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
