)]}'
{"id":"openstack%2Fkolla~832892","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Ib79b8486f4d5064e4f249201d28cf5d6541c69ef","project":"openstack/kolla","branch":"stable/wallaby","topic":"qemu-img-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-11 20:39:24.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"Ib79b8486f4d5064e4f249201d28cf5d6541c69ef","subject":"Add qemu-img also in nova-libvirt image","status":"MERGED","created":"2022-03-10 08:22:43.000000000","updated":"2022-03-11 22:11:36.000000000","submitted":"2022-03-11 22:09:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832892-qemu-img-stable/xena-stable/wallaby","cherry_pick_of_change":831411,"cherry_pick_of_patch_set":1,"meta_rev_id":"d7e4e944b9646a3ded46dfaff0a867cec631f2d5","_number":832892,"virtual_id_number":832892,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 22:09:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-10 09:20:07.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-03-11 20:39:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-10 08:43:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-03-11 20:39:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 08:43:59.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-03-10 09:20:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-10 10:17:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad1a724b189233a1fb121989242dc6e0eda4dbd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-10 08:22:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"af6ad161b4e24ee5e78d34454c3322b521ad6d93","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-03-10 08:43:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4b585747931eb2c27942a7e08ff8e87aab50ec60","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-10 09:20:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"00b75c49cb380fc277f89006c97441ab3e4b73ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 10:17:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ce1dc8e8627f44d58c76369d2c86fcc6 : SUCCESS in 30m 16s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/360d44afb6fc426b9b6875a2f8ccfd7b : SUCCESS in 49m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/45a7a1c8e6b143bf8a41aee38feb20e7 : SUCCESS in 1h 02m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/896a527663774540a77bc69c8aeea801 : SUCCESS in 1h 25m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c279396c2da54b069c2c4a95b4136f47 : SUCCESS in 1h 15m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2215f903a6904aeb96af643f1fafb17d : SUCCESS in 31m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/90e7628d01334139a46a2edb21189b85 : SUCCESS in 19m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d3557b93739c425e999b69cfcdf527e2 : SUCCESS in 52m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9ec356d72fc4551ac2201f7ef5f8725 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8dab6217da9479aab31ead1ccecfc29 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11393612ffbb41f49a5f6d3dc4fd7405 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16152c9083e24734afaca747d5427a3d : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dda60a3c2664e0f80a440c7da639631 : SUCCESS in 10m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bad292cccc1e47589821696dc6e39196 : SUCCESS in 21m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/caf5380e551345268430ab9784ba3530 : SUCCESS in 38m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c21ed4ee6e1e4dc18071b285c0439b1f : SUCCESS in 59m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/45cea179763f434a9a4b391c7dc15d5a : SUCCESS in 56m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/71aee76538aa4fdaa4ecada643ac6b43 : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7f3478a43f040e00fa5a130b7e5a3c3252bfb915","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 11:26:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/efb3c6ab426e4c12bc3f2c868acd5a03 : SUCCESS in 2h 20m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/279da23047084b6fa5a37cff4d1ec02c : FAILURE in 1h 41m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f753923427824553b4ac8a7d6ed4e291 : SUCCESS in 2h 37m 53s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3255c6684dcd4d15b9ca18c0d90b9781 : SUCCESS in 1h 46m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7632d1ae800d2417f60153f0eea4a0be387593b8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-11 20:39:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92d7aa4bd50b34b235e213cbde3b5df29837d88b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 20:39:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89167ddf8ee4cf3f83c9b202fe5c17b05261bdaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 22:09:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/432ed04964bb49bea9a91f3ac7f48c44 : SUCCESS in 37m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1f8f7d818ca349f6b0e286dfc34a8d8f : SUCCESS in 1h 01m 33s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/dc18edafd5684d05b4ffc960e88bbfc6 : SUCCESS in 1h 27m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d7f4648e554468b786f5d67ad3e983 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cbd98a63d0641b68dd53cebfe56260e : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e87bf3f8bbc45fbb35a6084b89bffb7 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f85343bfcce24f15b600fb78a89ee773 : SUCCESS in 6m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f090a0843e0943fabc2eb48a82ed5ffe : SUCCESS in 30m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/908d392f347f4b2fbacc4084df28a456 : SUCCESS in 52m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f16686fafe634fd98af025290f6b7435 : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":1},{"id":"603487c1af35158d8f7ba29615bd71f3339256b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 22:09:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d7e4e944b9646a3ded46dfaff0a867cec631f2d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 22:11:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3528ee02beb94e5aa67151d2f940b331 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3fab9d6aefa9879a940b8716a78aba4217713fb","revisions":{"a3fab9d6aefa9879a940b8716a78aba4217713fb":{"kind":"REWORK","_number":1,"created":"2022-03-10 08:22:43.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/92/832892/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/832892/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/832892/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/832892/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/832892/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/832892/1"}}},"commit":{"parents":[{"commit":"6b77f5bfc6db190952ad399ef32f7946207b53c2","subject":"Merge \"CI: Drop Ceph stream override\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b77f5bfc6db190952ad399ef32f7946207b53c2"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-02-22 12:14:39.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-10 08:22:43.000000000","tz":0},"subject":"Add qemu-img also in nova-libvirt image","message":"Add qemu-img also in nova-libvirt image\n\nThis is noop on CentOS (it was installed as a dep already - better\nbe explicit) but installs qemu-img (and other qemu utils) on\nDebian and Ubuntu.\n\nqemu-img may be used by libvirtd to create non-raw images as it\nhappens, e.g., when Kolla\u0027s libvirtd is used by tenks.\n\nChange-Id: Ib79b8486f4d5064e4f249201d28cf5d6541c69ef\n(cherry picked from commit 0236bc6d940ddb35c0fdbaac9dc55885b972d2b6)\n(cherry picked from commit df511582268f554701ff6f3954f558f86856ba6a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3fab9d6aefa9879a940b8716a78aba4217713fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3fab9d6aefa9879a940b8716a78aba4217713fb"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
