)]}'
{"id":"openstack%2Fkolla~832899","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I4658a06b077d096ab9f2cb5070e39c98b5a27254","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1964140-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-14 12:32:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4658a06b077d096ab9f2cb5070e39c98b5a27254","subject":"Install glusterfs-client in Debuntu","status":"MERGED","created":"2022-03-10 11:54:18.000000000","updated":"2022-03-14 12:33:56.000000000","submitted":"2022-03-14 12:32:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832899-bug/1964140-stable/xena-stable/wallaby","cherry_pick_of_change":832898,"cherry_pick_of_patch_set":1,"meta_rev_id":"8b597da9b20f384967adc4529754f94746111576","_number":832899,"virtual_id_number":832899,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-14 12:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-14 10:56:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-11 09:50:02.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-11 10:08:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-14 10:56:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-10 13:16:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-11 09:50:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-11 10:08:24.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-03-14 10:56:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2041b32605f806f8b015a2a80528c7f8f2f17238","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-10 11:54:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"7c99b766dab9d356399b29a61ad67d7c70973ef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 13:16:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/44f671dbec9c458ca80ce0bf45838388 : SUCCESS in 28m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b5a799b07a83410ca4e55f95f44ec502 : SUCCESS in 47m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8a2966338db042ac997ab89d4dbdea5d : SUCCESS in 1h 02m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d14a37408ffb42a581019d9c31d79298 : SUCCESS in 1h 02m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/85e0775715f74dc381484dacb6c01628 : SUCCESS in 1h 19m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3e785fdaa03c4734adbc0e29e263201c : SUCCESS in 27m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/61ad289509824356a4bf53b37630259b : SUCCESS in 19m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/70c414b784df4b5f99cd17ef2ed9a0ed : SUCCESS in 1h 02m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c848095c4fe4a81b09c415c07dfef1f : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/213c747e05d84bb8ab8cf86580b157c1 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6195634c4dc7480caa5ab1325f0b2153 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e3b632af5024c88b7ec6bd0e97ad641 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a06d71ac07b46a5a440216d07a1fe12 : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0a51f993e0b47eeb6d83afcbb19b716 : SUCCESS in 6m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/06adab050a6f4c4f8bde2a11ddb09733 : SUCCESS in 20m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2103d82456054af488a3ec268c6481a2 : SUCCESS in 37m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36bf87f138dc45a9a75c53240bad5623 : SUCCESS in 1h 01m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bc3ae83774ca462db962dba501877f3e : SUCCESS in 49m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/be8f00e36fd34819944ced11aed4cd60 : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0a67d448a99a08492b95b619973acc907c0ec045","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 18:30:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/81262e2c7ffe475eb13bf8c37830830a : SUCCESS in 2h 20m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/31244755a5874e71a8c279409a6ef0fa : FAILURE in 1h 35m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b72bff3087c4c90a2587e9a29ac65fa : SUCCESS in 2h 15m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/537ef96f34e4477ea7880eebaf88bffe : SUCCESS in 1h 41m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e433a1a60b359448627488ed94ec92da36347684","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-11 09:50:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8bbdcf3583a3a1a963df9b615ccc087562775851","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-03-11 10:08:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fd1e30602ee1860f7ccab2ed12dc05da6f5fa81a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-14 10:56:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1477f3382ecac353dc2f0cc33e4bc702cd45bc49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 10:57:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c7fd2bdaef359f287eaec6172c920d1c974c803","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 12:32:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6baa0e5df767483ebab340bdc674822b : SUCCESS in 47m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b75b70339c21463eb47abe1298c0cff6 : SUCCESS in 1h 12m 26s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f329466dae404f70b655579185f152f6 : SUCCESS in 1h 29m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d0568d46be74cbea886d70c3d599f2d : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2973e45019274fac9f4787b78f4818b9 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a864e1d1839471e89e48c1f0ea6b54d : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b8e89fd135420f92e7e9358e8e6661 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f6a8ecf48c24606aa8d6cb1f5e9e183 : SUCCESS in 5m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/31467db3ecf247e0a79a9ca406520755 : SUCCESS in 34m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dce474ecae6949099daae05f9dadc755 : SUCCESS in 1h 01m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/59f803e35f6f4a5cb808f97f4e441279 : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":1},{"id":"944c942d0af2d6dab76e75f0bfdfe88b69ec6917","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 12:32:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8b597da9b20f384967adc4529754f94746111576","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 12:33:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e3212d42c1f4a14b8f34e5c383d51fa : SUCCESS in 1m 27s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40b4784f10ec4c92ac94376ed1449f8d : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1cd4e91ac8d1c39c663b5dfa9f7af0755221985","revisions":{"d1cd4e91ac8d1c39c663b5dfa9f7af0755221985":{"kind":"REWORK","_number":1,"created":"2022-03-10 11:54:18.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/99/832899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/832899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/832899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/832899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/832899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/832899/1"}}},"commit":{"parents":[{"commit":"76d2e589d2443540e091ecc2147015a547d207d0","subject":"[CI] Test Ironic on Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76d2e589d2443540e091ecc2147015a547d207d0"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-08 15:23:26.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-10 11:54:18.000000000","tz":0},"subject":"Install glusterfs-client in Debuntu","message":"Install glusterfs-client in Debuntu\n\nCloses-Bug: #1964140\nChange-Id: I4658a06b077d096ab9f2cb5070e39c98b5a27254\n(cherry picked from commit da33e18ac83ff4696dcc599f46c12f007ba54856)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1cd4e91ac8d1c39c663b5dfa9f7af0755221985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1cd4e91ac8d1c39c663b5dfa9f7af0755221985"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
