)]}'
{"id":"openstack%2Fkolla~833414","triplet_id":"openstack%2Fkolla~stable%2Fxena~I13e19ca29eeab40cd08fa3afe2cdf7531867f81b","project":"openstack/kolla","branch":"stable/xena","topic":"bug/1964013-stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-11 23:26:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I13e19ca29eeab40cd08fa3afe2cdf7531867f81b","subject":"libvirt: add Cyrus SASL packages for DIGEST-MD5","status":"MERGED","created":"2022-03-11 20:34:37.000000000","updated":"2022-03-11 23:34:15.000000000","submitted":"2022-03-11 23:26:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833414-bug/1964013-stable/xena","cherry_pick_of_change":833021,"cherry_pick_of_patch_set":2,"meta_rev_id":"dc6d682aab571598bb77cdc7a3676178b490df73","_number":833414,"virtual_id_number":833414,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-11 23:26:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2022-03-11 20:36:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2022-03-11 20:36:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-11 20:36:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-03-11 22:00:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6dbe1baf750673b176352ac6c99c076aacd096f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-11 20:34:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9c2d240954514465016b344752c712a2a0815b2c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-11 20:36:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf181ec46b89f66b92b59c4d058d2d1e7c58bbaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 22:00:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/429b1f965df3418dac48e12644eae820 : SUCCESS in 24m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/29596310c13045ed88a06b1fad825db6 : SUCCESS in 36m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9224319236994306b5a79dfed272f8b8 : SUCCESS in 1h 08m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/703ae0b2cecf4d56a03bee18280b08e8 : SUCCESS in 1h 19m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a9a8fb53a5624f51adbc8245b8fdcf95 : SUCCESS in 1h 14m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/498a5ccf45f54a419da643e09f4e580e : SUCCESS in 27m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/97acd55d68e544d491bc12be893bd07a : SUCCESS in 21m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f03ddd9451a541c999041f449288ee97 : SUCCESS in 1h 03m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5488eff9b9544f7bd25f65c2f749b4f : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/843d09017b1b4486a1fad068755d64e5 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c87b36979144f4986a9260ea0a33a3f : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b080300c28594080885b858a4175edb1 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d116e4ebfb45adadffc15b6ed1bc0a : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88f0a146a8f243f9830241ff73c44e2d : SUCCESS in 5m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f6bc614d75af4df99f899d4eb87c8be7 : SUCCESS in 19m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf4ae74cd9aa4ca2a7d8c8a0f147ab3f : SUCCESS in 31m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4505c259fbf64049b329a9665b6e6dd7 : SUCCESS in 55m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2be1c08776a945d586fd26abc7be8157 : SUCCESS in 49m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a0a215a7ef3c4e12849130312502203b : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f3ae778dbb6774f22e024eaa0ddb2bf8c64b099b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 22:01:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1fad5c0aa9967c9b698d238805807d4348cedcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 23:26:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5a1276a139cf46d394130b1889cf8b9d : SUCCESS in 44m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4b03a3374a7f4840b3b47986d6db9cb2 : SUCCESS in 1h 22m 01s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1896d3f25ecf4ce3943bf77a3bdd6bf4 : SUCCESS in 1h 21m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9900a08342b54ffeac78643105a2093c : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4eedb407ebd14b72bfbdedc459ee2567 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfde105933cc40809c0d1bcc12a3201f : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/571c39f2b687497b86f09a8eb11ecd7e : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1891db2ee55a4522a7d3dbcf5e2813c6 : SUCCESS in 5m 47s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/afa281a1f2de4635999413603dc233ee : SUCCESS in 27m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbab7c4df21d4089beafe9775771cd23 : SUCCESS in 56m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/72218be5fd7f44d9aacae37e90a81da5 : SUCCESS in 1h 14m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7885068cbe039d66414e4d0aca339e168d9839b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 23:26:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e2cd399c7ce133b09245327bab397413114714c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 23:28:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20d490d1c2b48be90b70dde1bf86f59 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58ff09a1e5a24aaca6a881baac3d5a3b : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1},{"id":"dc6d682aab571598bb77cdc7a3676178b490df73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 23:34:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a957d251d2774acdbbbc5941fb9686ec : SUCCESS in 2h 09m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/778111a36c0b4b428ed220f4cae8c642 : FAILURE in 1h 49m 27s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f0f867104d8145b9961f576c0a5aad9a : SUCCESS in 2h 55m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64c5c54dc2dc4bce91ac400d96fcc09f : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4655cf7fc80b4de6a73f8fcb4771aa42 : SUCCESS in 8m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8bf63001905d45d5bf6e170e2dec87ee : SUCCESS in 1h 24m 34s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"292e78312a75dc7cb7c060d138eec7d34918d295","revisions":{"292e78312a75dc7cb7c060d138eec7d34918d295":{"kind":"REWORK","_number":1,"created":"2022-03-11 20:34:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/833414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/833414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/833414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/833414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/833414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/833414/1"}}},"commit":{"parents":[{"commit":"85f512847cf93ef599ba74855e7dd8a89c6bd385","subject":"Merge \"[CI] Test Ironic on Debian\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85f512847cf93ef599ba74855e7dd8a89c6bd385"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-09 11:48:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-11 20:34:37.000000000","tz":0},"subject":"libvirt: add Cyrus SASL packages for DIGEST-MD5","message":"libvirt: add Cyrus SASL packages for DIGEST-MD5\n\nAdds Cyrus SASL packages necessary for the DIGEST-MD5 and SCRAM-SHA-256\nmechanisms. These can be used for libvirt SASL authentication.\n\nChange-Id: I13e19ca29eeab40cd08fa3afe2cdf7531867f81b\nPartial-Bug: #1964013\n(cherry picked from commit e1efa9d257c45e5f99364974ea66a6d6ba5a5605)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/292e78312a75dc7cb7c060d138eec7d34918d295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/292e78312a75dc7cb7c060d138eec7d34918d295"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
