)]}'
{"id":"openstack%2Fkolla~833644","triplet_id":"openstack%2Fkolla~master~I31ee0744574fb3db96bee2e9f545600de53d9ee4","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-17 10:44:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I31ee0744574fb3db96bee2e9f545600de53d9ee4","subject":"Add enabled flag for sources","status":"MERGED","created":"2022-03-14 15:01:23.000000000","updated":"2022-03-17 10:51:27.000000000","submitted":"2022-03-17 10:44:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"833644","meta_rev_id":"caa32a9273a9ebde55513695a90527ce33378736","_number":833644,"virtual_id_number":833644,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-17 10:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-17 09:08:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-16 20:06:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-17 09:08:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-14 16:41:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-16 15:49:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-03-16 20:06:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-03-17 09:08:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"b14ddb219e1f07fe3f40497dd61f48afc66ff723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-14 15:01:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7eaead7a3cc392263da408a219f20085eed1830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 16:41:34.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2a45de6567fb42e5a1dd6400cf8c4baf : SUCCESS in 24m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/57b3990ad13a467a8e1783481871e8c1 : SUCCESS in 50m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/02b8ad0ff1994281b2f577a21d80ed8f : SUCCESS in 1h 23m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8c08ed51581c41d694fb7739abca0cf0 : SUCCESS in 1h 23m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/28f8202da59d4173bf67ada75ee46b6a : SUCCESS in 1h 32m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5332f4e8e2574c25937cfaa8e2f0c9c6 : SUCCESS in 37m 35s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3f1003c99a9642c5a3a4eca9c1e34262 : SUCCESS in 22m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b317cb04bf594235ad5cb77060630622 : SUCCESS in 1h 02m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5868d0c7a91d442a84940aaf08d60fc8 : SUCCESS in 1h 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdc3bff88fc2408388a24a23638424c4 : FAILURE in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b79f3e35b134389a5ec5e80327e0c5c : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e76b71976f1e4311a74f82308069076d : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e28132f4064da3830e959298c470f6 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21321518760f4d21ba1be57c8eb62585 : SUCCESS in 6m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0a6f502072fa4d2a958805f6d5bdae6b : SUCCESS in 21m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b50398c06ff410baeb5451b0cb337f8 : SUCCESS in 31m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dbf985123e02484ba98e12f93e802245 : SUCCESS in 53m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5cdd910ca17742acad221ee2cc480f1f : FAILURE in 47m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2a6860bb43e64d14bbd6a335f5fc360f : SUCCESS in 1h 05m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9360b778f7c94848c4413d9a9dac0db08a77d22b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-14 18:29:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e350d00959c478eac3e94297ec17029 : SUCCESS in 2h 06m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/006d08ee98cc42e290cc5430daf5e08d : SUCCESS in 2h 10m 22s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9db388768d364dbaa308bcc6235af895 : SUCCESS in 2h 31m 29s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f22b45475bae40ab856e5c9be867ffcc : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab270e50fff44dbd972ec7b054c58e18 : SUCCESS in 8m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ce38cce9662847acb18a79eab0d3cffc : SUCCESS in 2h 00m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72c1129bf58f7e609f1445d93f9f0b0d32fb76d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-15 10:44:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d463bdc60173ff1eaa704b1bbec5c636e5fd7fdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 12:05:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fee1a57e807d4b6bbb9f052036728cdb : SUCCESS in 30m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/908bd85e56a149c1a6059902829d5c51 : SUCCESS in 38m 38s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2198054603a8456cb2d63a594339c49a : SUCCESS in 1h 12m 01s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/96009c1301ae4ea09852ae4db11885e7 : SUCCESS in 1h 08m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/adfc30cdc6054440abbf93a378e40f89 : SUCCESS in 1h 11m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b1388afcc89446898e112e00f74165b9 : SUCCESS in 38m 37s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9582f71533874850b5282c6451549c6e : SUCCESS in 18m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d3dccab596b043e98bb90377dc098b40 : SUCCESS in 58m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/450d0664a32e423ea0644d5e09f0f29a : SUCCESS in 1h 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bb51de8f0d1479b87ce01f01fed4b00 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b661c19ce8d94715a30e4c5eb5e13690 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/362ce8c06fa646ad9df5443d4e91a3f5 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b01ca08f39446dbb18ebe05721b3e3e : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa5feb9868514e918ce595816be3d0f1 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/95c6c07bc5394bd3b3219acea12f0bf1 : SUCCESS in 20m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bbddbf0ec0a84f26817e0bf270c57a06 : SUCCESS in 38m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d6bc5cfc1ac34397bcedd719134cebdf : SUCCESS in 47m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/956dc24902d340d18b9b8232c3594434 : FAILURE in 45m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f8278c99e95e49148520e037cf237177 : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":2},{"id":"8a3f9bc1efa5ea8abb09b4c386af4320d90b5001","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 12:52:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/70cb374fe4dd4cf1acef088357560178 : SUCCESS in 1h 26m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf9041fa6886478cbeeba263b8a3a765 : SUCCESS in 1h 42m 14s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3cae22cdbcef48719b340b31a60690ea : FAILURE in 2h 02m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eebcdda0fd7b4ed5bcfec9a79060a11c : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a4fc968216b40a0b22b785bb9c2ccd7 : SUCCESS in 8m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/984b865b136e4d198a6fe99c110c8233 : SUCCESS in 1h 48m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02c23aa5da647f1bc0b064ee5901945053509793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-15 13:15:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6ebdf2aad2529726f4ac9cb5a4b6509ff47659c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 14:52:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b3c14fd4a8684adf9420cd9858c1a12c : SUCCESS in 35m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fd637595ef8044dc97eb2a4d4bf5fe3a : SUCCESS in 46m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b09eeabea66648859674abaf239abec6 : SUCCESS in 1h 11m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3855659be29a4d2bae91104ade9b281f : SUCCESS in 1h 15m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0c4da1fe916049378764966f193c8f60 : SUCCESS in 1h 30m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/477d6625ae5947ebbdc6407736358219 : SUCCESS in 36m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a264c8346d7f4d2f95b7487b8b73f494 : SUCCESS in 22m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4d75627c8e034d3d858b4c76a2003c31 : SUCCESS in 1h 12m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/35d2c1ec287142249475657c58f53348 : SUCCESS in 1h 03m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce83d76a4724509bd20081d28ff41e9 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db18e68ed82d4cf894225c0e16284df5 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbd986ac5c144b8785a9e7621e8306de : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78baf05f69c24fe188a7051c75976b28 : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3a66c15911540e2bcd9ef05b6bce169 : SUCCESS in 6m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/095ab98bded348ee83e2a9ee326e452a : SUCCESS in 24m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/abd103f065e34b4ea2dbec6324f75e02 : SUCCESS in 37m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b94f17619ec43feb2eadfbcab17b923 : SUCCESS in 51m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83df22647ff741219e7ff7fab272de6e : FAILURE in 44m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/30793ea4bd5c447496d5f2128a13ba78 : SUCCESS in 1h 00m 00s","accounts_in_message":[],"_revision_number":3},{"id":"08528b5572c549c6d0c6a9eee2bbc81916cfdd15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 17:24:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3334fdf3b5d34e6b879af8d6b8049eac : SUCCESS in 2h 25m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b66d888826374b29b072def9c2020bb6 : SUCCESS in 2h 10m 30s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/42684f5997db420f99788b80d9ef818b : SUCCESS in 2h 40m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9c00ea6ea96421594ecd6f7e4d8251a : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c36f0173b0624bb1975bcef0c8ef7f09 : SUCCESS in 8m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/18821b9e46204d41850f6df874d350d0 : SUCCESS in 1h 13m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d41d6be9f7bdd1d982d50aa7e466303986d508ad","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-16 15:49:51.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d8b6de8cbf46daa7efc3b853fa26b5d57e52359e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-16 17:07:01.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"037e66ac4f83907d162fbca1a7d36d752ac521bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 18:32:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a90ebcb5205a421abd43bd09e0b98615 : SUCCESS in 33m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a31ad9e800b84ad49e29dee68b28c09b : SUCCESS in 45m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/301d25726c844e37b8821a382ac46eb9 : SUCCESS in 1h 18m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/de44eee1653d4c669a2e8821bbe0ef57 : SUCCESS in 1h 10m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/66d7d2b5d5d24f39b22da981e0eef0ba : SUCCESS in 1h 23m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/de76185a45cf4ee9a9109fbcab21b808 : SUCCESS in 39m 56s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dc23d5c488024116822f60e9cc91fab1 : SUCCESS in 17m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aa729a628bed481a99d8ff65fb56aaf6 : SUCCESS in 1h 05m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5c40bd543679467aa3624c99a7d87276 : SUCCESS in 1h 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0fa56c9505643e4bed2fdb9fe51c11c : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6709061c48aa456f8a1cca349e985033 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6d6e469d2654cec827c55120936b212 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/513851b58a7a4f019a746c2dd63f83c8 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df32129e4dde4231b4a77bd032acf33d : SUCCESS in 5m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/42690f818d8a42519be8024b5e4a0d1b : SUCCESS in 20m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/577ba1ae4b874f2d9fb11bc67f530168 : SUCCESS in 33m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/53f148e3d6844883b9a7e69782b991d4 : SUCCESS in 1h 02m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eda82db8c54e4015804dba33d5decd8d : FAILURE in 44m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ce87533ed7b41339dd6d238097434c4 : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":4},{"id":"2fc6b31fe572abbb9d80a4215f36465594ef8e1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-16 19:41:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2daaf41b37d24418aee29f3866090b91 : SUCCESS in 2h 09m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ad2f06ea14a4e84b705d493d2834360 : SUCCESS in 1h 44m 44s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a107315172e74db6926cd3e393bde804 : SUCCESS in 2h 16m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02beae84fa314e31aef0e65eb6b2f82d : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3909d61fa684a82b6ca16f1b0e08b29 : SUCCESS in 9m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c1f3a4a0767c475a96743bfe325965f0 : SUCCESS in 1h 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"be817e7f9280b7fb35fb26eae8c10ef690ed4aff","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-16 20:06:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"fd4726963068275990d0996642daedf922c9e81d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-17 09:08:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"effc8fdb37ea7d8cca583e90ac09f9a2e7399d42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 09:09:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c090c2312f99fe253a6542477aab0b0c1c7afedc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 10:44:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a0ec92f598534df28c53234f7027ac04 : SUCCESS in 38m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/48be6722572a490eb8714e62ded43d7d : SUCCESS in 1h 18m 33s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/be6c2211b8e64e2cbb0eb93adf0263d5 : SUCCESS in 1h 15m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/20d7c1d8c9ab424184be0ba2a9c4f32b : SUCCESS in 49m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5235d87285a5492c8e984e35a0ba18c6 : SUCCESS in 1h 00m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c6d0fb30807d451980ff9ff254cd21a9 : SUCCESS in 1h 11m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/949f64c2d9674bb1938fb5326dcfe824 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/861abe82f08e470a9d886383bfaca78e : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd18894c832c4e60846aaaf2b19968eb : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36b41dbda93443ce97071a6ca86e39fc : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ffb7716457e435d807cd1cad0210ea7 : SUCCESS in 5m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/266741f1fb424014b6aa9f873b3e2e05 : SUCCESS in 49m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c3372b7a0aa4abaac852c8d14ccae25 : SUCCESS in 52m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8f34459956904a839595702ddd454be1 : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":4},{"id":"24a743fcaf61026b11ca25dd6d2a5b9fba7b32fd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 10:44:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"caa32a9273a9ebde55513695a90527ce33378736","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-17 10:51:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/405e960139974065a321cb02e1fbdaca : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3b5bbeb7f6e428e838a5afc93f04b93 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c9f93c30c1fa24e4949b2c34341cbbedf00421e9","revisions":{"0ac9b5bd5d03aa290e771bf063c67db15a6c7716":{"kind":"REWORK","_number":1,"created":"2022-03-14 15:01:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/833644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/833644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/833644/1"}}},"commit":{"parents":[{"commit":"5bb6489922aab2b431be8f121465eb7f3a49b5ae","subject":"Merge \"Add prometheus-msteams image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bb6489922aab2b431be8f121465eb7f3a49b5ae"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-14 13:57:12.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-14 14:59:35.000000000","tz":0},"subject":"Add enabled flag for sources","message":"Add enabled flag for sources\n\nSometimes it might be useful to disable a source for an image, e.g. for\na plugin that is not used. This may be to reduce image sizes, or\nrestrict the dependencies necessary to build images.\n\nThis change adds a new \u0027enabled\u0027 option to each source definition. This\nallows sources, plugins or additions for each source image to be\ndisabled individually.\n\nChange-Id: I31ee0744574fb3db96bee2e9f545600de53d9ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ac9b5bd5d03aa290e771bf063c67db15a6c7716"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ac9b5bd5d03aa290e771bf063c67db15a6c7716"}]},"branch":"refs/heads/master"},"ee8a2a288a9240e7a932847df0e5bb23c8e6e1ad":{"kind":"REWORK","_number":2,"created":"2022-03-15 10:44:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/833644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/833644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/833644/2"}}},"commit":{"parents":[{"commit":"5bb6489922aab2b431be8f121465eb7f3a49b5ae","subject":"Merge \"Add prometheus-msteams image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bb6489922aab2b431be8f121465eb7f3a49b5ae"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-14 13:57:12.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 10:40:41.000000000","tz":0},"subject":"Add enabled flag for sources","message":"Add enabled flag for sources\n\nSometimes it might be useful to disable a source for an image, e.g. for\na plugin that is not used. This may be to reduce image sizes, or\nrestrict the dependencies necessary to build images.\n\nThis change adds a new \u0027enabled\u0027 option to each source definition. This\nallows sources, plugins or additions for each source image to be\ndisabled individually.\n\nChange-Id: I31ee0744574fb3db96bee2e9f545600de53d9ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee8a2a288a9240e7a932847df0e5bb23c8e6e1ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee8a2a288a9240e7a932847df0e5bb23c8e6e1ad"}]},"branch":"refs/heads/master"},"e74f63cd7258cca6a83049d32df9bd347215be9c":{"kind":"REWORK","_number":3,"created":"2022-03-15 13:15:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/833644/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/833644/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/833644/3"}}},"commit":{"parents":[{"commit":"5bb6489922aab2b431be8f121465eb7f3a49b5ae","subject":"Merge \"Add prometheus-msteams image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bb6489922aab2b431be8f121465eb7f3a49b5ae"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-14 13:57:12.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-15 13:15:48.000000000","tz":0},"subject":"Add enabled flag for sources","message":"Add enabled flag for sources\n\nSometimes it might be useful to disable a source for an image, e.g. for\na plugin that is not used. This may be to reduce image sizes, or\nrestrict the dependencies necessary to build images.\n\nThis change adds a new \u0027enabled\u0027 option to each source definition. This\nallows sources, plugins or additions for each source image to be\ndisabled individually.\n\nChange-Id: I31ee0744574fb3db96bee2e9f545600de53d9ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e74f63cd7258cca6a83049d32df9bd347215be9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e74f63cd7258cca6a83049d32df9bd347215be9c"}]},"branch":"refs/heads/master"},"c9f93c30c1fa24e4949b2c34341cbbedf00421e9":{"kind":"REWORK","_number":4,"created":"2022-03-16 17:07:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/833644/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/833644/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/833644/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/833644/4"}}},"commit":{"parents":[{"commit":"5bb6489922aab2b431be8f121465eb7f3a49b5ae","subject":"Merge \"Add prometheus-msteams image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bb6489922aab2b431be8f121465eb7f3a49b5ae"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-14 13:57:12.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-16 17:07:01.000000000","tz":0},"subject":"Add enabled flag for sources","message":"Add enabled flag for sources\n\nSometimes it might be useful to disable a source for an image, e.g. for\na plugin that is not used. This may be to reduce image sizes, or\nrestrict the dependencies necessary to build images.\n\nThis change adds a new \u0027enabled\u0027 option to each source definition. This\nallows sources, plugins or additions for each source image to be\ndisabled individually.\n\nChange-Id: I31ee0744574fb3db96bee2e9f545600de53d9ee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9f93c30c1fa24e4949b2c34341cbbedf00421e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9f93c30c1fa24e4949b2c34341cbbedf00421e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
