)]}'
{"id":"openstack%2Fkolla~833797","triplet_id":"openstack%2Fkolla~stable%2Fxena~I1fbad333b1bc95484e6f4c5145d5936a0e2db84f","project":"openstack/kolla","branch":"stable/xena","topic":"unify-curl-use","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-06-23 18:45:34.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2022-07-18 09:39:30.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-06-23 18:45:34.000000000","reason":"Change was abandoned"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-06-23 18:45:34.000000000","reason":"Change was abandoned"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-29 11:51:21.000000000","reason":"Removed by Radosław Piliszek by clicking the attention icon"}},"hashtags":[],"change_id":"I1fbad333b1bc95484e6f4c5145d5936a0e2db84f","subject":"Unify curl use (customisation options)","status":"MERGED","created":"2022-03-15 14:32:36.000000000","updated":"2022-07-18 09:40:31.000000000","submitted":"2022-07-18 09:39:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"833797-unify-curl-use","meta_rev_id":"c30ba7d9b1c4054023c8e9f5242007542e01fa89","_number":833797,"virtual_id_number":833797,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-18 09:39:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},{"value":2,"date":"2022-07-18 07:59:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":2,"date":"2022-07-13 09:27:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},{"value":1,"date":"2022-07-18 07:59:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-15 15:53:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-15 23:15:56.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-03-16 10:04:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-16 10:04:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-05-04 07:22:51.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-06-25 18:31:14.000000000","updated_by":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2022-06-25 18:31:55.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2022-06-26 14:14:47.000000000","updated_by":{"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},"reviewer":{"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},"state":"REVIEWER"}],"messages":[{"id":"6cab2b64a7f746208cfd4bfda936bae4f7dcc5f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-15 14:32:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72f373e0e22ff81f4c25449d6f7b1a5d7d2ef912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 15:53:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/630492b091214d4b976d0cf4c0693f5d : SUCCESS in 26m 03s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e38116da14ce49559248d56c339f70b8 : SUCCESS in 37m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f82f694914174101a192f7258df411ce : SUCCESS in 1h 07m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6a31f9dcec6b4d5aa961abd691c308e5 : SUCCESS in 1h 01m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ed93411ed263474787f052e34a34fd68 : SUCCESS in 1h 14m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/369298364fea4f96ab88e8a431afd086 : SUCCESS in 22m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b1f610a74e324497bb3b7451472c190e : SUCCESS in 17m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/19c9a34072cd4fa2ac76f5304eabcaf6 : SUCCESS in 1h 03m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/501cd4740d57496c92f55a30ca136d9b : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cffbd8d6ebe4ba9a070d6d1d70c51a1 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18bc2baabdce4244bc77188e62e4c39c : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eff9dc0138794f92b0dd2994d3ee90a7 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1c5b0805354d6282637337b5b2abd2 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3562c11748b640cc9ee9c46fd2c63cf4 : SUCCESS in 6m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1cfa872d47b4921b603aaa36811a48d : SUCCESS in 23m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c0920bcf83f746f091f31fcbf2805260 : SUCCESS in 28m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3ede1c9be0d4bfa895d4b34e7362e28 : SUCCESS in 46m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c8ad0c1f7067489e966e71766ac52eb8 : SUCCESS in 52m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1a9a86e884444f43b065e5d3e70f6324 : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"efe19805ffdbfb1e8d242887d62db3750d51ab28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 18:25:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/160887bcfbf44d3d828983272bff9b0b : SUCCESS in 2h 28m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c2a4847f584e4c0295b833623df306b4 : FAILURE in 1h 28m 14s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c726029ee3544b7a937f605de299db3 : TIMED_OUT in 3h 04m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8be4b9a859a74b24983fe2c4b38d529d : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ccb9b5483b1a47f49901c1de70b0e70c : SUCCESS in 8m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/07691b4622d6480681c4c35d73ad54a0 : FAILURE in 56m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9cd5cc8d945fc5297f7ab5d3864e812662b5e50","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-16 10:04:52.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1f204181771af5ad97c18df6382b30eb420167d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-03-16 10:24:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"12e4ac59a903d1f6def98fd32bd63cb5befd4194","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-23 18:45:34.000000000","message":"Abandoned\n\nthe change was stale","accounts_in_message":[],"_revision_number":1},{"id":"a648a362464a56d8b3fc4e14475ca2810b3d995b","tag":"autogenerated:gerrit:restore","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-25 10:07:36.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"ab5d47472bbb7714ea3a6b830f66b9a714138258","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 10:07:53.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 833797,1","accounts_in_message":[],"_revision_number":1},{"id":"6aa6a62bb2528f3992fd38466ec31a440057bc83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 10:07:56.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 833797,1","accounts_in_message":[],"_revision_number":1},{"id":"6a062c268c0ac731d5f7b9754226a4f5f6784889","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-25 10:09:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce8a912bcf38338e147799d0219635e9d7222c7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-25 10:26:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ea6ac08c5768a6ccc3c1a6b2970bf1d4591ee37","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-06-25 10:28:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91dca4106ba779b252939cacd73a311183d7d0e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 11:56:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9cb4a56c5cda40e7b14375fc58efd5c0 : SUCCESS in 28m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8fe045c8d27940daaf285d4134f1d81a : SUCCESS in 33m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/32d7cfcaf64744b79ea30f7f8de53669 : SUCCESS in 1h 10m 07s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ed79817abf4b4910b5cf17e1a1ecd758 : SUCCESS in 1h 15m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/429ed3e5c0fa4515a5bbab35e6f62d2d : SUCCESS in 1h 25m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2699ef14381b4af9866cf89a99af5576 : SUCCESS in 15m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/51ba0c57d29845a2b99c44a817c86bb9 : SUCCESS in 17m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/052868f0100e4b7186d1f1940d8ae82e : SUCCESS in 1h 03m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d514d6a46fd403f8d5517dc60bf0a95 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91f99f58878a42c8a6d4a0069cc49483 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/890904cce8064dca8b98290ec2e78b85 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/245a00431ae14b29aec83909c5b3909f : SUCCESS in 3m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f765eaaa3eba403da29aec964d4069ca : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25458d4fd2054e559df89bd1691da26e : SUCCESS in 3m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/172bef23d18e47d28fa72235c5448418 : SUCCESS in 19m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e5404eec459a4e648251e19b6d6819b9 : SUCCESS in 19m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/18903202b14049cc928e62178d574a05 : SUCCESS in 57m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/48ccda9f950444eab454f8133a92aa6a : SUCCESS in 55m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4379912f8c6c44608b5260b46c06d62c : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":2},{"id":"1599f912ce1a4ed0943f9a1eb483078ca4d4f5c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 12:35:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f07cdb25778446cab6c3f3d141484c88 : SUCCESS in 1h 44m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f605c08fcae94bd9ad150556460f9987 : SUCCESS in 25m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/025f3a0cb2504b899603bf8776173b5b : SUCCESS in 1h 59m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a77df4f2c38d4af29a28b88df90905a3 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be7d815e84b9486f930c5cc0a09c7f40 : SUCCESS in 4m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c62f45277a51435baea142ccb273dd22 : FAILURE in 1h 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"950bd9a66d6fb1b26b79f6ec528d6a1534dd0fa1","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2022-06-25 18:31:14.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e835be990e900a6e05f8a5f3a6ea02ea2d987415","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2022-06-25 18:31:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a9e2b2f49e48c1ae1e44d1dbfa8b58b302540213","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-26 13:28:45.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e41b4f68d05e1bc8ed800867ffe7825fe0b43fb5","author":{"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},"date":"2022-06-26 14:14:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4e6cdadf47cf6e1105dfdccdb9e3268bb96ed5d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-07-12 13:06:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dad63019ba53075cd278b983e335d6a1eda0f6c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 14:28:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be9437bb350c40968b77533b533589e4\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1e344e09e76c48dfba4f1d26187296dc : SUCCESS in 24m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3f24160c3f674aaab35d652f73280ff0 : SUCCESS in 26m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6a5df275d7be4ecebf6ddfa9c78a78cf : SUCCESS in 1h 08m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/39b9ef9604c54ca7a9d97b36b7cf57e9 : SUCCESS in 1h 11m 00s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/90e7bf744d634727b728c3c9a71f379c : SUCCESS in 1h 17m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b4c402e9aa614ce188eada79bea10348 : SUCCESS in 16m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a8a694ec83e0432db938b79f3d2ee97c : SUCCESS in 17m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e764f3dfe13c4f3f9e20a56148ef25d4 : SUCCESS in 1h 02m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042040c334fa47aeaa7d42aac192f9df : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6ad030f99434a70bc92caef88e6bf5c : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/caf43c02c1b845cca632e5bb0990117d : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/384d419d8c7549279ea89c61c25a3886 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84441efd975847389d9bfc8a639395c7 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f41a819575d249c494a300663820f412 : SUCCESS in 5m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d4550171b1504ef6bfd98f800d6fe1ad : SUCCESS in 21m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c25da13572f4ffda6197ed66f93fa13 : SUCCESS in 21m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b62aac3e73c483ca0f1d26dc0ee8937 : SUCCESS in 54m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4eb7384aeaef481cab0ba640d452ba15 : SUCCESS in 52m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f440a8d1faeb4cbbac00b59d4599cd92 : SUCCESS in 1h 05m 29s","accounts_in_message":[],"_revision_number":3},{"id":"6aa67688237835237581513b832a4035ad45a373","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:05:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d529bfab77904917aa3a1a36d63b441d\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed70a73e1db54992970207ba110cbd7d : SUCCESS in 2h 56m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d5678e8a8d604633bc0aed38294b44a4 : SUCCESS in 31m 24s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fdac4b7222a1479d817e6d5209c77f99 : SUCCESS in 1h 26m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9c51bdae4d746388dfab623dd868d02 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55c3aa7522374a05b889d2bc86c9ce69 : SUCCESS in 5m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c0191954af834d7593f9a4dee0538786 : FAILURE in 1h 35m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7f5740a316d4b48efa41d4e9d24b980058bc3915","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-13 09:27:27.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a8ce995dbe613bf62e341e462d1e8f42b9e49e1e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-18 07:59:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b817795244eea636c24a9fac798a6317dacc9e67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 07:59:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"462e34f46f4d3dab89b82cb1cd8b44f24c8f3fae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:39:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d91d2d9563ca45478aa8a7cf29c12bb2\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cd8581a9458b429ea5f922d1b0d33060 : SUCCESS in 31m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4d10bdc884034858a4b64ee7f8f56fc5 : SUCCESS in 1h 11m 11s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d295a108212d41f29b2ec9c30826a334 : SUCCESS in 1h 33m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7cf9769d18d47e3a717677fc54080e0 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7c523359a31498eaa84db7445c1c6e3 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/766636153ede4aa48baae068d845d2cc : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb15139564064bd794dba21cf315276e : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da5520fa87854a118c487a04ab7376c1 : SUCCESS in 3m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ff476594bfb34d9b97c02b265a9ba600 : SUCCESS in 17m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e9c9fe0fcdd44dbfb122be146bb50380 : SUCCESS in 58m 11s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2adc2bc5251a4ed882b01c0b39a9df79 : SUCCESS in 1h 09m 16s","accounts_in_message":[],"_revision_number":3},{"id":"9f93d34021bb4b7df592d8a906e686aec75ef0ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:39:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c30ba7d9b1c4054023c8e9f5242007542e01fa89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:40:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e428e9ff04a4f4693482749e349ddca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/808ad4514a9c4eb888769eca681c6632 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/028e9b9f73764c6f893992ac218395b1 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6486f75afb26af7a4d33aa56717ae8aca80aef9d","revisions":{"56c7dde89946627ea9854b41f46ff0eb6a1982b7":{"kind":"REWORK","_number":1,"created":"2022-03-15 14:32:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/833797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/833797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/833797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/833797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/833797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/833797/1"}}},"commit":{"parents":[{"commit":"ebb2b095587968417fa780191f06d7c9a7ac35a6","subject":"Merge \"Add Prometheus libvirt exporter image\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ebb2b095587968417fa780191f06d7c9a7ac35a6"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2020-12-16 18:26:56.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-03-15 14:31:18.000000000","tz":180},"subject":"Unify curl use (customisation options)","message":"Unify curl use (customisation options)\n\nThis change adds and unifies an ability to override hardcoded URLs for\nair-gapped environments via the usual blocks mechanism. Also, this\nchange replaces ENV with ARG instruction for the variables used only\nin building images, and uses bash variable expansion where it possible.\n\nThis change is a continuation of the I46b77978926fc2b578a68d1aaa944b2198af0685\n\nChange-Id: I1fbad333b1bc95484e6f4c5145d5936a0e2db84f\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 83efe325462e683e2c99d088012cd611d7a0b138)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56c7dde89946627ea9854b41f46ff0eb6a1982b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56c7dde89946627ea9854b41f46ff0eb6a1982b7"}]},"branch":"refs/heads/stable/xena"},"c1bb500ba5ee1995c00fd107ed58b49b798331e0":{"kind":"REWORK","_number":2,"created":"2022-06-25 10:26:11.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/833797/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/833797/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/833797/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/833797/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/833797/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/833797/2"}}},"commit":{"parents":[{"commit":"9adffe329cd296edb8ecc0a49f503e28b1ee5623","subject":"Bump skydive version to 0.28","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9adffe329cd296edb8ecc0a49f503e28b1ee5623"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2020-12-16 18:26:56.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-25 10:25:31.000000000","tz":180},"subject":"Unify curl use (customisation options)","message":"Unify curl use (customisation options)\n\nThis change adds and unifies an ability to override hardcoded URLs for\nair-gapped environments via the usual blocks mechanism. Also, this\nchange replaces ENV with ARG instruction for the variables used only\nin building images, and uses bash variable expansion where it possible.\n\nThis change is a continuation of the I46b77978926fc2b578a68d1aaa944b2198af0685\n\nChange-Id: I1fbad333b1bc95484e6f4c5145d5936a0e2db84f\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 83efe325462e683e2c99d088012cd611d7a0b138)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1bb500ba5ee1995c00fd107ed58b49b798331e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1bb500ba5ee1995c00fd107ed58b49b798331e0"}]},"branch":"refs/heads/stable/xena"},"6486f75afb26af7a4d33aa56717ae8aca80aef9d":{"kind":"REWORK","_number":3,"created":"2022-07-12 13:06:39.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/833797/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/833797/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/833797/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/833797/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/833797/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/833797/3"}}},"commit":{"parents":[{"commit":"51d5e442c105bd5dd74c983ef277812b8915eb52","subject":"Bump prometheus-openstack-exporter version to 1.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51d5e442c105bd5dd74c983ef277812b8915eb52"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2020-12-16 18:26:56.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-07-12 13:06:06.000000000","tz":180},"subject":"Unify curl use (customisation options)","message":"Unify curl use (customisation options)\n\nThis change adds and unifies an ability to override hardcoded URLs for\nair-gapped environments via the usual blocks mechanism. Also, this\nchange replaces ENV with ARG instruction for the variables used only\nin building images, and uses bash variable expansion where it possible.\n\nThis change is a continuation of the I46b77978926fc2b578a68d1aaa944b2198af0685\n\nChange-Id: I1fbad333b1bc95484e6f4c5145d5936a0e2db84f\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 83efe325462e683e2c99d088012cd611d7a0b138)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6486f75afb26af7a4d33aa56717ae8aca80aef9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6486f75afb26af7a4d33aa56717ae8aca80aef9d"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
