)]}'
{"id":"openstack%2Fkolla~834692","triplet_id":"openstack%2Fkolla~master~Ie90f13a568b4120531bbbf5982d2cbbf1f9c5995","project":"openstack/kolla","branch":"master","topic":"toolbox-ironicclient","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-23 15:54:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie90f13a568b4120531bbbf5982d2cbbf1f9c5995","subject":"toolbox: Add ironic client","status":"MERGED","created":"2022-03-22 14:32:55.000000000","updated":"2022-03-23 15:56:12.000000000","submitted":"2022-03-23 15:54:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834692-toolbox-ironicclient","meta_rev_id":"0af1d31bcac5803fc88cb13b4c0a1ae304ee3c47","_number":834692,"virtual_id_number":834692,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-23 15:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-03-23 14:07:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-03-23 08:23:02.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-03-23 14:07:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-22 15:58:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 08:23:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-23 14:07:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"7babf83547d7abdc5a351d95aa35a15d06e0d202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-22 14:32:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be4f308944d13c96a2456ca46781ee1f17524ffe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 15:58:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/94da7738489b4c109a75eed5512d21e0 : SUCCESS in 48m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0a70b74e87474fbf95c34f868b9dbd3f : SUCCESS in 49m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d5e75b6e1cfb462491280aa2750445d8 : SUCCESS in 1h 16m 53s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/df0955fda63e4263aff94c3624551be3 : SUCCESS in 1h 14m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/32fe16eb7bae4b369b8879e8eeb84d5f : SUCCESS in 1h 22m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1121abddb42c4cf3909aa0e9c4410ff0 : SUCCESS in 35m 07s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/67bf4e7407ca49cdb6c4be552562ed23 : SUCCESS in 19m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/50b3beda1a9745458047903f95b87231 : SUCCESS in 1h 02m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5542c2cae71c4fd39375df7df8aa1f20 : SUCCESS in 1h 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37e2376384564b74b8dd622b0ba34a34 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9cefc54e3ac4586980f050c859356c5 : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bfafe8be19d42298423d32ecd9956a8 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5376bd963fe4475db2ef27ff8b41c763 : SUCCESS in 8m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fa2de38bdcb444458ba3f10a7a9ef945 : SUCCESS in 21m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c35ec73620947fc912a0f6d49cec941 : SUCCESS in 35m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/78eaa55118e4457ab4172f8685b4a8ac : SUCCESS in 56m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4ceedfcbdd594a72b2c3e876ca56b0ef : FAILURE in 46m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7b5748c488614e35be71b2a482cb3fb0 : SUCCESS in 1h 05m 38s","accounts_in_message":[],"_revision_number":1},{"id":"089204d8b36782db86f0108680a6f7e4c9af5c8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 17:13:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0bcf02db29014ff797e5f00da35fabbe : SUCCESS in 2h 37m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2180c977362145fcaef889fc3866a77c : SUCCESS in 1h 43m 08s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/73a2b471e8454e6bb7a90c926b8f0e53 : FAILURE in 2h 07m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d40e187ba9a34f13a03006c78736371c : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46df8a78211c4b3fbb5f0e151450acf3 : SUCCESS in 8m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/61284aa8c1ea412ebf584be8e914c199 : SUCCESS in 1h 54m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4115b49dc703624cda6a49fefbff3d77eeeecc4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-23 08:23:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f216d8bfc5aa2bea58039a81185dc08e512319d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-03-23 14:07:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01b2bee0e616b065cd60050ed0f123835d91acf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 14:08:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd3e2915be0ce00b1bfe9fd25427aacfc051e7a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 15:54:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9e8018de34d244b8894d59c9cd8f7d56 : SUCCESS in 46m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/59071288b4b24f0f91aaed3e348a948d : SUCCESS in 1h 21m 52s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1c3613c50a52466bb39825013044663d : SUCCESS in 1h 40m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/97de60824edc462e805bcdae9558c286 : SUCCESS in 37m 12s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/82c4d720067c492c8e970952343ad7fb : SUCCESS in 55m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0e0cda854092464ea13b76ff7ed52371 : SUCCESS in 1h 15m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a74ba6bc68ca46e2b85a6db541858645 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d5ce4aa31dc4c2f93e84f397b517c86 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4df001268ff1485c9a3d7cd0c7ca23c5 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37989e63a83a4e0694b0e8c23f2d1a90 : SUCCESS in 7m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1588e7b7075a439aa6a35556c20d6318 : SUCCESS in 36m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/74f813fd32de4f05870afb13e2e381f4 : SUCCESS in 56m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/72949962f4ee4548aac5bacdca0dc109 : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ba4c4390069ae1c6e39c66f8f7c20b857e73ce62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 15:54:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0af1d31bcac5803fc88cb13b4c0a1ae304ee3c47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 15:56:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab1d241516994602985c62b94730651d : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac371617c9f124e9115ad8112020254477e11f24","revisions":{"ac371617c9f124e9115ad8112020254477e11f24":{"kind":"REWORK","_number":1,"created":"2022-03-22 14:32:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/92/834692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/834692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/834692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/834692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/834692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/834692/1"}}},"commit":{"parents":[{"commit":"5cc480a617c68ed20c80bffec16321a937f5a32b","subject":"masakari: add Cyrus SASL packages to monitors image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cc480a617c68ed20c80bffec16321a937f5a32b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-22 14:32:13.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-22 14:32:13.000000000","tz":60},"subject":"toolbox: Add ironic client","message":"toolbox: Add ironic client\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/832159\nChange-Id: Ie90f13a568b4120531bbbf5982d2cbbf1f9c5995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac371617c9f124e9115ad8112020254477e11f24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac371617c9f124e9115ad8112020254477e11f24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
