)]}'
{"id":"openstack%2Fkolla~835341","triplet_id":"openstack%2Fkolla~master~If666bb4aedd7bb4e3041f0672c3b70e1bb5d36b6","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-03-28 15:42:45.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-03-28 15:42:45.000000000","reason":"Change was submitted"},"16312":{"account":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"last_update":"2022-03-28 15:42:45.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-28 14:03:42.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-03-28 11:43:00.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"If666bb4aedd7bb4e3041f0672c3b70e1bb5d36b6","subject":"Revert \"Revert \"Update code to Jinja 3.0 style\"\"","status":"MERGED","created":"2022-03-28 11:39:58.000000000","updated":"2022-03-28 15:46:35.000000000","submitted":"2022-03-28 15:42:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":805897,"submission_id":"835341","meta_rev_id":"1439816f6710c4754a9dcc44aa51145324df4111","_number":835341,"virtual_id_number":835341,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-28 15:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-03-28 14:03:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-03-28 11:43:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-03-28 14:03:42.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"}],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-28 11:39:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-03-28 11:39:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-03-28 11:39:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-03-28 11:39:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2022-03-28 11:39:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b20017885cca91fb493597a779fe91b696b7b32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-28 11:39:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"119fd73fc0fad9cf2ab0ded57242a301adb8ac3d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 11:43:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3e34730bcb5fa9fc2f8a51f52d8583073178845f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 13:18:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b76989611e2e453cb2b181fccb3c06ec : SUCCESS in 30m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d720b8eeff9f467896c1afedb7e2169d : SUCCESS in 35m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/eb8882f1e795469b85a00e43da4a1128 : SUCCESS in 1h 21m 54s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8083d5d134eb4c82964f4a084f8e2cc4 : SUCCESS in 1h 15m 48s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/42a357e291db49fa917c2ceb33176bf5 : SUCCESS in 1h 33m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/31a0e3ff78c446918a2bf76a62669c7a : SUCCESS in 48m 40s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cee7a97d6a354a51a2f1e58e41ad0680 : SUCCESS in 19m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3872961737c64fd9818b1f2722429ca7 : SUCCESS in 1h 10m 03s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/01eea1faefb44bae91d14dd5b9f472a6 : SUCCESS in 1h 15m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c0691756824f05b80447b455ecc091 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4aa7ed42576244bab659ca19da983481 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a379e150329472f983449332a42727b : SUCCESS in 5m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0a1e72bae3c54acf9a506d3beffb610a : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28a42ec9e2934705aee2273f29e82460 : SUCCESS in 9m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8ff5d3abe2e4b588e5d3621dcae16b5 : SUCCESS in 24m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2db4e57255e94975a2fc1417f13a7039 : SUCCESS in 44m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/905f4a1c46d649129f356b120a725f69 : SUCCESS in 57m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/be43d859c2eb47588dd806ac9a1126e4 : FAILURE in 51m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5fcf50c1518b480e8a1a933b42e3792c : SUCCESS in 1h 09m 57s","accounts_in_message":[],"_revision_number":1},{"id":"05d5c96f776c92f33360341ae9b01d083474a1b2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-28 14:03:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b76c4d99f995a64ef95bbdd573b9d3d92951f38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 14:04:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b2178a3c591e74a8f33da4f30ff401f2c77b14d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 14:27:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ee9a9b37b87144c8a890b062639defa6 : SUCCESS in 2h 16m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e33b38fdb5144e64987a189ab5fffdd8 : SUCCESS in 2h 01m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c7ec0cc9b0148ce85041d6f7837f217 : SUCCESS in 2h 09m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc7475b868c949e09d736dcddcaeed9a : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e37dc3ebecd4eababd1ad1281b88d6b : SUCCESS in 7m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/975a67b41ccc459ca87349434aa49811 : SUCCESS in 2h 45m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e304bc7738200294e930f01f8bab125d2f16231a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 15:42:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bb178c6b7ae14136b2500d4e82443227 : SUCCESS in 41m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7e9375ec872b4541b3752e543a1640be : SUCCESS in 1h 20m 04s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5632e14fbb5f41b7a196233189ca2f91 : SUCCESS in 1h 22m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/02469e7be6ba4f7cb88b3e06b80e1aa6 : SUCCESS in 39m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0a0b439638f646e089056446d0852459 : SUCCESS in 58m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b42e75c82d774612b5472faeaca7ae0b : SUCCESS in 1h 08m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef3e2ffafb5943cea17a238a709e7c87 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7891ba0b02fb4fcbbe4046572710096a : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e88fd31226334d8e9fff35296bdf9428 : SUCCESS in 5m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c56b35de5aa44b39d6eac2dbbd16a47 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb1b2ddecb5848719c02b94ceaafeeed : SUCCESS in 7m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5015a52f81d14cbebad42bbf2e33e324 : SUCCESS in 37m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0cf4b8982ac4f7286aeb77414545268 : SUCCESS in 51m 51s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8156e9dcc2c24bf8967a20145d1292ae : SUCCESS in 1h 00m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4491f0eecdc06fece3b567d4059ebd18b46f7c6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 15:42:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1439816f6710c4754a9dcc44aa51145324df4111","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 15:46:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d1354afac194f2baef19a873fe50364 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16c230fb114d81a2a09187580875f8b9959ab8d4","revisions":{"16c230fb114d81a2a09187580875f8b9959ab8d4":{"kind":"REWORK","_number":1,"created":"2022-03-28 11:39:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/41/835341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/835341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/835341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/835341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/835341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/835341/1"}}},"commit":{"parents":[{"commit":"25e9602845fca7ca5b7a5c55fe5b71be0b137f04","subject":"Revert \"Update code to Jinja 3.0 style\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25e9602845fca7ca5b7a5c55fe5b71be0b137f04"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-28 11:39:58.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-28 11:39:58.000000000","tz":0},"subject":"Revert \"Revert \"Update code to Jinja 3.0 style\"\"","message":"Revert \"Revert \"Update code to Jinja 3.0 style\"\"\n\nThis reverts commit 25e9602845fca7ca5b7a5c55fe5b71be0b137f04.\n\nReason for revert: we updated to newer Jinja.\n\nChange-Id: If666bb4aedd7bb4e3041f0672c3b70e1bb5d36b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16c230fb114d81a2a09187580875f8b9959ab8d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16c230fb114d81a2a09187580875f8b9959ab8d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
