)]}'
{"id":"openstack%2Fkolla~835343","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I5efab66e487e06abd1a56af97d7e7caa1ebc880d","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-19 09:13:09.000000000","reason":"Change was abandoned"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-19 09:13:09.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5efab66e487e06abd1a56af97d7e7caa1ebc880d","subject":"Use jinja2.pass_context instead of contextfilter","status":"ABANDONED","created":"2022-03-28 11:48:07.000000000","updated":"2022-04-19 09:13:09.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"cherry_pick_of_change":835342,"cherry_pick_of_patch_set":3,"meta_rev_id":"d768e0b2b0c9603dc45b261dc9c73f3980bee45b","_number":835343,"virtual_id_number":835343,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-28 20:42:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"date":"2022-04-05 16:27:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-04-05 15:20:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-28 12:08:22.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-03-28 13:13:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 15:20:59.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"4cb53834a0fef12726619cf6b43eabae1407714a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-28 11:48:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"a9053476710ac5e109a27e309a79f1f477faadbf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 12:08:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fd1b70cf7e9691ced9c6a78958eb5be2539c0dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 13:13:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/82f85a921cc847dcaf1d4a0408c4105b : FAILURE in 33m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b4704500a80c4f14a1b0952fef94d431 : FAILURE in 38m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5c5d4df778674f56a76ab7b1a00a752a : SUCCESS in 1h 15m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/14584fb3aa58493f98eb413bcf30dbbe : SUCCESS in 1h 10m 18s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/65d651a7bc244d399b201f02be9ad2ff : SUCCESS in 1h 22m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/efda740519e3431dbfda2362f3281415 : FAILURE in 5m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1519a22a45664653926e2fb0423f30aa : FAILURE in 5m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1567a3e7e37a4a90a36b139e31cb62aa : FAILURE in 10m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91841894360c4215a562c6edead4e67f : FAILURE in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fb3de230bdf4b738a879e2a0ecaaa86 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/278148978d5e4130bea8dc1ac8cc9748 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5836d65679a4c0eaa88e970deca4c7d : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e24ec2363673474eb119cf333375a0da : FAILURE in 5m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e55059e1319647cf9abb3d6e501d2653 : FAILURE in 5m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c16da407f914e1db5e3a7000b14d6c5 : FAILURE in 5m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/10134eb2d7da419cb871f98ba70bf951 : FAILURE in 9m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5b4c539f70fe4768bab820de705bb5a2 : FAILURE in 10m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a50e13bf867d4465b2af55bdb0432797 : FAILURE in 13m 36s","accounts_in_message":[],"_revision_number":1},{"id":"711615fad8d5585669f9a179b109f58abaedab4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 14:35:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/21d4e3d9c98e4cd3b65ac1165addcd6c : FAILURE in 2h 25m 39s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bf353c75fabb4e21b2a924d2df464351 : FAILURE in 7m 49s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/63cc91921e5042339991c25932e1c550 : FAILURE in 15m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f80632d8a3824af6917ce506285779cc : FAILURE in 6m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c899b828c401f0b654f857c5a30551dbfb4b8c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 17:25:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"75572d4519a16c2f6e2726ba7ec86db6c2dc9e64","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 17:25:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"70137aa206a381b4a4eae56800ccedf5f3bef79b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 19:03:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5478dbaff67445bf9a841c5811d4eda6 : FAILURE in 30m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ab92ce26c9b048688ed05190701ba1f2 : FAILURE in 35m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3b140ec658604ea6beef723f878f4d87 : SUCCESS in 1h 31m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/dd33a58bd85e47d18009ca8e527b0d9b : SUCCESS in 1h 12m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2ac30cdede9b4d8880f19151eb7253fb : SUCCESS in 1h 19m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/da6729cedde14ad08214dd66a5817e30 : SUCCESS in 23m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/89358bc9ccfa4bc7a5e1d823bea1a502 : SUCCESS in 32m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/60b7f60a6e6b4afeaebe0a604153b0f7 : SUCCESS in 51m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd19c439f1d54ef88aaab8c41d462f54 : FAILURE in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b233038125dc43ef86de414cd72c62b7 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9f71d8b87e14cb4985607c4fd9e3dde : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17d0517a8f646dfa6e58aa6bc9c039a : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6375b4f1e5df4ae8bab9fa121e2b703a : FAILURE in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/76090267a5054987b8e1f313344f6cdb : SUCCESS in 21m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a9df7a6adef64589ae9b61f6ee5e4bd5 : SUCCESS in 27m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f9b102fbd579418aa9521c2b5bc33d9d : SUCCESS in 54m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4121845661242528a655a49075e8d54 : SUCCESS in 56m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/beab145117ee4b11bbeb79a5d7830d9b : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":2},{"id":"8eb57fea6aad6ca048dcbdb758e0d57dcf8fc3e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 19:17:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"db999a5d3376f3b64be3ea6ed5e995eecea1f6c0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 19:17:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"453ce8d0bdba58b9f73802405246b38dc998a745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 20:42:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/dc19aea7aaa84a169076fccab79e6349 : FAILURE in 34m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8bb466daafb4429eb1a0805130a10790 : FAILURE in 30m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5fe11a6e07e4491b9a4cd84d659889f7 : SUCCESS in 1h 17m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e3d9517bd4ae4aabb1ca86a872033257 : SUCCESS in 1h 15m 02s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7602a684acf54e9d82a1f329144b18f1 : SUCCESS in 1h 23m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/387255debe2b48a5a2350d7d13fc3b75 : SUCCESS in 18m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4082950399a845aea31b753d3432107e : SUCCESS in 20m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/97f11136b04b4467a47352c268d52181 : SUCCESS in 52m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56c34bd5e144db5896172ff9e355010 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07e8c0a900f04a63b0e819b16f742695 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bade2ae28a64ff99f0cdeab755665ac : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7d1b86afae543389bd0245a6b7bd2fe : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e53568bb28e496a920d33132789873e : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb93bd2ffc0f48e1b00ac2ac69e69179 : SUCCESS in 41m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/beddc52c878e4df29857dbf014160c1f : SUCCESS in 27m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe19621d6c1b4f738cb40fd8f5567d29 : SUCCESS in 1h 05m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e0baa9befb754bf0abf37ad855322795 : SUCCESS in 54m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/700ab500726548db8fc528e99c17082d : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":3},{"id":"526bc0b2960668b3f94465ff13ce350817d67920","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 22:20:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/078f4f7b2c264dfaa4b59721d34f1b05 : FAILURE in 2h 38m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c6aa0cf0dd674d49bf5375df07556307 : SUCCESS in 1h 44m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/14f0db0a9480450689b62e05c6604099 : SUCCESS in 2h 24m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f9ba3b61b7ce4630bf9bdb05f4cc7eea : SUCCESS in 1h 44m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5a536c1824766c15bb0d6c9be360136828c136df","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-05 15:20:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"207b4495695ab21c3167e6e54d8e2c0fa837c2dc","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-05 16:27:34.000000000","message":"Removed Code-Review+2 by Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"d768e0b2b0c9603dc45b261dc9c73f3980bee45b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 09:13:09.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/kolla/+/836591","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"252f3ade922cf5367388cf7b1d8747809c4bfaa9","revisions":{"677346f149d287fc0db6bb586ab22b91a4c9f984":{"kind":"REWORK","_number":1,"created":"2022-03-28 11:48:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/43/835343/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/835343/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/835343/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/835343/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/835343/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/835343/1"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-28 11:42:51.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-28 11:48:07.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (filters.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change has been updated to also support Jinja2 2.x releases,\nsince the Wallaby upper constraints specify 2.11.3. In practice, most\nusers will not use UC to install kolla.\n\nCoAuthored-by: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I5efab66e487e06abd1a56af97d7e7caa1ebc880d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/677346f149d287fc0db6bb586ab22b91a4c9f984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/677346f149d287fc0db6bb586ab22b91a4c9f984"}]},"branch":"refs/heads/stable/victoria"},"f06b5468a0ba80d9a907af3e4085b72217787bc9":{"kind":"REWORK","_number":2,"created":"2022-03-28 17:25:50.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/43/835343/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/835343/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/835343/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/835343/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/835343/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/835343/2"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-28 11:42:51.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-28 17:25:50.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (filters.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change has been updated to also support Jinja2 2.x releases,\nsince the Wallaby upper constraints specify 2.11.3. In practice, most\nusers will not use UC to install kolla.\n\nCoAuthored-by: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I5efab66e487e06abd1a56af97d7e7caa1ebc880d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f06b5468a0ba80d9a907af3e4085b72217787bc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f06b5468a0ba80d9a907af3e4085b72217787bc9"}]},"branch":"refs/heads/stable/victoria"},"252f3ade922cf5367388cf7b1d8747809c4bfaa9":{"kind":"REWORK","_number":3,"created":"2022-03-28 19:17:05.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/43/835343/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/835343/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/835343/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/835343/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/835343/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/835343/3"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-28 11:42:51.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-28 19:17:05.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (filters.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change has been updated to also support Jinja2 2.x releases,\nsince the Wallaby upper constraints specify 2.11.3. In practice, most\nusers will not use UC to install kolla.\n\nCoAuthored-by: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I5efab66e487e06abd1a56af97d7e7caa1ebc880d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/252f3ade922cf5367388cf7b1d8747809c4bfaa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/252f3ade922cf5367388cf7b1d8747809c4bfaa9"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
