)]}'
{"id":"openstack%2Fkolla~835345","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I5efab66e487e06abd1a56af97d7e7caa1ebc880d","project":"openstack/kolla","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-06-15 11:11:39.000000000","reason":"Change was abandoned"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-30 10:08:29.000000000","reason":"Removed by Radosław Piliszek by clicking the attention icon"}},"hashtags":[],"change_id":"I5efab66e487e06abd1a56af97d7e7caa1ebc880d","subject":"Use jinja2.pass_context instead of contextfilter","status":"ABANDONED","created":"2022-03-28 11:48:32.000000000","updated":"2022-06-15 11:11:39.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":835344,"cherry_pick_of_patch_set":2,"meta_rev_id":"767a90cd9b53cca1057be37aa5a0f1346b53347b","_number":835345,"virtual_id_number":835345,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-28 20:36:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2022-03-28 19:17:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-28 12:08:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-03-28 12:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-19 15:03:22.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"}],"messages":[{"id":"348b7e17fb0be4cecff125d870c2678e3be1fc28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-28 11:48:32.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"d42f338683da3c29e20b212c218cc8c255171ca3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 12:08:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"854883b2c910c7a8c2c7ae03bb4df245cf015f1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 12:58:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e8d157c801004c8ba9c21e30b5abefce : FAILURE in 7m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/00553838a25f45058cc043540543ae14 : FAILURE in 7m 35s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cccfd06eae444d829cf3bd2d95afd021 : FAILURE in 34m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d39d5ae839dd4462ac32931656cec60f : FAILURE in 10m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cc974778c93a4db0b9128f1f746c31cf : FAILURE in 10m 30s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2ca9fb43d3a441dd9561ab01d0cef501 : SUCCESS in 1h 03m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6ace3f91f8c440298e019a498842c41b : SUCCESS in 58m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4f3707d702dc4e52833f6905299e43e2 : RETRY_LIMIT in 4m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/76e12f3dd75e4733a0274104baf6f5b7 : RETRY_LIMIT in 5m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/159688a34bbf476193315eda55801c7c : SUCCESS in 59m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f553274f7d647328276363a1cb33e55 : FAILURE in 7m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e9e0cb908f85451fb82dd859fbcd4788 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb15ee49ae5a49ef82fa5f9a6bdc6c61 : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40475e5962244b1d9cbbceabab1a76e3 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf8f633c71645dbb03ee2311f3d499f : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/404b241caa9d479782a98a21bdfb435a : FAILURE in 5m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e31d39688afa40068a9ce83ae7eaa58a : FAILURE in 21m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0163b157a3bd4bf8acfc14f2893f1d95 : FAILURE in 23m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9add1dfbdb1446dbad85e53540687a28 : FAILURE in 16m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/771f80da4651412d8e4a0f278df5e901 : FAILURE in 19m 26s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"e455b399bc70032e17ed5795ae36ed7bdca11565","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 17:26:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"43b1039ae893789d40af26c917e8a6370a3287f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 17:27:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3c06e74bb5d6954cd981921288e0b506c3d8309a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 17:27:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c25555d7145bd7e0fb052b958d93e7a138d1e481","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 19:17:31.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"15c4d8f1b045adf3629145bdce6f8122fd44e6a1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-28 19:17:36.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8eb22876315e2a77d46bb2ac47ac5873b094cf32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 20:36:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2ce673c79e2c425d901a3f72e77677e5 : SUCCESS in 50m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0f8ba50853fc418980107a35eaeeb70e : SUCCESS in 52m 48s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e65c0a2a08f245b98346e0e6fa133818 : FAILURE in 47m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/356488ee9ed24ec79d615cc588f85dfb : SUCCESS in 1h 07m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/414ee3050c5042edacdc9271ffcfb02d : SUCCESS in 1h 15m 59s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9d92b4991d3c4a4289acbfe6a9f73bfb : SUCCESS in 1h 06m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a5593b72d40a4a2f8ce83d170de0c818 : SUCCESS in 45m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c8c83d02314e4551bb45a8ac9ae81fb7 : RETRY_LIMIT in 4m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0c6175eedbf34671aaa2f949ad721628 : RETRY_LIMIT in 4m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9e8b4def7a614a78b86cfc1849f26e17 : SUCCESS in 55m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a93bd5295694ce78c5e33b5f346042c : SUCCESS in 6m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/749111a5c87847819ad6c355f8ff7117 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee4f37d2600541638871f0d6aa6ee6eb : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33d5e9beff3b45c19964cf8d258810e4 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/935adbe9448941c3a7bd68646893073f : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edf1250537ea4784a7f4db8400098cd2 : SUCCESS in 5m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71ac5556fa6948bca83b5076922089b4 : FAILURE in 40m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d889ad0862784b268515014dc958c61c : SUCCESS in 25m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0fbd874e4ed44178ade111c5a5fcefca : SUCCESS in 54m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/da54ac9fb6c54fa7bf166d636dc2a75c : SUCCESS in 1h 00m 10s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"0aaf005c0cb6c83b6a231962934553d44edfb1f6","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-19 15:03:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0a9d7c7a3c4c794bb101e3427c00b0bd26b04a40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 16:31:30.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e5f4cbe10d4c4651986b50b187a47e79 : RETRY_LIMIT in 4m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4e2df83197e14569950258b0882c151e : RETRY_LIMIT in 2m 58s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6fd7ab10a7c54b2b96b35f9d4d8af36c : RETRY_LIMIT in 3m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5c430eee282a42b4a374e4ce5e9526b2 : SUCCESS in 1h 20m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b9679d164b71453a968e4168e4f58791 : SUCCESS in 1h 21m 31s (non-voting)\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/86ce40d1ba314754b9e3351db14982f1 : SUCCESS in 1h 03m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fb04d8d25be248f795f927c138b8ebec : SUCCESS in 57m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/214555fd39414e198a0aa74e63b548b4 : RETRY_LIMIT in 2m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dad3feada9e1463987e56829c33a2586 : RETRY_LIMIT in 3m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ad05d752ff7346cab72b7fcd8ed8b414 : FAILURE in 13m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7945ec4fc0d4030ba7b9767963c97b8 : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/370d5ba59c1b4f318e09fd98b9c91006 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdd53031c0834e6aa55f28ebd7c46142 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59683b3e077443478ed5dcbe32ed8e65 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52501c5141334805859b8be057d7eec7 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/246a24f0e4814e60b3604af92f2fcfad : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4cb8e338860e42b2bee65e3752044f27 : FAILURE in 26m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/16c37701bd3f494ab5cfc7c5aeea4d3f : FAILURE in 22m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/df2e51081ad2455190717c1cf5ac4d84 : SUCCESS in 47m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6f8c0b7005ba42bab65788c1ba284fc5 : SUCCESS in 46m 40s (non-voting)\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":4},{"id":"767a90cd9b53cca1057be37aa5a0f1346b53347b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-15 11:11:39.000000000","message":"Abandoned\n\ntrain is gone","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"48c28005d7f5aed8b4b21c9dbe5e502dee86b0b2","revisions":{"139c0b4ff8abdb3243aecb278d747c67f649e619":{"kind":"REWORK","_number":1,"created":"2022-03-28 11:48:32.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/45/835345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/835345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/835345/1"}}},"commit":{"parents":[{"commit":"d967f400e2761e9022a6b2dedc2d307d32d1ea27","subject":"libvirt: add Cyrus SASL packages for DIGEST-MD5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d967f400e2761e9022a6b2dedc2d307d32d1ea27"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-28 11:42:51.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-28 11:48:32.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (filters.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change has been updated to also support Jinja2 2.x releases,\nsince the Wallaby upper constraints specify 2.11.3. In practice, most\nusers will not use UC to install kolla.\n\nCoAuthored-by: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I5efab66e487e06abd1a56af97d7e7caa1ebc880d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/139c0b4ff8abdb3243aecb278d747c67f649e619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/139c0b4ff8abdb3243aecb278d747c67f649e619"}]},"branch":"refs/heads/stable/train"},"c43abdae5af8c6935c5f0bff465680e53b821027":{"kind":"REWORK","_number":2,"created":"2022-03-28 17:26:17.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/45/835345/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/835345/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/835345/2"}}},"commit":{"parents":[{"commit":"d967f400e2761e9022a6b2dedc2d307d32d1ea27","subject":"libvirt: add Cyrus SASL packages for DIGEST-MD5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d967f400e2761e9022a6b2dedc2d307d32d1ea27"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-28 11:42:51.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-28 17:26:17.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (filters.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change has been updated to also support Jinja2 2.x releases,\nsince the Wallaby upper constraints specify 2.11.3. In practice, most\nusers will not use UC to install kolla.\n\nCoAuthored-by: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I5efab66e487e06abd1a56af97d7e7caa1ebc880d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c43abdae5af8c6935c5f0bff465680e53b821027"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c43abdae5af8c6935c5f0bff465680e53b821027"}]},"branch":"refs/heads/stable/train"},"f8d18aa7e69524065cc5f5919b0e0299fef80b9b":{"kind":"REWORK","_number":3,"created":"2022-03-28 17:27:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/45/835345/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/835345/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/835345/3"}}},"commit":{"parents":[{"commit":"d967f400e2761e9022a6b2dedc2d307d32d1ea27","subject":"libvirt: add Cyrus SASL packages for DIGEST-MD5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d967f400e2761e9022a6b2dedc2d307d32d1ea27"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-28 11:42:51.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-28 17:27:09.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (filters.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change has been updated to also support Jinja2 2.x releases,\nsince the Wallaby upper constraints specify 2.11.3. In practice, most\nusers will not use UC to install kolla.\n\nCoAuthored-by: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I5efab66e487e06abd1a56af97d7e7caa1ebc880d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8d18aa7e69524065cc5f5919b0e0299fef80b9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8d18aa7e69524065cc5f5919b0e0299fef80b9b"}]},"branch":"refs/heads/stable/train"},"48c28005d7f5aed8b4b21c9dbe5e502dee86b0b2":{"kind":"REWORK","_number":4,"created":"2022-03-28 19:17:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/45/835345/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/835345/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/835345/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/835345/4"}}},"commit":{"parents":[{"commit":"d967f400e2761e9022a6b2dedc2d307d32d1ea27","subject":"libvirt: add Cyrus SASL packages for DIGEST-MD5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d967f400e2761e9022a6b2dedc2d307d32d1ea27"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-03-28 11:42:51.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-03-28 19:17:31.000000000","tz":0},"subject":"Use jinja2.pass_context instead of contextfilter","message":"Use jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (filters.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change has been updated to also support Jinja2 2.x releases,\nsince the Wallaby upper constraints specify 2.11.3. In practice, most\nusers will not use UC to install kolla.\n\nCoAuthored-by: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I5efab66e487e06abd1a56af97d7e7caa1ebc880d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48c28005d7f5aed8b4b21c9dbe5e502dee86b0b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48c28005d7f5aed8b4b21c9dbe5e502dee86b0b2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
