)]}'
{"id":"openstack%2Fkolla~835846","triplet_id":"openstack%2Fkolla~master~I8bdce3b76bc02fdcc029f90cf1984e6b3d4eb961","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-03-31 17:26:11.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-03-31 15:46:50.000000000","reason":"Marcin Juszkiewicz replied on the change"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-31 17:26:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8bdce3b76bc02fdcc029f90cf1984e6b3d4eb961","subject":"[release] Use Yoga sources by default","status":"MERGED","created":"2022-03-30 15:07:58.000000000","updated":"2022-03-31 17:28:04.000000000","submitted":"2022-03-31 17:26:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835846","meta_rev_id":"195a1fcf86f7eb05588a3f3af457ad849c4e3bcc","_number":835846,"virtual_id_number":835846,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-31 17:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-03-31 15:46:50.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-03-31 13:13:18.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-03-31 15:46:50.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-30 15:08:37.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-03-30 15:13:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-03-30 16:32:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 13:13:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"0ff8690355b5e717ac4409ef7a5ae04f5f88dcad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 15:07:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f20e57d68593e0842ebdd8fac63d6e5d5f31384","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-30 15:08:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b17ef4e81228d56042ebf46bbc0fa460d205603","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-30 15:13:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b838e25937bcbaed03ffabfc095995db14a443db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 16:32:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/666bbd19cb6642a2b1cb6c0d628dc752 : SUCCESS in 36m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/583e712b492248ef8847372471370564 : FAILURE in 37m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0c35656ad131414b82e1ce6eb7d98995 : FAILURE in 29m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6d1aac332e724ed69b9dc00d516dc3e6 : SUCCESS in 1h 17m 46s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/af0e3dc25181435ca362d0c7050dc02f : FAILURE in 29m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b542edec9ee949c9a017461971a24fd0 : FAILURE in 30m 19s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f005b74ebf144c0a9d10448201daf211 : SUCCESS in 17m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bd8747c3ee66421d8c585026d4be6eed : FAILURE in 24m 57s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d887d593e8ad42f494b34a1e5987192e : FAILURE in 21m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8836b7ebeceb4df1989d5eae99d180e9 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec9c6a743a284143b60c9ba90d56c61b : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4360dce2674c40339c7394d555d89277 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c3944a416a94538bc6b76a2d6188a2c : SUCCESS in 6m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/78445d18a3344d4f8923a4b1036d4fc9 : SUCCESS in 19m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/716fab11d2ed45ef894ed956a362470e : FAILURE in 24m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c6f1574fb22482eaf4c765d76b3df31 : FAILURE in 19m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3dbc25f4669b4c23b013139301b5bfe6 : FAILURE in 47m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/38ebd0488bbe48a887c9e75da82cda1d : FAILURE in 18m 22s","accounts_in_message":[],"_revision_number":1},{"id":"47e7d1ee394141102c28d48c0b802cbf72bd9968","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 18:22:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c455e59b1acb4a0991fd6328fd98f49f : FAILURE in 2h 43m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3bf1183fbf2a4e6b847bed10e0a49c3e : FAILURE in 2h 00m 11s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c7e57f89f94472bbdb52c6cfbfa53f0 : FAILURE in 1h 12m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ce0d305a4a24bb39585f5de21f0c2b7 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34cff8ce18c64656af46f69aa96b79ac : SUCCESS in 8m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dac11e1818904af0a9773224d4fb425e : FAILURE in 1h 34m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f09321e1db43167f18a1f9410df697a780e03bd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 09:25:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f62006786711d02276cc65072afa820b55f1485c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 09:26:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"14ab009641953dce8d02040682031ccb1537ff6d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 09:32:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9f753f232ea4dc93cc2034ccf33879acf4226c8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 10:57:17.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a17ad5f4b03c4fc9a7af84cf1853c37e : SUCCESS in 40m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/31a405b98cdc421785ddab8746f4e19f : SUCCESS in 45m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/49880ba990cb441c9eab1eb2b5471787 : SUCCESS in 1h 25m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e941d6af769f4af9bef9a03148ef8da5 : SUCCESS in 1h 13m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8ff97e4287a648a783467689fd38be61 : SUCCESS in 1h 21m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5a702ef674d346c69ff2b8e1974dd7b1 : SUCCESS in 38m 22s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fa2f3ca4f1724132a1f9a503c7d50487 : SUCCESS in 18m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4a9f89ce5e31490cb99cdd9311b9041e : SUCCESS in 47m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f8e47a14919e40e58bc3a207f0559225 : SUCCESS in 1h 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67cff36026dc41efa70d46e0713d5f01 : FAILURE in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0a6fe3705244435bf4cc497788c3f47 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adfedfd605144f69a6ffab26ba2c27df : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eef15156fa84f12b4fc8a6325f94f98 : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d4365516a884441eb2ce0c345d30c583 : SUCCESS in 19m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/56c8cbbb495449c998e00a1944c06457 : SUCCESS in 37m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/63379701cb814b308b3c81e7e5e9940e : SUCCESS in 50m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bcca973086b64def9765d95e867c8ff9 : FAILURE in 42m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f49b51b09b164b2da76bbcf3d7738cf1 : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":2},{"id":"40116aac01b3bf8429eb2db312bded11df2c9851","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 11:41:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f644fd8f9d7e4c6ea63cd53412427b2a : SUCCESS in 1h 57m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f886429cd7794f08ad3090fb51177531 : SUCCESS in 1h 42m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/edf6aea904e4455a8060fbff4cfb2cae : SUCCESS in 2h 02m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb8904d5aa7545f682d1776828f69fa6 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6df2706c52bb4c11beaec2d7d4b9c75c : SUCCESS in 6m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c46b83b4b1a34bbe8f6489c79fb28b6a : SUCCESS in 1h 19m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32e17d8fa621c9b300d0a904977f675ce86720fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 13:11:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce188e0b215211388a2419bac72b36ec8226630c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-31 13:13:18.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9e573274a45dbd3ae6a04afa3d421d02b2e6f6a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 14:49:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/15b52c8923ac4bc3bed4433ae58b10bb : SUCCESS in 35m 56s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b888cfa41fbd4ea3b400aee0fd374f2f : SUCCESS in 41m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/03cda60c90b64d2a8c42b6a03eac2bf9 : SUCCESS in 1h 09m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d2877ce7af3846888bf45f3f2685064c : SUCCESS in 1h 08m 42s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/17b002893f06474d9a35be05aa85e0d9 : SUCCESS in 1h 30m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6945491f60314e559bd7b4801899f0ee : SUCCESS in 35m 00s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d6edc6af62fe4598ba4add915a8bd595 : SUCCESS in 19m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/07a04a57c0d64c7186b9260cc58ad52e : SUCCESS in 57m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/04956f88f91a490ab58927b3309384d8 : SUCCESS in 1h 08m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c468a24e6b9447d18151b4beb05f34bf : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0f5e5f444d54f10a2cd6501d46c6c0b : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea2d40d68eca4a59891c4e62afafb507 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d31f2421024918915e220014fd6c93 : SUCCESS in 5m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/269e518d7ff443f3866ea353d4303660 : SUCCESS in 19m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c7038852dd247898d45e00996a31b5d : SUCCESS in 42m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d8415b4e0d848c3a700a27ae7d04079 : SUCCESS in 50m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ed617d681944ad0a1cdef1bffa9fbcd : FAILURE in 45m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4ea305f19b924c6e92790ccb787bab88 : SUCCESS in 1h 03m 39s","accounts_in_message":[],"_revision_number":3},{"id":"86984dacff1db75ef9cbcad9f5cce16562dd7350","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-31 15:46:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f9d4a8d85ed4e8b3978ebe5d794380a4e2bc98c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 15:47:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fe8dea94fe03b9a2591d5350b2dd03bd9074be48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 16:38:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3418ee46725c4de9b6e6578c49897dce : SUCCESS in 2h 14m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e7d62e22d9c14b70b0114c66a4b9f1d6 : SUCCESS in 1h 36m 49s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f4e36c080e2446858894bcafb89bc58c : SUCCESS in 2h 06m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac33399f07904bca907106fe124b26d9 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/468bd597013149c2a447d941dce19944 : SUCCESS in 6m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d1502df909f94624a3989fa10c929892 : SUCCESS in 1h 43m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f26c4f5a04f2efb03ab98546fad19562a00cd1f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 17:26:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4013f10b8c194a59b3af066c20ae6135 : SUCCESS in 46m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1295c3a4306d4eb99752b3bd91586928 : SUCCESS in 1h 18m 20s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/418b2f901e40459fb44adc657d98b9cf : SUCCESS in 1h 36m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b53a314bba704248b0855c9addb260ff : SUCCESS in 29m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/23100524e9644b4a84788eadf931e427 : SUCCESS in 1h 02m 10s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b58430b1022d4927bba7313754d6094b : SUCCESS in 1h 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969a836381404b7e99180090489a87c2 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47eb01de95674d20815491d9d9808281 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70a059a964ca41db8294bf136a00d546 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ac0998c0f149b98beec353c48dd156 : SUCCESS in 6m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1b10cc9e58e4542973dedd92d7bd9ee : SUCCESS in 53m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8968baa509db49008b2361127be01e9d : SUCCESS in 55m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3dc4761344eb4315b66ed62adae9e449 : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":3},{"id":"e0ced40da2df2eefbcdad79569557322bc7c59b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 17:26:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"195a1fcf86f7eb05588a3f3af457ad849c4e3bcc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 17:28:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293f485b3518463092bbcaa6b2f6effa : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bb65c5667423707dc92b3cdb3993523ccca0f150","revisions":{"ab02ddf51a7f8b147d1478202e333305f70e3afe":{"kind":"REWORK","_number":1,"created":"2022-03-30 15:07:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/46/835846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/835846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/835846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/835846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/835846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/835846/1"}}},"commit":{"parents":[{"commit":"c572fc61d02071a4fc5a950d12b783e9b8a998ae","subject":"Merge \"Revert \"Revert \"Update code to Jinja 3.0 style\"\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c572fc61d02071a4fc5a950d12b783e9b8a998ae"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 15:05:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 15:05:15.000000000","tz":60},"subject":"[release] Use Yoga sources by default","message":"[release] Use Yoga sources by default\n\nChange-Id: I8bdce3b76bc02fdcc029f90cf1984e6b3d4eb961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab02ddf51a7f8b147d1478202e333305f70e3afe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab02ddf51a7f8b147d1478202e333305f70e3afe"}]},"branch":"refs/heads/master"},"38d3605166c79198aa704760d982838f248fe433":{"kind":"REWORK","_number":2,"created":"2022-03-31 09:26:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/46/835846/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/835846/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/835846/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/835846/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/835846/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/835846/2"}}},"commit":{"parents":[{"commit":"c572fc61d02071a4fc5a950d12b783e9b8a998ae","subject":"Merge \"Revert \"Revert \"Update code to Jinja 3.0 style\"\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c572fc61d02071a4fc5a950d12b783e9b8a998ae"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 15:05:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-31 09:24:01.000000000","tz":60},"subject":"[release] Use Yoga sources by default","message":"[release] Use Yoga sources by default\n\nThe following projects do not have stable-yoga tarballs yet:\n\n* monasca-api: stable/yoga exists, but no additional commits yet, so no\n  tarball. Use 8.0.0.\n* neutron-vpnaas-dashboard: stable/yoga exists, but no additional\n  commits yet, so no tarball. Use 6.0.0.\n* venus: no stable/yoga branch yet. Not in releases repo. Use master\n  branch.\n\nChange-Id: I8bdce3b76bc02fdcc029f90cf1984e6b3d4eb961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38d3605166c79198aa704760d982838f248fe433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38d3605166c79198aa704760d982838f248fe433"}]},"branch":"refs/heads/master"},"bb65c5667423707dc92b3cdb3993523ccca0f150":{"kind":"REWORK","_number":3,"created":"2022-03-31 13:11:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/46/835846/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/835846/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/835846/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/835846/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/835846/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/835846/3"}}},"commit":{"parents":[{"commit":"c572fc61d02071a4fc5a950d12b783e9b8a998ae","subject":"Merge \"Revert \"Revert \"Update code to Jinja 3.0 style\"\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c572fc61d02071a4fc5a950d12b783e9b8a998ae"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 15:05:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-31 12:55:30.000000000","tz":60},"subject":"[release] Use Yoga sources by default","message":"[release] Use Yoga sources by default\n\nThe following projects do not have stable-yoga tarballs yet:\n\n* monasca-api: stable/yoga exists, but no additional commits yet, so no\n  tarball. Use 8.0.0.\n* neutron-vpnaas-dashboard: stable/yoga exists, but no additional\n  commits yet, so no tarball. Use 6.0.0.\n* venus: no stable/yoga branch yet. Not in releases repo. Use master\n  branch.\n\nChange-Id: I8bdce3b76bc02fdcc029f90cf1984e6b3d4eb961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb65c5667423707dc92b3cdb3993523ccca0f150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb65c5667423707dc92b3cdb3993523ccca0f150"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
