)]}'
{"id":"openstack%2Fkolla~835847","triplet_id":"openstack%2Fkolla~master~I69a1cda22bda2b10ae66f8c5ade39f4d38b714b2","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-03-31 13:13:54.000000000","reason":"Removed by Radosław Piliszek by clicking the attention icon"}},"hashtags":[],"change_id":"I69a1cda22bda2b10ae66f8c5ade39f4d38b714b2","subject":"[release] Use RDO\u0027s Yoga Delorean packages","status":"MERGED","created":"2022-03-30 15:07:58.000000000","updated":"2022-03-31 18:14:20.000000000","submitted":"2022-03-31 18:00:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835847","meta_rev_id":"c17c217c3b935cf2dcf99199e9dd34992e6c0b42","_number":835847,"virtual_id_number":835847,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-31 18:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-03-31 18:14:20.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-03-31 13:13:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-03-31 13:13:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-30 15:08:48.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-03-30 15:13:27.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-03-30 16:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09255fdee001f1ff7ebad7cd94662a9bb0dba91a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-30 15:07:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a4b748107dce815896063370182d9e203a6e81e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-30 15:08:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c5db5e601a83aad02c060ee5a49fe2015e48d7c2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-03-30 15:13:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"416d4f737b99a21f8888421871ecb2a1dfbdf945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 16:22:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ea17c4046991485794b1aea85ab6acf0 : SUCCESS in 26m 59s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9d91a40c33384491b528c2f2b6d2bc34 : FAILURE in 38m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d39b175825bb4ae4aef96e1214c3c176 : FAILURE in 27m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c3f73140f7bd47f5bf409a4376d3bce5 : SUCCESS in 1h 08m 19s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/55de133c59ea49ca8f27f6f0226cc232 : FAILURE in 27m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bc8d356631a047eaade533b821f71e84 : FAILURE in 27m 56s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7bf9b68057354ad8b2dab6f4abd70e4f : SUCCESS in 19m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a29c3194ea864c8b9233f592f97689ff : FAILURE in 19m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/64091e5800c24a9e800696637a3f8539 : FAILURE in 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e09b01725e64fe098a9930f992be5a1 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7298adaedf642e9a6dbfecce25ebe46 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/170344662dec4fc4aaa8a3a5f42c9f62 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42de292381164561a7791227a3e2c8d5 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/351ac2ab192a47a4b88ef76fc688d4c2 : SUCCESS in 22m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/14f97143ca56464ca35df537acad3f98 : FAILURE in 32m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/91f3ff6f2075437cbaef3f146387c697 : FAILURE in 20m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/547ee3dba91b47e6a528c37a323a9dd7 : FAILURE in 43m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/66840a8b3d49495e99bc3335b03c2fb2 : FAILURE in 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5796bd709cb81fade4ee47d0e878e025fd51e145","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-30 18:43:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f65b238f63b4ab5a26d4639b0a0f116 : FAILURE in 2h 41m 11s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/600020cfb1ca4867a2c8776417eefa2c : FAILURE in 1h 52m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb40a547b53b4cf2a762678c7d2c8dc0 : FAILURE in 58m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a67246c8867749c98335edf6e9f097a5 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6c26e4525c946488368cb0079cf354b : SUCCESS in 9m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2b274943f33f4e65a4ec649b77c33fbe : FAILURE in 1h 48m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e19f76037a9946ea7665ff83e9114c4d2a03b0bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 09:26:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8ba3375473ad78805599eea1ab9fea51a4b9a82a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 10:49:28.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/51747d51c94548e0ae411bc8b8ad6a28 : SUCCESS in 33m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1d3baac9ce864702945c71f99869c08d : SUCCESS in 46m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ebd5cba6405a49ee8445565224054c83 : SUCCESS in 1h 09m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c85a08b5912b4e2baccf2f3d067dc5f8 : SUCCESS in 1h 17m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1056cce620e6461f82e1ec67ed8d326c : SUCCESS in 1h 11m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/089840ae5d4b49c29999591735c649de : SUCCESS in 39m 06s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/217050d7005d4a1c8365d129013295e9 : SUCCESS in 19m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8ad39e7cc3034fab811edd336fb45dee : SUCCESS in 57m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3cde4ec66cf74127b056e3b59d7c6d38 : SUCCESS in 1h 07m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcf7bd66d4f746b6a17b4acb51ccf718 : FAILURE in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d0144585c3b4ffcac2760efbe20aacc : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f1f7fd228b14e3bb07b7cd752e07472 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc39ba6ac2e244baadf68f8e0c4d9700 : SUCCESS in 6m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/79d6613b76b149b2ac02fce4565451ca : SUCCESS in 20m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e5bc204929df40ddbb27337f23c6ed23 : SUCCESS in 32m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4047a771c88f48e68eaafaa73d20ef45 : SUCCESS in 52m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/272c42d64f9c4949ae4e4a0665c0c7cd : FAILURE in 58m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/558b2fbef93e4d45bf7a8aa540c94de6 : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":2},{"id":"e256f649b8bce3852288637b6f1f4f2cf736675f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 12:18:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8e8bff0ab87490a9404d04187c2a8ce : SUCCESS in 1h 59m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/461dc74f6b5d4898bd6fdeff9f9bf7dc : SUCCESS in 1h 27m 59s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ea717586c034926bb5da61611efa125 : SUCCESS in 2h 02m 01s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1949cdbb4c3443db9fcaf2228bd4b0dc : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/809a8490934240bb9d69df71dadb3ec4 : SUCCESS in 6m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3005852c70e24deaa97ec2a532d4b973 : SUCCESS in 1h 30m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f28a39d274ad761e018c355f56f160501a8c064f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-03-31 13:11:02.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c7fe402e1417b7098af07118eb49b4679301f559","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-03-31 13:13:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1eacbc86d77a0e027ccb303ba8ed143478e3ee23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 15:05:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1dd6ba9f5c5b4f8bb3c4aceae4bb0f49 : SUCCESS in 31m 41s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c0b702ce68944b108405ea33a86042fe : SUCCESS in 45m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/22b81710e9664f85a11ab99f51db7065 : SUCCESS in 1h 21m 59s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b8eb96aa167845df886c8f53f79cc837 : SUCCESS in 1h 23m 37s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/48f2ab4b96fa4e879abb735787c91bc9 : SUCCESS in 1h 35m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1d462c29f4c2411c937c7e2a53c955cb : SUCCESS in 39m 33s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6bebb6adb6304e62b7ba9bf7a0826a7b : SUCCESS in 19m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/086d4cf6aa334d5091941379b39d252e : SUCCESS in 1h 05m 50s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/251ed4c81c3f4c379f56790995dcb3c6 : SUCCESS in 1h 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d80cb19b0ee4468926b5eb13af85b8d : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0f581c6e85f404e98a625ad9c96d02c : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/600afdac90954bf981e63d99f3c013f8 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6976dbac23764206870d8f476f958116 : SUCCESS in 6m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2b69cc46d80d4332ab1cee2dce4501c5 : SUCCESS in 19m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/079fc95660fb40029b2760d63968c0a6 : SUCCESS in 31m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de860a985612404aa200a66371411748 : SUCCESS in 48m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2398049a62084f649d9f93fb771867a6 : FAILURE in 41m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bf511e677ee344ca83af689d88f13a9a : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":3},{"id":"9cfd0e2d1b22a37c08727e97886f832270b13921","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 15:47:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b0e3b282fe2453b23b1127eebb82a2433fe08156","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 18:00:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/bca14f15cc0b41ec82e2796ee7be597a : SUCCESS in 42m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d43f299e7c2b4abd80727dda06e7c7cd : SUCCESS in 1h 16m 34s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0a30a15217f040d8ab7c52b922b5606a : SUCCESS in 1h 27m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/66b0ae6969d34adba631caff4cbda60c : SUCCESS in 30m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a05b59fa1def4b649926e52382adfe01 : SUCCESS in 1h 10m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0d048fd9326241f2ae86c3953b86401b : SUCCESS in 1h 23m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7654cf0917c4315ace4358cceed1f5b : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17fa8000a43e401d9db9e6f22b04fd3c : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85d6c19bcc88441793ec350a9c5011ea : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bcfb5eb91e146dbb0c648207b159f97 : SUCCESS in 8m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/40bfc247d615414aa2ed7bc0c2102b7f : SUCCESS in 34m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e21bb338a86a4e29a33f7b12c12c6dc4 : SUCCESS in 59m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c763f44306fc40398be8e2ac79127400 : SUCCESS in 1h 09m 44s","accounts_in_message":[],"_revision_number":3},{"id":"bac666746072369346fb10911d1ca1ac51ed35b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 18:00:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ceb89111b65a6dca4dbbf59efddd624196936680","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 18:02:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb9bb67739e41cbb0df0407d6209ab1 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":3},{"id":"8930b6725d181b2369c31ed018f6b3e85d2df98c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 18:14:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/70788868d98c44eaa867221b98d809ec : SUCCESS in 2h 34m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bd783b235ba8412cb4404e8733e8f2a5 : SUCCESS in 1h 44m 12s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b6993e58c534ae48cba528ed4e11f52 : SUCCESS in 2h 07m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cecf2aed651a44cf8476fa1bd1784b3a : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4d06dfa27354bcb9518b6d9003b4a16 : SUCCESS in 7m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/15a9118a321d4d6581c5c7ac72d3a0a0 : SUCCESS in 1h 57m 45s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2c4b43e562f9cfa86cf7a1d9ca5eaae6ded2095f","revisions":{"42e4f5251905ef6c5b928f40943b74cf92a2a6e2":{"kind":"REWORK","_number":1,"created":"2022-03-30 15:07:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/47/835847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/835847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/835847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/835847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/835847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/835847/1"}}},"commit":{"parents":[{"commit":"ab02ddf51a7f8b147d1478202e333305f70e3afe","subject":"[release] Use Yoga sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab02ddf51a7f8b147d1478202e333305f70e3afe"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 15:07:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 15:07:39.000000000","tz":60},"subject":"[release] Use RDO\u0027s Yoga Delorean packages","message":"[release] Use RDO\u0027s Yoga Delorean packages\n\nThis is a temporary solution until centos-release-openstack-yoga is\navailable.\n\nChange-Id: I69a1cda22bda2b10ae66f8c5ade39f4d38b714b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42e4f5251905ef6c5b928f40943b74cf92a2a6e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42e4f5251905ef6c5b928f40943b74cf92a2a6e2"}]},"branch":"refs/heads/master"},"5bd626876ab7cd1f67c8ed808b70b087d9dc275d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-31 09:26:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/47/835847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/835847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/835847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/835847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/835847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/835847/2"}}},"commit":{"parents":[{"commit":"38d3605166c79198aa704760d982838f248fe433","subject":"[release] Use Yoga sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38d3605166c79198aa704760d982838f248fe433"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 15:07:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-31 09:25:04.000000000","tz":60},"subject":"[release] Use RDO\u0027s Yoga Delorean packages","message":"[release] Use RDO\u0027s Yoga Delorean packages\n\nThis is a temporary solution until centos-release-openstack-yoga is\navailable.\n\nChange-Id: I69a1cda22bda2b10ae66f8c5ade39f4d38b714b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bd626876ab7cd1f67c8ed808b70b087d9dc275d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bd626876ab7cd1f67c8ed808b70b087d9dc275d"}]},"branch":"refs/heads/master"},"2c4b43e562f9cfa86cf7a1d9ca5eaae6ded2095f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-31 13:11:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/47/835847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/835847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/835847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/835847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/835847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/835847/3"}}},"commit":{"parents":[{"commit":"bb65c5667423707dc92b3cdb3993523ccca0f150","subject":"[release] Use Yoga sources by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb65c5667423707dc92b3cdb3993523ccca0f150"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-30 15:07:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-31 12:55:34.000000000","tz":60},"subject":"[release] Use RDO\u0027s Yoga Delorean packages","message":"[release] Use RDO\u0027s Yoga Delorean packages\n\nThis is a temporary solution until centos-release-openstack-yoga is\navailable.\n\nChange-Id: I69a1cda22bda2b10ae66f8c5ade39f4d38b714b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c4b43e562f9cfa86cf7a1d9ca5eaae6ded2095f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c4b43e562f9cfa86cf7a1d9ca5eaae6ded2095f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
