)]}'
{"id":"openstack%2Fkolla~835987","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f","project":"openstack/kolla","branch":"stable/wallaby","topic":"healthcheck-disable-curlrc-stable/wallaby","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-05 09:35:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f","subject":"Fix disabling of curlrc in healthcheck_curl","status":"MERGED","created":"2022-04-01 10:44:53.000000000","updated":"2022-04-05 09:38:11.000000000","submitted":"2022-04-05 09:35:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835987-healthcheck-disable-curlrc-stable/wallaby","cherry_pick_of_change":835957,"cherry_pick_of_patch_set":1,"meta_rev_id":"becff24fb1f38956bd91f4323cf4060c5c380550","_number":835987,"virtual_id_number":835987,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 09:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-05 08:07:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-04-01 13:57:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-04-05 08:07:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 12:01:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-01 13:57:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-05 08:07:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"ae589743443b3ac6424e2130eb1c75284ba4ed75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-01 10:44:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ebdf6362570bd341183e9f90fe5d5b081212711f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 12:01:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/be519a635fab4f549d6885a633ac4675 : SUCCESS in 29m 11s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6803302363ed463da5196616e269046a : FAILURE in 49m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/808f24da40ab4d5db056a1d1386e6cbf : SUCCESS in 1h 06m 16s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/59863864f08343ad9570afe2845b7a9a : SUCCESS in 1h 02m 47s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/857972b757924be8a3660500dc19e6d5 : SUCCESS in 1h 11m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/93f86bb3c5824dbf8fcba9724e156d87 : FAILURE in 3m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b3f3e485d6494d0d9da40e23cda4dbed : FAILURE in 3m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/de3de5f348274268a6fae0618a00d0bb : FAILURE in 14m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec72790ed5414db6904eb370958b1202 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ce310cfc77b947e6a85c4a36a53d27c6 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6a7de84d06a42398e4bf2ef5b605224 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e84cae0d5e1242a29dcdb62993e1d6d2 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a29db0b36c6477099a1753e4bca1932 : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b526886d69840b992b60d17e6dba171 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dde3a1f330784562a41352265d910018 : FAILURE in 3m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/98cfc6a845bb4988b7c77fa50e9a22bd : FAILURE in 2m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/08d3d4950f8a431aac43b116086830ea : FAILURE in 9m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5b33d7bc2e484564961f7d457230580d : FAILURE in 10m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee9ac21cbb20440d9ec94c53d7032deb : FAILURE in 7m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7cb89b1f7b797c7407c46e577bd45c69be37c16a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-01 12:41:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5a059a1cd8e577f6c89a5b8b4c47d27db86e98c4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-01 13:57:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"35ab3a48a7c840745e1c5bb93bdfd49412ab8086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 13:58:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2ffaa1b33bf14ff7aa40c9ed76bac577 : SUCCESS in 28m 12s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cca68d61683d4e3c87e14af5e6b0b587 : SUCCESS in 49m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/635d82c3cb004120aff629446f397749 : SUCCESS in 54m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3e8bd159baa44d9b9e197b4fd6953e7f : SUCCESS in 1h 03m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c04353141b2d4b8d9ba6e712d0d77158 : SUCCESS in 1h 03m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/98d46188a43e41548c2d760cc5edc555 : SUCCESS in 25m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/37d068996e1543e79f1b5d45fb790f43 : SUCCESS in 18m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fc0edcd9c94a468381f58f32acb15ff0 : SUCCESS in 1h 00m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a872709ad5247cbbf6c7bfbc44b89f2 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/612c50b9485a4056a8419b0931b4c5c3 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f19db53c33841d0a51d317bb6bcf2b9 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eea846abf4fb476b86a7b4f7e19fd123 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16cfb22972474f2ebfffb1a1c1bf2762 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf360b9bda1940829b1f091cdd6eb16a : SUCCESS in 4m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/007831a3a2f041d99b244aebb6c14c2b : SUCCESS in 20m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9538aecba6d543b8bdfa2be7ea70c471 : SUCCESS in 39m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/63078dcd85e5449ea4844f523a909d55 : SUCCESS in 58m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d689701592f4fe3b6bbb7f104df3e39 : SUCCESS in 52m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d8be0d42037d4b6c8dcb6f7e37618671 : SUCCESS in 1h 08m 28s","accounts_in_message":[],"_revision_number":2},{"id":"db2757a9310d149ee7e8aa2083c2cdb236c13db5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 16:40:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dfc69b28ff634b728f975e81836ed1ed : SUCCESS in 2h 18m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/085d2ba55f1346a191342fce4deecd99 : FAILURE in 2h 03m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a9aeced36894c0c9a2a121c75973575 : SUCCESS in 2h 13m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ac2635ac9bad4daea89e4b1c46e4c91f : SUCCESS in 1h 56m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4cd5f9fd34431e223850435365823b25d699c55f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-05 08:07:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0dc6baafffba3527db564fdadfe570ab8ef3c962","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 08:08:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5b9b294f4c47a47b0352e46355e97a9e4337599f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 09:35:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fc5b692dd84244aea4a7bb6c8c9055cb : SUCCESS in 36m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/adcfaaa276bd4592ba2a48bfb129786e : SUCCESS in 1h 02m 12s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/02b6b8b8a84e4b3f9c9843558616c62f : SUCCESS in 1h 22m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b1c9b041923443eb4a52bcd1093bac7 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1e8aa63f6434033a5db01f17c01292d : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea4dbe815b074e6da6f734cf6f0012b2 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/865db3fb30534e8db54ad3adf001e765 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16bd67dc41d04a6ebb37452755fd106e : SUCCESS in 3m 55s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8039e776d3124a2eacf37a33f5669d93 : SUCCESS in 32m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6eacfcf3a82e49fab933364ec3d0e59c : SUCCESS in 55m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/be344eb87134468e9f57541d80680951 : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":2},{"id":"5503eb23486fd501eaa31f3e28a762e69ffadb6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 09:35:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"becff24fb1f38956bd91f4323cf4060c5c380550","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 09:38:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27849cba5ebf4c8f8a6237d05a706155 : SUCCESS in 1m 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0a99f7197e241bfac0db2dd50050205 : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f3a066673a2e536f6d51eb5a6f9abf19ee30f083","revisions":{"c82f1f620593e4e711ff0d8330a1155e93ac63ad":{"kind":"REWORK","_number":1,"created":"2022-04-01 10:44:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/835987/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/835987/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/835987/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/835987/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/835987/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/835987/1"}}},"commit":{"parents":[{"commit":"ba22f9e6c2936528f875b062d9986cca02c2da45","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba22f9e6c2936528f875b062d9986cca02c2da45"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 10:24:04.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-01 10:44:53.000000000","tz":0},"subject":"Fix disabling of curlrc in healthcheck_curl","message":"Fix disabling of curlrc in healthcheck_curl\n\nThe -q (aka --disable) option only works if it is the first parameter on\nthe command line.\n\nChange-Id: Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f\nCloses-Bug: #1967272\n(cherry picked from commit 196b742ba271fd7e3521448f83e8a2914f5c7dc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c82f1f620593e4e711ff0d8330a1155e93ac63ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c82f1f620593e4e711ff0d8330a1155e93ac63ad"}]},"branch":"refs/heads/stable/wallaby"},"f3a066673a2e536f6d51eb5a6f9abf19ee30f083":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-01 12:41:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/835987/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/835987/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/835987/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/835987/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/835987/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/835987/2"}}},"commit":{"parents":[{"commit":"41b43807bcd77e330fe255fd5e1fdff428fa45c9","subject":"Use jinja2.pass_context instead of contextfilter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41b43807bcd77e330fe255fd5e1fdff428fa45c9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 10:24:04.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-01 12:41:43.000000000","tz":0},"subject":"Fix disabling of curlrc in healthcheck_curl","message":"Fix disabling of curlrc in healthcheck_curl\n\nThe -q (aka --disable) option only works if it is the first parameter on\nthe command line.\n\nChange-Id: Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f\nCloses-Bug: #1967272\n(cherry picked from commit 196b742ba271fd7e3521448f83e8a2914f5c7dc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3a066673a2e536f6d51eb5a6f9abf19ee30f083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3a066673a2e536f6d51eb5a6f9abf19ee30f083"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
