)]}'
{"id":"openstack%2Fkolla~835988","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f","project":"openstack/kolla","branch":"stable/victoria","topic":"healthcheck-disable-curlrc-stable/victoria","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-19 13:53:21.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-19 13:53:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f","subject":"Fix disabling of curlrc in healthcheck_curl","status":"MERGED","created":"2022-04-01 10:45:15.000000000","updated":"2022-04-19 18:07:24.000000000","submitted":"2022-04-19 13:53:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835988-healthcheck-disable-curlrc-stable/victoria","cherry_pick_of_change":835957,"cherry_pick_of_patch_set":1,"meta_rev_id":"63ceff2cd6a0d46b1c2fb6267ae308258af734c4","_number":835988,"virtual_id_number":835988,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-19 13:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-04-19 09:33:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-19 18:07:24.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-04-19 09:33:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 13:52:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-01 13:58:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-19 09:33:55.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"29cb6c608209d4b656316b68881baf00e292f75c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-01 10:45:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cc5f5fab1026800138f2a4735e8174194cc5786d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 13:52:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5a8a41c1e0a142a7a7254a6d3e5233bf : TIMED_OUT in 3h 01m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5163b87fc48d4569b31411ba7b4165c8 : FAILURE in 43m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/49cfff32e3554c5ebe12477a518942fc : SUCCESS in 1h 06m 58s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/23ef3b2b56da4e9a93b044bc469f2099 : SUCCESS in 1h 18m 42s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e2753ee5acef401fb80f01585d55578d : SUCCESS in 1h 42m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8273c02219f44a05905cc5d433a43791 : FAILURE in 4m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1d34e3dc50704936a4403a2e02113b3c : FAILURE in 3m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/25f92d49e9b44da886f592cc8bfaf152 : FAILURE in 7m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a57705ad228f42b7b5926363bcfd9a6a : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80e183ebde5b47f3940218f3e56f0f14 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adff59fe882f4691865333c417b24036 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb5136c1a33419089fa3734c975ceae : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea6dc18cc70a4ef5a8a167c4b5ca82c4 : SUCCESS in 3m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc3f5d67f165433d870bdf7210adfb38 : FAILURE in 3m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/82560afb03b84e2ebdc8253794293d33 : FAILURE in 3m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4238e1b7e6948e998038dfdb0233eab : FAILURE in 7m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ec945280cd114a9fb257cc7bb61604c5 : FAILURE in 8m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/162d4469b9bf43e6bfdf1e69affe6394 : SUCCESS in 1h 04m 34s","accounts_in_message":[],"_revision_number":1},{"id":"22326b5f305b49c3cbb636963b7d8d6900141540","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-01 13:58:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4819c6346132afe5cb9527ad4783e32d4afb8670","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 14:03:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/101df5a15f3a48c0b192f0459ed2b1cf : FAILURE in 2h 49m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a09e9be6da9e45898372e22a8deedda8 : FAILURE in 6m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/add37a50f2234ec0a54d3ae81e99c741 : FAILURE in 14m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5eca23840d754a98897da541f7ac7f21 : FAILURE in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"900748d681c0b3d05e6281b2eafa59f5a37180dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 09:13:25.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a9032457f88edd54cc0edef6a606a859e3e40df9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 09:17:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"967f4e753908da9682a2589be3031d84e2ddf5a8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 09:33:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f725de1250bcfecde7038e6643e9f3b5fc277203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 12:23:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b2941c5e6eaa4691a67f9f4887f9df29 : TIMED_OUT in 3h 01m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/573126fa2e1d4dba963f9cd348eb9bc5 : SUCCESS in 29m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b0f97fbd47d74748b43bf2398816af41 : SUCCESS in 1h 09m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8064835e4375432c917c10f732cd87ac : SUCCESS in 1h 06m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d007aca2167b4130b52af5ced909071e : SUCCESS in 1h 10m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d2856798b2714b159e3389f1d0e54c6f : SUCCESS in 21m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2436c1cc89f64b7090c24063cc584041 : SUCCESS in 19m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4b36e527bbcd44deb8dbc31a80089e24 : SUCCESS in 54m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f35454d0f92a4bdea9675b3b65aae938 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bef34fd61574da797268730d6597190 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e310b46af934a3c82e764747ee5e823 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c24a963273247f2927c19e09dcb2d8e : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/534542daef1b40c7afeebd61325297a2 : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fa6f7fd82e9d4fbbac07a92a9c2ae4f8 : SUCCESS in 23m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ea97873cbd2458ab0384eb5c7ebd23c : SUCCESS in 21m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b4e09b4867448da9f8f3ac632fb019b : SUCCESS in 55m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3a6ceaf3f108449e9fe20e42382a91aa : SUCCESS in 48m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7e8db922673f444fa5c39c310bab7581 : SUCCESS in 1h 05m 38s","accounts_in_message":[],"_revision_number":3},{"id":"27c27a43c5bdb048069464a0d37fd908cac06307","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 12:24:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fe6fa99f78cb3ce38b505813d3c00eb7f535b918","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 13:53:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/baf656e63d1c42a48056f10b1c0e95d1 : SUCCESS in 33m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5072a0909b55423798aaa07cf7e0a0e8 : SUCCESS in 1h 09m 25s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9afa71e2be864b438f1b9a663cf7b029 : SUCCESS in 1h 25m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac839a8f17ef48fdb44c1a8a162ced66 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83881773c1a24dcdbd03708b3d2d4bc8 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db9e1fb95b2b4a5087e5d80a7e50520a : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdeac484c23b4ae2ab2b72d919d12144 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e115035bc650447ea949ede6bfc2ac6a : SUCCESS in 4m 17s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1117bac2c1d04097906a554ae7cd157b : SUCCESS in 23m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f97eb8174ceb419e8bd1440e853ecbfb : SUCCESS in 53m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cdf6bb2d1fcd4b40a6b0e9104214b8a8 : SUCCESS in 58m 51s","accounts_in_message":[],"_revision_number":3},{"id":"7f1e518e69d5603c26924abbb62636718a00745d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 13:53:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1a77f867afc36f772212799ed0ef3d96f2996ed0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 13:55:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82dc2fb3d81f47f98415243982298dc0 : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6a853cf93ca4ec38a40bd49e438a390 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":3},{"id":"64fc3ce8971e0a3b626d7f9f1ab0f37156926b58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 18:07:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a7beb0d1067d40f2ae4cf333f248b74a : TIMED_OUT in 3h 05m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d7a851b79294e36bcffb0518b9b2e78 : SUCCESS in 2h 10m 56s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/98b8e8ecc8e34d978b72c475af274998 : SUCCESS in 2h 27m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fc4b9a7bbe4e47a4a8f8cffd77313314 : SUCCESS in 2h 02m 31s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"238c56f53020da918f3c840eeffe9f41225892d7","revisions":{"af476e763987ab6ff88f52ac81c782bb4147376a":{"kind":"REWORK","_number":1,"created":"2022-04-01 10:45:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/88/835988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/835988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/835988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/835988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/835988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/835988/1"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 10:24:04.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-01 10:45:15.000000000","tz":0},"subject":"Fix disabling of curlrc in healthcheck_curl","message":"Fix disabling of curlrc in healthcheck_curl\n\nThe -q (aka --disable) option only works if it is the first parameter on\nthe command line.\n\nChange-Id: Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f\nCloses-Bug: #1967272\n(cherry picked from commit 196b742ba271fd7e3521448f83e8a2914f5c7dc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af476e763987ab6ff88f52ac81c782bb4147376a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af476e763987ab6ff88f52ac81c782bb4147376a"}]},"branch":"refs/heads/stable/victoria"},"3e20d4ca9e0e181c5f465b86fd8c27de7287ab33":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-19 09:13:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/88/835988/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/835988/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/835988/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/835988/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/835988/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/835988/2"}}},"commit":{"parents":[{"commit":"15d3f1b3130e4e09cc470871162656359a9f1af6","subject":"elasticsearch: install Java first on CentOS too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15d3f1b3130e4e09cc470871162656359a9f1af6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 10:24:04.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-19 09:13:25.000000000","tz":0},"subject":"Fix disabling of curlrc in healthcheck_curl","message":"Fix disabling of curlrc in healthcheck_curl\n\nThe -q (aka --disable) option only works if it is the first parameter on\nthe command line.\n\nChange-Id: Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f\nCloses-Bug: #1967272\n(cherry picked from commit 196b742ba271fd7e3521448f83e8a2914f5c7dc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e20d4ca9e0e181c5f465b86fd8c27de7287ab33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e20d4ca9e0e181c5f465b86fd8c27de7287ab33"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"238c56f53020da918f3c840eeffe9f41225892d7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-19 09:17:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/88/835988/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/835988/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/835988/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/835988/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/835988/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/835988/3"}}},"commit":{"parents":[{"commit":"a7ff9f6c0d2d5af14a0c9bebd0ceb932bbfeffb5","subject":"elasticsearch: install Java first on CentOS too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ff9f6c0d2d5af14a0c9bebd0ceb932bbfeffb5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 10:24:04.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-19 09:17:25.000000000","tz":0},"subject":"Fix disabling of curlrc in healthcheck_curl","message":"Fix disabling of curlrc in healthcheck_curl\n\nThe -q (aka --disable) option only works if it is the first parameter on\nthe command line.\n\nChange-Id: Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f\nCloses-Bug: #1967272\n(cherry picked from commit 196b742ba271fd7e3521448f83e8a2914f5c7dc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/238c56f53020da918f3c840eeffe9f41225892d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/238c56f53020da918f3c840eeffe9f41225892d7"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
