)]}'
{"id":"openstack%2Fkolla~835989","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f","project":"openstack/kolla","branch":"stable/ussuri","topic":"healthcheck-disable-curlrc-stable/ussuri","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-19 11:58:44.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-19 11:58:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f","subject":"Fix disabling of curlrc in healthcheck_curl","status":"MERGED","created":"2022-04-01 10:46:43.000000000","updated":"2022-04-19 19:18:48.000000000","submitted":"2022-04-19 11:58:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835989-healthcheck-disable-curlrc-stable/ussuri","cherry_pick_of_change":835957,"cherry_pick_of_patch_set":1,"meta_rev_id":"9fd88f1f5e590d6919fbf84a9db9ce6809d2169c","_number":835989,"virtual_id_number":835989,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-19 11:58:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-04-19 09:33:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-19 19:18:48.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-04-19 09:33:46.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 11:59:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-01 13:58:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-19 09:33:46.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"a2ab45ec5d52017b78b24c6ca77d343b5a937047","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-01 10:46:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"07cbaf057c4c8478ef811919139476bacfea220d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:59:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3e7af52a61ca4176b41ee128ad6446f3 : SUCCESS in 1h 10m 17s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7210c345fd1645b094b3fbb2ecd18e89 : FAILURE in 47m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1cdaa8463dd7428f94ed8ab587e608b4 : FAILURE in 4m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/31b3bcd7a5754caca774966d2e89be90 : FAILURE in 3m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/de997513d0b14394a6c9f168c3676271 : FAILURE in 7m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/866a33696b26458b934fa2c3b9837df4 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbbbfbfdda414db9803801b5389fe842 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8728b1638714c3aa5351ac6c633a754 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf10c7e3ac8d433e852d9eb130b217fe : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81c7ab7fe27048daac1552a0d88ed69f : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e2eab4562894284a45b958b63a18932 : SUCCESS in 21m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/315623e0ad664b139ed4e9ce5bf8c031 : SUCCESS in 32m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cafe09a23bb34afab989d9e76791b252 : SUCCESS in 50m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/215522143a8f48bdb690becac5542d25 : SUCCESS in 51m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6e33cade9eec44279faf00952c604bb8 : SUCCESS in 1h 01m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d83c5d58a6563d35f51ca99e7b8f907cbe8b65f8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-01 13:58:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cf237a4de6054bc16db2cbf8957a7c662ad84fde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 15:09:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b659a53fda834bcfa3bcd3aea7e0f1eb : TIMED_OUT in 3h 06m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/de93b16b39de4f6bb06c15161ee733df : FAILURE in 5m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ab3073d481e4fb2a9151be788c404ab : FAILURE in 16m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/24cdc3e8a7d34e1aa2c9b25ee4fabb40 : SUCCESS in 1h 44m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60f4b4b614c0693b66d6d17c10411ccc0a68ba65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 09:17:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"eeafaf06b5a2c33dae73c64a1f1e2c5f0322c3b4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 09:33:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6383a4279f74cb2b66301ad17222230352374a5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 10:21:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fef7768e93d54a909a5aabf7755c10d9 : SUCCESS in 53m 31s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6745ba2f2c9a4988a4201743e253fb53 : SUCCESS in 29m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4200866b45354abbb6987d774082e783 : SUCCESS in 25m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8e23120749654296ac733af1bc59cfbc : SUCCESS in 19m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4691f6b0b35e444e95fd7f87104ad93d : SUCCESS in 47m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fb9668e920e416f9f022af9f3f77089 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eacb55f40f9e4d63af8ca91a7c5c7ef7 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3280022942ba4a5b9f0c23f1f7bf5750 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0958272808e4b60824b5f30b734290d : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4974b240099041b0bef62927233e2d70 : SUCCESS in 3m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ab7e77bd1148497fa5a683a5ef918134 : SUCCESS in 21m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/90e76980968943d7953aea89043f5d72 : SUCCESS in 24m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/70ec50e08b1c4895a9ebc8ab88704c32 : SUCCESS in 43m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/45ab461d3b43468eb1f05e66bee9b47c : SUCCESS in 41m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6c8a54c9f4704cfeae95b5cb6c02d736 : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":2},{"id":"c9707534b10f9987c52048bb40221621fd7b6d2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 10:35:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e4c716125d296eca05c482405b33c0f516f9db8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 11:58:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/273659326a5b4f27859c8eefb6bfb1e7 : SUCCESS in 1h 22m 26s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f1ec773e777940d5855c6d6f2899116a : SUCCESS in 58m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/798d9c80ac944cfca09ba4124a7bc041 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebca6f60ada64e1a99749ce15247a70d : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9775859a8ba24d83a586bbe1f90961e2 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7b1969086e4e97932f7d2fd6aa596d : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4691f96abbb4ddc833649ec52b75746 : SUCCESS in 5m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/37fd1abed4b0470fad2895456ffb7835 : SUCCESS in 29m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2692dd3326444b48acf49e5918f353b : SUCCESS in 53m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5ba49810a7374ca0a170e5caa30d303d : SUCCESS in 1h 15m 10s","accounts_in_message":[],"_revision_number":2},{"id":"77a018dc6341be51d340880195f8b93c893213df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 11:58:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e2465f8969d856863cbd2b6a4ac824eeb9090688","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 12:01:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ddcf331a8e44e0abc416b5b4c90d2f : SUCCESS in 1m 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c82baf3069a14f568128934a3d4e2dd9 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2},{"id":"10c4739b09040d9cc2f772b3921450cfdb9de679","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 19:18:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3dcebda4425f4ed1934ebe7546838980 : TIMED_OUT in 3h 06m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e346537b79043a6befbf9f8354ce9b4 : FAILURE in 2h 12m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/82f18438094a43c3acd11bac17b69801 : FAILURE in 1h 17m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8394e40896c146989d399ac872e1178e : FAILURE in 2h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"42b45677e2992c4d37c64fa57da96d410a3ae43e","revisions":{"1723c1318098a85d6836ca11151720d34ae86013":{"kind":"REWORK","_number":1,"created":"2022-04-01 10:46:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/835989/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/835989/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/835989/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/835989/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/835989/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/835989/1"}}},"commit":{"parents":[{"commit":"74f707eaed4a5d4ff4203b0efac5afafcc802bed","subject":"libvirt: add Cyrus SASL packages for DIGEST-MD5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74f707eaed4a5d4ff4203b0efac5afafcc802bed"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 10:24:04.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-04-01 10:46:43.000000000","tz":0},"subject":"Fix disabling of curlrc in healthcheck_curl","message":"Fix disabling of curlrc in healthcheck_curl\n\nThe -q (aka --disable) option only works if it is the first parameter on\nthe command line.\n\nChange-Id: Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f\nCloses-Bug: #1967272\n(cherry picked from commit 196b742ba271fd7e3521448f83e8a2914f5c7dc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1723c1318098a85d6836ca11151720d34ae86013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1723c1318098a85d6836ca11151720d34ae86013"}]},"branch":"refs/heads/stable/ussuri"},"42b45677e2992c4d37c64fa57da96d410a3ae43e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-19 09:17:48.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/89/835989/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/835989/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/835989/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/835989/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/835989/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/835989/2"}}},"commit":{"parents":[{"commit":"a12d88c253b7b46e58a9b2c239b90b6b550c8f01","subject":"elasticsearch: install Java first on CentOS too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a12d88c253b7b46e58a9b2c239b90b6b550c8f01"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-03-31 10:24:04.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-19 09:17:48.000000000","tz":0},"subject":"Fix disabling of curlrc in healthcheck_curl","message":"Fix disabling of curlrc in healthcheck_curl\n\nThe -q (aka --disable) option only works if it is the first parameter on\nthe command line.\n\nChange-Id: Ia9ab0bdc95e658c17e5be5abefcf96f1c05ee84f\nCloses-Bug: #1967272\n(cherry picked from commit 196b742ba271fd7e3521448f83e8a2914f5c7dc6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42b45677e2992c4d37c64fa57da96d410a3ae43e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42b45677e2992c4d37c64fa57da96d410a3ae43e"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
