)]}'
{"id":"openstack%2Fkolla~836138","triplet_id":"openstack%2Fkolla~master~Ie28e8580406f3f805500ed4020895e5e9ebf7c31","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-04 09:58:22.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-03 09:55:47.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"Ie28e8580406f3f805500ed4020895e5e9ebf7c31","subject":"cloudkitty-api: make sure that we install packages","status":"MERGED","created":"2022-04-01 10:57:42.000000000","updated":"2022-04-04 10:00:07.000000000","submitted":"2022-04-04 09:58:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836138","meta_rev_id":"898887ebb075a6b1a2da6d153289d97a0c6990e9","_number":836138,"virtual_id_number":836138,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-04 09:58:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-04 08:37:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-03 09:55:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-04 08:37:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 12:27:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-01 15:27:55.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-04 08:37:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"106b19a5b6b751c40d429f588f9d3757ff500417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-01 10:57:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ada77229841a1f71b7baff4452af19790f90879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 12:27:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/23d52022774643fb9d50f54055f756de : SUCCESS in 28m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4305771b55cf40488395bfc2d6c88963 : SUCCESS in 44m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e3128cef8f254f7189a4e747b95bbde4 : SUCCESS in 1h 11m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/53e86ba761eb4aef81334df2d818c0e5 : SUCCESS in 1h 20m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4378b6e6b74644d1a4c598903e215f27 : SUCCESS in 1h 21m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1c15c7acdade428e8a060a01b968cf7f : SUCCESS in 26m 45s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fff76928260b4e648cbf7d42d5bfdec7 : SUCCESS in 16m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a6007a5ef99344d6978386808642b4a0 : SUCCESS in 1h 00m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5d4f075aecd94d8d9230054bb1f40a8f : SUCCESS in 1h 23m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc4607279d840d88c912fb67ebe99b0 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dc5eb816f494953bf4a54695053f26c : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f804e0f37b754bf1b6eeb9fb2c846b8f : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42871e5cc9d646c290cb3647d49eb36a : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6addc382b6d4478b816e22c15c5200b7 : SUCCESS in 19m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/109b4a1845544daba7135e93f9742898 : SUCCESS in 30m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8fee8350bb334eb8b72c4f86d44093e2 : SUCCESS in 50m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e39dbf2d7fec43478cedee803c4cc047 : FAILURE in 42m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fb0dd9cbcba94800b2ed2a3b1228af5a : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e081e10f7562bdcee3b06bf03988c98d756dc76b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-01 15:27:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d804b8bf102a1633006a6d989dd34eea321c180a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 16:10:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8946e55826a44e692b927eb7322d2fb : SUCCESS in 2h 35m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8dcbe8a1c0524a88812ac87552d28b6b : SUCCESS in 2h 22m 25s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f544a31a966498cbcd32135b490c09f : SUCCESS in 2h 19m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03d47a6372e2423097ff530feaa47213 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31f46100cd364acfa42b6329db0f7626 : SUCCESS in 6m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e85d31dcb8704b5aab74ff52b044c42c : SUCCESS in 1h 59m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19f62bd778e42bff3e871f328b2c15ce3700ad2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-02 13:46:51.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fe2dbe842f6cbc482e55a8590d655c22491e051e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-02 13:47:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a84bd75024d2f38d228ea7f9128ca1d6dc83e0ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-02 15:17:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1c8f9264ec484b6f81424c8185d71d37 : SUCCESS in 31m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e96482dfe7b744548dbfab992f7005de : SUCCESS in 44m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fc890bb799294448bd4ae92e29756242 : SUCCESS in 1h 14m 05s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e3f4dcb201294023a82c943f9821b0d1 : SUCCESS in 1h 10m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7adeb6021caf4137ba48003c64b5e428 : SUCCESS in 1h 27m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/23126c9db22e40da98df39f3a2802a64 : SUCCESS in 32m 01s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/eed6dd08022b4626a19b2bed199faefb : SUCCESS in 20m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a7c02ea38fae438bb43819702c203906 : SUCCESS in 1h 01m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2cd0066edfc745d0b18d4721e8bd1ed4 : SUCCESS in 1h 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0efbd8585f9e46afbc92f3464063ebbb : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/201bbe8d71bd4dd79107aa0f214909e0 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d20d7eeb38c46d4b6267e5b314ed7aa : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bbfc5e4e44b4621b7e9d2c442787514 : SUCCESS in 6m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6bbaaaf1eeb34779a1fec9481381d291 : SUCCESS in 22m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/44f201602a9e4496bcbc141e743163ab : SUCCESS in 37m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c827a78397941ccba3f79aa7a3b761d : SUCCESS in 56m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae40ad18851f458a8585887dec3945ec : FAILURE in 46m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a96b47176c2b485ea7405c5146cb0841 : SUCCESS in 1h 04m 07s","accounts_in_message":[],"_revision_number":2},{"id":"bde94ef1c11e1717bb8cef053065eabb451e0b63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-02 15:27:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/88a32848ca744e9eab8bc82923a4b590 : SUCCESS in 1h 27m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e2006b8c24540ef9cc2d6c9790e9487 : SUCCESS in 1h 02m 52s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ec948370ddc406b814db2228a4645dd : SUCCESS in 1h 38m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cfa3b5ddba2404fa68c99b4cf420404 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1f9de955fc74405b0a0d7d0d9c3b959 : SUCCESS in 6m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/824994f7ceab41f1935df5b7318b8b41 : SUCCESS in 1h 03m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"330c337b27cd26cd78384715882b0f2155f12a4f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-03 09:55:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8b59ec6b488d1f2414e3f598abd40e5405b52bae","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-04 08:37:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d3669c379034b2e985f029f03b30b4169e2fdbed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 08:38:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0e88d1a2f93c2b8699d620b0c4f74d17afe05887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 09:58:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/29a1edf7c4fa4a74948deba0dab40432 : SUCCESS in 33m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/990d2aaee267432ca98205f3336d80da : SUCCESS in 1h 03m 37s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9bd7fef0b8ce4f96a9945258c1367b74 : SUCCESS in 1h 08m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2e63f86bfbba428284eab1e951da02f5 : SUCCESS in 27m 34s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4afcd922d0524cb292d7d7c752742816 : SUCCESS in 50m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/20b701bfdb0e4148a983a73127094c4e : SUCCESS in 59m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7409609a80be480b931ddc9fad6ddc95 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/472c77ef6efe47bdb94235329d983f0f : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4aee002031d847ea8f57cbde1754cc0a : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce305abaeffa4f5b9faaad9b87ced9f1 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ed61de76eb3144398e6fc6dd71b954c0 : SUCCESS in 32m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9970b7e7b49c45e4a39a64968d13d698 : SUCCESS in 43m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ad50c4c307e647c9a5493caed954e8dd : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":2},{"id":"85d0a8551a3088ea62cd3d0a00e1c1893d5856c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 09:58:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"898887ebb075a6b1a2da6d153289d97a0c6990e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 10:00:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1dc7c3d9884535b5521c6a9e2cbbdc : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a8fcb02f40c94c2101edff9ec4a3101d1d14fe97","revisions":{"379860c381cd2618bfc0226d92c17d83f49e74d4":{"kind":"REWORK","_number":1,"created":"2022-04-01 10:57:42.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/836138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/836138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/836138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/836138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/836138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/836138/1"}}},"commit":{"parents":[{"commit":"089899d8804fbc9f5d2509fe6c50a7a25f45231f","subject":"Merge \"Fix disabling of curlrc in healthcheck_curl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/089899d8804fbc9f5d2509fe6c50a7a25f45231f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-04-01 10:56:56.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-04-01 10:56:56.000000000","tz":120},"subject":"cloudkitty-api: make sure that we install packages","message":"cloudkitty-api: make sure that we install packages\n\nDebuntu images did not got packages installed.\n\nChange-Id: Ie28e8580406f3f805500ed4020895e5e9ebf7c31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/379860c381cd2618bfc0226d92c17d83f49e74d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/379860c381cd2618bfc0226d92c17d83f49e74d4"}]},"branch":"refs/heads/master"},"a8fcb02f40c94c2101edff9ec4a3101d1d14fe97":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-02 13:46:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/38/836138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/836138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/836138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/836138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/836138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/836138/2"}}},"commit":{"parents":[{"commit":"089899d8804fbc9f5d2509fe6c50a7a25f45231f","subject":"Merge \"Fix disabling of curlrc in healthcheck_curl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/089899d8804fbc9f5d2509fe6c50a7a25f45231f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-04-01 10:56:56.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-02 13:46:51.000000000","tz":0},"subject":"cloudkitty-api: make sure that we install packages","message":"cloudkitty-api: make sure that we install packages\n\nDebuntu binary images did not got packages installed.\n\nChange-Id: Ie28e8580406f3f805500ed4020895e5e9ebf7c31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8fcb02f40c94c2101edff9ec4a3101d1d14fe97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8fcb02f40c94c2101edff9ec4a3101d1d14fe97"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
