)]}'
{"id":"openstack%2Fkolla~836212","triplet_id":"openstack%2Fkolla~master~I5c2deb4a2e33575d57c852089f856a9acc6818d0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-12 08:30:27.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-12 10:09:59.000000000","reason":"Change was submitted"},"29771":{"account":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"last_update":"2022-04-12 10:09:59.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-12 10:09:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c2deb4a2e33575d57c852089f856a9acc6818d0","subject":"Adjust permissions on _extend_start files.","status":"MERGED","created":"2022-04-01 20:52:51.000000000","updated":"2022-04-12 10:11:55.000000000","submitted":"2022-04-12 10:09:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"836212","meta_rev_id":"fd2d5d4fdfa31aad2f12af5abd63d5828cf22aed","_number":836212,"virtual_id_number":836212,"owner":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-12 10:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-12 08:30:27.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-04-11 15:22:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-04-12 08:30:27.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 22:34:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-03 22:37:01.000000000","updated_by":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-04 09:10:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-05 15:17:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"d1cb82f46e074a1328330ccfe463bea065d0a3f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"date":"2022-04-01 20:52:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce60847dcd1517d34de6d84d872afe3c9df115c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 22:34:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/5915a0e59d8641a69fb37fdd05f2aeca : SUCCESS in 31m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/366bdd95adeb4bbaa258d5df3654b654 : SUCCESS in 44m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/34fca6dbcdc44c8189455a8c4ccc8e6e : SUCCESS in 1h 03m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9d32e81d833a42a5b62d10acd44a70a1 : SUCCESS in 59m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3cee852598ff42eba6cd15b91914d6bc : SUCCESS in 1h 36m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fc8174747b68472897b972231fc31781 : SUCCESS in 31m 04s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4010bb1de74c4bae981867e1d9cf1819 : SUCCESS in 16m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3a8914187b7e4cb2a8d567e2e0f07d83 : SUCCESS in 49m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fff293dec4c64a70a5293a6c3de4a417 : SUCCESS in 56m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c297da9d4ea1435aac4175b22a975598 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24758bb149d949e19e71e69d44a64f64 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/247696187ec4423f94ed3238c60f0b22 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62718e341a334cb28d9df6a5ba9b4297 : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b52ec6c468004dcfac59e19fd4964334 : SUCCESS in 27m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f66f43d011d54e16bf0b7e07104ae12b : SUCCESS in 38m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4e3d2762550400a80dd7914a4e07934 : SUCCESS in 56m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f03454705acd4f528fa67159f8078fb6 : FAILURE in 54m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/764ba50c7b4d44b197ff756d332c26a2 : SUCCESS in 54m 37s","accounts_in_message":[],"_revision_number":1},{"id":"838b105378106d0fba8d6381afc546859fb51618","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 22:45:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/89bcc0f365264823b90a88679889215d : SUCCESS in 1h 27m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/aaf74750903a4c9f8aae6f888aa8b895 : SUCCESS in 1h 03m 15s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6816d6a0fab840479a8c912df1b691fb : SUCCESS in 1h 50m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1a4c93da5a448f6a901d405a256b357 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e0a820758ae4ce09455ad9ffa071812 : SUCCESS in 6m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a7d59129a17b4d95b2b6a216160f8102 : SUCCESS in 1h 00m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c289d86d2dc01125ad4e23f0245577b996d0d47","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-04 09:10:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6e54c379106bdecf42100bd0100647bef23b83e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-04 19:20:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97f13447c909e27e982cf8f1763103af6afd5f5e","author":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"date":"2022-04-05 03:02:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0e77a15df50ddd32261bae43f1aead6de6dc8c1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-05 08:30:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea25d0ce132c85b955275077440a75ea2fc0d5ad","author":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"date":"2022-04-05 13:49:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a0a301c1e831bc810ede2a98162eab1b419ddb3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-05 14:15:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba85727169303187923957947a4f1bdc11cab6ff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-05 15:17:17.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9c59acb60ac4478e9ba33159391a69f43f5c4f4e","author":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"date":"2022-04-05 16:31:18.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"0af7f28a098abad19a1f2e04e91bcda44ae0e523","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"date":"2022-04-08 20:48:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"545a75a5b0e079648d95c6649f15e21174f5953a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"date":"2022-04-08 20:49:55.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bf2cd62d5666e6ea7b391f05d8478a1e663b9a6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 22:32:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ff084fb9eae24f4fa8495e646981d281 : SUCCESS in 52m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d9614ce9e45e49f0a57c47538cfcbb0b : SUCCESS in 1h 08m 10s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b62a1bda2d3d47ba8eb49da50b6d5aef : SUCCESS in 1h 28m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/aed74201f803412187a96a28f997cdbf : SUCCESS in 30m 40s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/da6473eea2f84a969a7c6170a5155574 : SUCCESS in 1h 00m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e876ebdb38e843ab879aa3785fbea8e7 : SUCCESS in 1h 37m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ffede0982d4c049624991377148aa2 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a58c54b6e9b84a188bf4f673bb19a1d8 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2201529f07db4825995a1f719444d61b : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64941fef55244ac9811f398935a6e87b : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a720a899d8d74bd2a171a82577665e5f : SUCCESS in 6m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d40a6884729b43f9aebddcedffe8fda5 : SUCCESS in 31m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2067d1d8119349de9346a5a51a1212e9 : SUCCESS in 49m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/968dade7257f4eef8354a95a6187fb53 : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":3},{"id":"d82fffa7d0be65470dffaa93a0e3057f81391d2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 22:49:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fc1953c78670460094785a7b9190e33c : SUCCESS in 1h 55m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7a6970c58ca744fcb6ad13961cc046fc : SUCCESS in 1h 26m 56s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5261119fd48148a197947c556cbe7382 : SUCCESS in 1h 57m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c0e2388231c45e78b1f55c590e51006 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42c60433bb6348aebaff3b9c7e995008 : SUCCESS in 6m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e3a40c68902f47f19e1a7785dad4ecf8 : SUCCESS in 1h 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0952e7a12efeba92ef798528ab4a768b81b31c4d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-11 08:57:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"64be020fdf927caf91ed8d5be3447f01985b15c5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 10:33:38.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"86be44241532523a1828fab459a968a7d2bd2ca8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 10:34:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"19484cbfb5be129d118854aa9d5d3728ce039f56","author":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"date":"2022-04-11 11:28:44.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"908e60b917294d05381fbf9a6d0b92f08d49880e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:32:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/a061b4a59b084f168c89c86a2342067f : SUCCESS in 54m 15s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/33b0563419e843a8bb905565232285ac : SUCCESS in 42m 29s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/44d8dd9567544435a3b7410f9fb669c7 : SUCCESS in 1h 35m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/dd46f7576efa4b32b703cefcb431d40b : SUCCESS in 1h 41m 25s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/962fab86ed7942c8b44c4557a5833c79 : SUCCESS in 1h 39m 09s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/852020a72572432eb69c4f940289ee2b : SUCCESS in 1h 47m 18s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/6f4d1f1d8baf4050a750eb79d175a3cb : SUCCESS in 1h 22m 30s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/a0fbf7bf0f914329ad102165e852cafa : SUCCESS in 1h 12m 36s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/9866f7a742dd45b09aaf9d5ae55afaf3 : SUCCESS in 29m 08s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/b476ad7285be46bcb977e2020289fd39 : FAILURE in 55m 01s (non-voting)\n- kolla-ansible-centos8s-source-octavia https://zuul.opendev.org/t/openstack/build/e9926a5bec3c454996d051e275791465 : SUCCESS in 1h 05m 20s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/ad92863cdc4a426b87201028051913e5 : SUCCESS in 1h 24m 04s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/436f31fd68d14b29bb6773b35e6285eb : SUCCESS in 38m 34s (non-voting)\n- kolla-ansible-centos8s-source-kvm https://zuul.opendev.org/t/openstack/build/3ed00cb76df742ac93d177279a04baf2 : SUCCESS in 54m 26s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/64bd0bcd703a4ab3b0fa2cf2b46c0c0d : SUCCESS in 29m 18s\n- kolla-ansible-debian-source-ironic https://zuul.opendev.org/t/openstack/build/17d432ec7db84930ae69e59aea1f0b9e : SUCCESS in 1h 09m 40s (non-voting)\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/3375fc49cf4f497c8dcea604cafa9991 : SUCCESS in 23m 12s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/f31b71a23b5f45859650e7203b5a951d : SUCCESS in 1h 09m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4091f1ee4f8146e99abfbd40eb5405a5 : FAILURE in 22m 40s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0b06f59569284a20a1b979d4ca4332c9 : SUCCESS in 1h 09m 32s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/2109ed47da0d4a14994835148f60e3ff : SUCCESS in 1h 00m 42s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/a7da823f9f3945c2b1521c433549e220 : SUCCESS in 1h 13m 05s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/697f0fc5d1d34a47868636be7cd5729d : SUCCESS in 27m 29s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/600f5629bd4c44a0b07055a3467efce4 : FAILURE in 42m 34s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/73aaa79ed92947aea3db8600c47153ca : SUCCESS in 51m 16s (non-voting)\n- kolla-ansible-ubuntu-source-octavia https://zuul.opendev.org/t/openstack/build/41109036f6e3407797408606e504e134 : SUCCESS in 52m 45s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/7c7a6cdc28e94795ac6d05d2f3dfe8cf : FAILURE in 48m 16s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/da390faed19745fd9c404c139c2ca5e1 : SUCCESS in 24m 22s (non-voting)\n- kolla-ansible-ubuntu-source-kvm https://zuul.opendev.org/t/openstack/build/096b7af4de864a2c8c85c410a4ac228a : SUCCESS in 45m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ac0794f297cd02c192155e78e2eafd967787e8f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 12:42:24.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a9da748dd67742ceb164e1e13dfee4468e32d7aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:51:12.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 836212,3","accounts_in_message":[],"_revision_number":3},{"id":"3eadf8bf75506b19bad5a2730824775c026f5633","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 15:22:32.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"28b60ea5b2a63e72f85851d0cf71bc3634d53f40","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 15:22:49.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ae05c6ea00588b3620bf8ee85680acb33e4878a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 16:48:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ca04dbd290264a9a9adf31790ba0178e : SUCCESS in 47m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/822488aad73e4e538864feb6a1cdcc8d : SUCCESS in 1h 08m 38s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/98cd91a2a9a4404588fa1ab565007a2a : SUCCESS in 1h 19m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5a724cd614944a07a7e7efe85715906d : SUCCESS in 30m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eed01ae3b0154c3cb3483135c5ee3bec : SUCCESS in 1h 05m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0cbbcf73ecca4c8c82839a840e2fac29 : SUCCESS in 1h 01m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77a318d2829644f2846150cf628907bb : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab429cb8f8f943fc8c0db10c6564580d : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03127bcbd58142a5ae8793a04820b6bf : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/411557d64cd745ae9d75b87ffc3b4a28 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2091c19ca124722a0063ba93e6216cd : SUCCESS in 6m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c287a684f5ab4b9d880d7b2cf6f1eec6 : SUCCESS in 27m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e9909b172bf94ad99c9314f1ca2ed171 : SUCCESS in 50m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e892a2bb0723489785d79847146bb8ad : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":4},{"id":"e77cf0652b722c898c9ae9cef9d74b9a13101698","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 23:58:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c1db55a71155469bb21548615cbaf81f : SUCCESS in 2h 36m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c8a541d921a24936a2825a60d09b7e50 : SUCCESS in 2h 02m 30s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad0c46b808e34ff28eac69071ce6cf2d : SUCCESS in 2h 10m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3905e55c4cc4d90af78b99880cc7e7f : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68feae50fc1245069ce2ee4a6bbeae43 : SUCCESS in 7m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/67dbe886d94d4aa49f300d3db552df6e : SUCCESS in 1h 56m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f9c2ccb6fa9e97185a419b8bbc682b8e26e99b8b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-12 08:30:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"53f3cc3de5d045b6e1d9a7b01023c415988329a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 08:31:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fdb8dc368ef6c4d627a36be474ecc6db24fa8e33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:09:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b8f8662bf93b4e3fa7698a6375cbbe40 : SUCCESS in 30m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/320d7cbd1ac5404baa726439a728c921 : SUCCESS in 1h 15m 12s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e4bc9a5a54b04282be20e850a5afa6ef : SUCCESS in 1h 17m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/62f81f4c60b541aba72119c1461add80 : SUCCESS in 34m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/320210516f644e11a512bf49fb1c7915 : SUCCESS in 1h 02m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9f13595c470f43a4a4a7b15fcbd50d27 : SUCCESS in 1h 09m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e70432a49dd41c79fefb6f2ede7790e : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9f149ea6a3443ef8eccf0eccc2c2321 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c51bb6c4926c4049a26e9fc30960ff9c : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a6622549f4b46399e4ce2465628e30d : SUCCESS in 5m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/135e6ddce3f2422b8fd36243c7967f1b : SUCCESS in 34m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5cbec26b7e84476a8cd594400f587585 : SUCCESS in 43m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee87870136ef422aa839f7afd87558ad : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":4},{"id":"10158ef8374eec66cbed34f33003d8965e6240f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:09:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fd2d5d4fdfa31aad2f12af5abd63d5828cf22aed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:11:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fcac93561854b0cb2be38df28ab9952 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1d96a2bbe1b536b7d7f4cdbf55c6dabae6d058ae","revisions":{"7034cc73d6f4bee6d68cdba7ae98c5606fceb515":{"kind":"REWORK","_number":1,"created":"2022-04-01 20:52:51.000000000","uploader":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"ref":"refs/changes/12/836212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/836212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/836212/1"}}},"commit":{"parents":[{"commit":"089899d8804fbc9f5d2509fe6c50a7a25f45231f","subject":"Merge \"Fix disabling of curlrc in healthcheck_curl\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/089899d8804fbc9f5d2509fe6c50a7a25f45231f"}]}],"author":{"name":"Tim Shearer","email":"tshearer@advaoptical.com","date":"2022-04-01 20:22:28.000000000","tz":-240},"committer":{"name":"Tim Shearer","email":"tshearer@advaoptical.com","date":"2022-04-01 20:49:54.000000000","tz":-240},"subject":"Set correct file permissions on kolla_extend_start binary in kolla-toolbox image.","message":"Set correct file permissions on kolla_extend_start binary in kolla-toolbox image.\n\nChange-Id: I5c2deb4a2e33575d57c852089f856a9acc6818d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7034cc73d6f4bee6d68cdba7ae98c5606fceb515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7034cc73d6f4bee6d68cdba7ae98c5606fceb515"}]},"branch":"refs/heads/master"},"ad72f485e8ed254a4053be05a4307169869d4806":{"kind":"REWORK","_number":2,"created":"2022-04-08 20:48:02.000000000","uploader":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"ref":"refs/changes/12/836212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/836212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/836212/2"}}},"commit":{"parents":[{"commit":"46d7f1dc10bf67f6dde3beaca59d1ef8b07acef9","subject":"Merge \"add releasenote about dropping binary images support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46d7f1dc10bf67f6dde3beaca59d1ef8b07acef9"}]}],"author":{"name":"Tim Shearer","email":"tshearer@advaoptical.com","date":"2022-04-01 20:22:28.000000000","tz":-240},"committer":{"name":"Tim Shearer","email":"tshearer@advaoptical.com","date":"2022-04-08 20:09:20.000000000","tz":-240},"subject":"Adjust file permissions on _extend_start files.","message":"Adjust file permissions on _extend_start files.\n\nExplicitly set the permissions on the kolla-toolbox kolla_extend_start\nfile. Also, since all extend_start files are sourced rather than\nexecuted, the executable bits are now cleared throughout the project.\n\nChange-Id: I5c2deb4a2e33575d57c852089f856a9acc6818d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad72f485e8ed254a4053be05a4307169869d4806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad72f485e8ed254a4053be05a4307169869d4806"}]},"branch":"refs/heads/master"},"6ef9cbeb2e261290590b8adf883b98353789e241":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-04-08 20:49:55.000000000","uploader":{"_account_id":29771,"name":"Tim Shearer","email":"tshearer@advaoptical.com","username":"timmins"},"ref":"refs/changes/12/836212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/836212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/836212/3"}}},"commit":{"parents":[{"commit":"46d7f1dc10bf67f6dde3beaca59d1ef8b07acef9","subject":"Merge \"add releasenote about dropping binary images support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46d7f1dc10bf67f6dde3beaca59d1ef8b07acef9"}]}],"author":{"name":"Tim Shearer","email":"tshearer@advaoptical.com","date":"2022-04-01 20:22:28.000000000","tz":-240},"committer":{"name":"Tim Shearer","email":"tshearer@advaoptical.com","date":"2022-04-08 20:49:37.000000000","tz":-240},"subject":"Adjust permissions on _extend_start files.","message":"Adjust permissions on _extend_start files.\n\nExplicitly set the permissions on the kolla-toolbox kolla_extend_start\nfile. Also, since all extend_start files are sourced rather than\nexecuted, the executable bits are now cleared throughout the project.\n\nChange-Id: I5c2deb4a2e33575d57c852089f856a9acc6818d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ef9cbeb2e261290590b8adf883b98353789e241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ef9cbeb2e261290590b8adf883b98353789e241"}]},"branch":"refs/heads/master"},"1d96a2bbe1b536b7d7f4cdbf55c6dabae6d058ae":{"kind":"REWORK","_number":4,"created":"2022-04-11 15:22:32.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/12/836212/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/836212/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/836212/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/836212/4"}}},"commit":{"parents":[{"commit":"19765c629734fe699871301be23fb308c14b2639","subject":"Merge \"for Zed cycle minimal Python version is 3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19765c629734fe699871301be23fb308c14b2639"}]}],"author":{"name":"Tim Shearer","email":"tshearer@advaoptical.com","date":"2022-04-01 20:22:28.000000000","tz":-240},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-11 15:22:24.000000000","tz":120},"subject":"Adjust permissions on _extend_start files.","message":"Adjust permissions on _extend_start files.\n\nExplicitly set the permissions on the kolla-toolbox kolla_extend_start\nfile. Also, since all extend_start files are sourced rather than\nexecuted, the executable bits are now cleared throughout the project.\n\nChange-Id: I5c2deb4a2e33575d57c852089f856a9acc6818d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d96a2bbe1b536b7d7f4cdbf55c6dabae6d058ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d96a2bbe1b536b7d7f4cdbf55c6dabae6d058ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
