)]}'
{"id":"openstack%2Fkolla~836317","triplet_id":"openstack%2Fkolla~master~I72d7920acd8d15941c8c57a4186186212b273a38","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-05 09:35:20.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-04 19:16:51.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I72d7920acd8d15941c8c57a4186186212b273a38","subject":"elasticsearch: install Java first on CentOS too","status":"MERGED","created":"2022-04-04 11:37:37.000000000","updated":"2022-04-05 09:38:10.000000000","submitted":"2022-04-05 09:35:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836317","meta_rev_id":"d4f0b605c14d16c0d0b8e774edc0282e9ad3a40c","_number":836317,"virtual_id_number":836317,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 09:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-05 08:06:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-04-04 19:16:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-04-05 08:06:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-04-04 11:38:29.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-04 13:00:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 17:07:48.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-04-04 19:16:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-05 08:06:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"683005614b7002c7dfbebb9ae774c3773dd34ec5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-04 11:37:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a0ca409980142ca4a8f17ac8e64129791b98403","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-04 11:38:29.000000000","message":"Patch Set 1: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"837a1513a19a8c548b141db49534c3be81904ec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:00:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e9dc5518fd0348828e6f4d027161af2a : SUCCESS in 24m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2341480489664653bd2368290cc76a6d : SUCCESS in 43m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d881533ac09a4184ae4e35d0c8e2b998 : SUCCESS in 1h 12m 25s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a1e31d1ee3c441b5a736f62dab36dec2 : SUCCESS in 59m 00s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d941113c8b1e44c8959e542202e7774c : SUCCESS in 1h 17m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9d9c4c7f16d54e4c857cbedf5282966f : SUCCESS in 28m 31s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a755a8125c69445182f450c91bbd620b : SUCCESS in 20m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/29c873ff08f14a7baf368537c91f9230 : SUCCESS in 1h 14m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/57e7a6b9c9794e8c96db3cde55e6cda3 : SUCCESS in 1h 02m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07a8460b6f33499ea070bd85bc17f58e : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bb042a65ac649a6aa4cca2b7ca5d868 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec55ca60ee4748f09557eadac86c4fd3 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/896ac87a1b554b8abe095a706f3bdb33 : SUCCESS in 6m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e40a0fac47242fdab600a62c7a727dc : SUCCESS in 15m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9563c23609b04ac1972bfa9f3848aeb0 : SUCCESS in 33m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/34b6c82279c445fe9a5e9c7d2da0c97c : SUCCESS in 50m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a5d1ba59ba154fd68c6e3732d770435f : FAILURE in 42m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7db0bb4bbb1d4d6b8cfd53353abe6b00 : SUCCESS in 1h 03m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a3ff94fdb5a214a5e41a2329cfd3233af60e2bb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:48:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c5cfdd256334838bf12fc6062ba3879 : SUCCESS in 1h 53m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/46750b86cbc54d5f8e56f931af92c162 : SUCCESS in 1h 19m 39s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f86ee663d74242eaadcfa2e2876948b8 : SUCCESS in 1h 56m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e08045a84d74bada19feff4455fe79b : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ce8f88d105d4a199720e0533a59b233 : SUCCESS in 7m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4cfa4dfe354a44279e112fae13db7a6e : SUCCESS in 1h 32m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d763c69cb19e23f2f0412a933023adb6b6f61657","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-04 17:07:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff7c0ab119bb905aaade60a661a7d24fad894169","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-04 17:26:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f60fd46317655486a803398b0379ba844672e167","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-04 19:16:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b473920648eb727d366d9c1af7112cfcb71764b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-05 08:06:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e57fd41d00d68ce23740568740ff11a0eb034a8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 08:07:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f76b0363d9cb85cebc5fbd1e5c2601123eb5041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 09:35:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3d9842da0cb54160b03b729edcca2bc6 : SUCCESS in 45m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/655abe66b21d4e229c2daef593259f77 : SUCCESS in 1h 16m 57s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fa6cd74845f54caa95275c16d6e80d32 : SUCCESS in 1h 17m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/87feed76c6e8476484d920a4a1505977 : SUCCESS in 33m 50s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/97f9823afbaf48e19833e0545a61e453 : SUCCESS in 1h 00m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c898be876cbb41b19955e39ef7e72863 : SUCCESS in 1h 21m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2ff0f0cfe6547ac831408def99e4487 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e29bba936394970ba38d5eb12f21bf0 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c6cd741bece4bf3b74c0ed1ab2eb7c0 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8afb6e091af450fbe61f2ee93ef5a83 : SUCCESS in 5m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4fbb3a5a5cc416f918432cbcccb0a6e : SUCCESS in 37m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e781e36f983f4eaaa9d14d1583619d4b : SUCCESS in 52m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/83d563bfa9544a988fb4aa637c2f9318 : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":1},{"id":"506f1cd4770df962c4c70474512732670dadc5f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 09:35:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d4f0b605c14d16c0d0b8e774edc0282e9ad3a40c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 09:38:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ada145d350164818b109e9460b365111 : SUCCESS in 2m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"976465c448c4de2bd323021b946770b26ab1e912","revisions":{"976465c448c4de2bd323021b946770b26ab1e912":{"kind":"REWORK","_number":1,"created":"2022-04-04 11:37:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/17/836317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/836317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/836317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/836317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/836317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/836317/1"}}},"commit":{"parents":[{"commit":"a8fcb02f40c94c2101edff9ec4a3101d1d14fe97","subject":"cloudkitty-api: make sure that we install packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8fcb02f40c94c2101edff9ec4a3101d1d14fe97"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-04 11:35:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-04 11:37:32.000000000","tz":120},"subject":"elasticsearch: install Java first on CentOS too","message":"elasticsearch: install Java first on CentOS too\n\nelasticsearch-oss package fails to install if we install it with Java:\n\nINFO:kolla.common.utils.elasticsearch:could not find java; set JAVA_HOME\nINFO:kolla.common.utils.elasticsearch:error: %prein(elasticsearch-oss-0:6.8.23-1.noarch) scriptlet failed, exit status 1\nINFO:kolla.common.utils.elasticsearch:\u001b[91mError in PREIN scriptlet in rpm package elasticsearch-oss\n\nBackport down to ussuri needed.\n\nChange-Id: I72d7920acd8d15941c8c57a4186186212b273a38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/976465c448c4de2bd323021b946770b26ab1e912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/976465c448c4de2bd323021b946770b26ab1e912"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
