)]}'
{"id":"openstack%2Fkolla~836345","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Ie28e8580406f3f805500ed4020895e5e9ebf7c31","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-05 09:28:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie28e8580406f3f805500ed4020895e5e9ebf7c31","subject":"cloudkitty-api: make sure that we install packages","status":"MERGED","created":"2022-04-04 19:20:10.000000000","updated":"2022-04-05 09:30:38.000000000","submitted":"2022-04-05 09:28:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836345","cherry_pick_of_change":836344,"cherry_pick_of_patch_set":1,"meta_rev_id":"5a38cb5cf9ede23555206ef3671d547ac6b7ee38","_number":836345,"virtual_id_number":836345,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 09:28:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-05 07:42:40.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-04-04 19:20:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-04-05 07:42:40.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-04 19:20:10.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-04-04 20:42:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 07:42:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"77cf9c87436a552c635dcd531a9c83e94439b740","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-04 19:20:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"8c9b99cd421a799f10cfbc2d6f56e07ae2748a5e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-04 19:20:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1d03b73eec77f683c50aac2b470dbe393e52cdf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 20:42:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4419db0fa5914da894bced98e848cd25 : SUCCESS in 28m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b2509ebf06344e42bacfb4f31238bade : SUCCESS in 48m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/504cf9b6ab634da085fe85525d82924f : SUCCESS in 1h 05m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e0e881dfe621487ab6cc155b5818b48c : SUCCESS in 1h 15m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b998fabe3b0a4ad9a3a880d783ba3bfa : SUCCESS in 1h 12m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8fba1589c3914045b4f7b3586e05569e : SUCCESS in 29m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/565e3cb2bea64809846eb2a9b7b853bb : SUCCESS in 14m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/39ab4deb9a5a4f0bbd44a5632751c353 : SUCCESS in 57m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d4f573f67414fcb8cc442dcad5f2314 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2d5c95828c24f3591d9c0e7ce245a9c : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f4417cd7f384067aa474debc323cb84 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fb0c1534b344527836201a4ef7690cb : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07e2fa39d8ad46d0855d31c3c4c41421 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c0dc40e124e74d07a39dbf6de9557a0a : SUCCESS in 20m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d41a5147d5ab4c78bdb7fb8ca8eabfcd : SUCCESS in 34m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a8510c130584ce2a4d174f57d4dea09 : SUCCESS in 55m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c82530ed8c9f463da7c8de724c7b6fda : SUCCESS in 58m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e4b99aab97384ec3a630c2f123a14457 : SUCCESS in 1h 09m 35s","accounts_in_message":[],"_revision_number":1},{"id":"43f076417f898c4e2caaab1471a09f6c90a8d3c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 22:37:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/70013911fdaf40bc9a7bbf79ec31cbac : FAILURE in 1h 06m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a861054a61a54dde89a4287a1fda6a2b : SUCCESS in 1h 52m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/822e4113f6474b7ebc055e8942ebd2bb : SUCCESS in 2h 34m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/38ba7dc61fe14c55909b954be3de07ed : SUCCESS in 2h 05m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c6c3308964d6c86589ec2733b00c7752be71b83","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-05 07:42:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff6b54b717727cf23e548303ca7d0872e2092a45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 07:43:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7edd02d1f57c689280e80a7cec1736f75184f812","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 09:28:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7401721bc67c4b1ebb706170c9841470 : SUCCESS in 31m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b4ede5ca80794293a4c92ee35b9f8723 : SUCCESS in 49m 03s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/133bae9e88bd40078e201f13786883ab : SUCCESS in 59m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f2b2767357248b7a6ef02ffedf6b3b1 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77f24bd46c4f4cc39cfad36a7e6e8007 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdfdf1acf34d4a9fbb9384c172bbe572 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d3b4f33fcf04ea5979c842dbe257bde : SUCCESS in 5m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7f8951946c54a6dbac204fea2c2ceb1 : SUCCESS in 23m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad416f301c8e4d40bd88ae7b80f19162 : SUCCESS in 37m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/74499660edc04532b9a4cb080b7a9641 : SUCCESS in 1h 09m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7ea751a82a714adfef438cb120b9a445d44131e3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 09:28:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5a38cb5cf9ede23555206ef3671d547ac6b7ee38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 09:30:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c69ed406110b4213aa51c5c4692d978b : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8708b713afa96f23b00d53197811eb04270a6ff","revisions":{"d8708b713afa96f23b00d53197811eb04270a6ff":{"kind":"REWORK","_number":1,"created":"2022-04-04 19:20:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/45/836345/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/836345/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/836345/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/836345/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/836345/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/836345/1"}}},"commit":{"parents":[{"commit":"4ac3ae7186a86a4fa8452b51187ea3df05d5bed7","subject":"macros/pip: revert to old setuptools way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ac3ae7186a86a4fa8452b51187ea3df05d5bed7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-04-01 10:56:56.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-04 19:20:10.000000000","tz":0},"subject":"cloudkitty-api: make sure that we install packages","message":"cloudkitty-api: make sure that we install packages\n\nDebuntu binary images did not got packages installed.\n\nChange-Id: Ie28e8580406f3f805500ed4020895e5e9ebf7c31\n(cherry picked from commit a8fcb02f40c94c2101edff9ec4a3101d1d14fe97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8708b713afa96f23b00d53197811eb04270a6ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8708b713afa96f23b00d53197811eb04270a6ff"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
