)]}'
{"id":"openstack%2Fkolla~836428","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I366a092f7657dc179867322683767e0c9acf5bd3","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-05 10:13:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I366a092f7657dc179867322683767e0c9acf5bd3","subject":"DNM: test jinja\u003c3","status":"ABANDONED","created":"2022-04-05 08:00:17.000000000","updated":"2022-04-05 10:13:37.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6b3eada7787531dd36e2a8d7a85c134d4abde6e2","_number":836428,"virtual_id_number":836428,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-05 08:43:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 08:12:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 08:44:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"498d214b020c502ceb827d62c62d864b594ff1da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-05 08:00:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7bf109943e43741baad8f24c8edbd5a460f729c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 08:12:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e6bc5886744b4139bd96114f2c702b77 : FAILURE in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1a86f51760c023c9e94d19739f1a9ad48a675a0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-05 08:35:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ba7de1953c3bdeba671a85fd7beb9dee8c78890c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 08:43:47.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"06f4f9828bcfacd02174f6d18a5ef2f85169b793","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 08:43:51.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"14dc920325d9b4dd89cd9145c774590f5963e2ea","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-04-05 08:44:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a599ed78ae0c0a41a9624b78e0cc286adb87f4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 09:38:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b1eb45e8394a4a46a453fa2057ac4ce2 : SUCCESS in 47m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d27a66178322484d94394f61d3f914a6 : FAILURE in 4m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a9cd2167c67446a9a1892e560508dd11 : FAILURE in 3m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bc43bd6980f34fe695655f876717fec1 : FAILURE in 12m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d25817902a12494d99ff841e65d9c862 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a8839ab6f484c5aabfb11a421237e9c : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac6c5fcd629243899efe82d33b08e8c8 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bce19eb9836a448dbedd26e4a815fbfd : SUCCESS in 4m 23s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/42038a370e0e4789939d2bb9f27a0edf : FAILURE in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21860b657b064cd59a28b067a0b8382e : SUCCESS in 5m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a47407b343634c2ca82bbc6ee1b5a26d : FAILURE in 3m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce0ae51e068e471796540597b62de17c : FAILURE in 3m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c021d40e4d234313960cfbb4d576696f : FAILURE in 9m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/111bb614e52f475db2bf4489836a7c06 : FAILURE in 9m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e241ba8a3ea841d2abf71faee23040b1 : FAILURE in 7m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6b3eada7787531dd36e2a8d7a85c134d4abde6e2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-05 10:13:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1777f78fc010571c4b56815b34c5811fcbf029f3","revisions":{"3c9de856b7834a18f42ba63de290ec7b939b6d42":{"kind":"REWORK","_number":1,"created":"2022-04-05 08:00:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/836428/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/836428/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/836428/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/836428/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/836428/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/836428/1"}}},"commit":{"parents":[{"commit":"4ac3ae7186a86a4fa8452b51187ea3df05d5bed7","subject":"macros/pip: revert to old setuptools way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ac3ae7186a86a4fa8452b51187ea3df05d5bed7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-05 08:00:00.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-05 08:00:00.000000000","tz":60},"subject":"DNM: test jinja\u003c3","message":"DNM: test jinja\u003c3\n\nChange-Id: I366a092f7657dc179867322683767e0c9acf5bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c9de856b7834a18f42ba63de290ec7b939b6d42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c9de856b7834a18f42ba63de290ec7b939b6d42"}]},"branch":"refs/heads/stable/wallaby"},"1777f78fc010571c4b56815b34c5811fcbf029f3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-05 08:35:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/836428/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/836428/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/836428/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/836428/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/836428/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/836428/2"}}},"commit":{"parents":[{"commit":"4ac3ae7186a86a4fa8452b51187ea3df05d5bed7","subject":"macros/pip: revert to old setuptools way","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ac3ae7186a86a4fa8452b51187ea3df05d5bed7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-05 08:00:00.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-05 08:35:00.000000000","tz":0},"subject":"DNM: test jinja\u003c3","message":"DNM: test jinja\u003c3\n\nChange-Id: I366a092f7657dc179867322683767e0c9acf5bd3\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/836427\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1777f78fc010571c4b56815b34c5811fcbf029f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1777f78fc010571c4b56815b34c5811fcbf029f3"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
