)]}'
{"id":"openstack%2Fkolla~836573","triplet_id":"openstack%2Fkolla~master~I98b8edb575c69e59a1f342237a249bad049c8bcd","project":"openstack/kolla","branch":"master","topic":"zed-remove-support-for-binary-images","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-08 09:45:55.000000000","reason":"Change was abandoned"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-08 08:43:12.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I98b8edb575c69e59a1f342237a249bad049c8bcd","subject":"tests: drop testing for binary install type","status":"ABANDONED","created":"2022-04-05 09:20:48.000000000","updated":"2022-04-08 09:45:55.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5e9cd6ee40999ea1adc3d76f6c74dff6270ff66e","_number":836573,"virtual_id_number":836573,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-08 08:38:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-04-08 07:22:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-08 08:43:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 10:52:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-07 17:46:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-04-08 06:58:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"176170cc8cc11ba311bd48ed14edf9bc731daf51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffb5f1a6ffeb5e80c6d7799c5f2238331cb0aedf","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:21:43.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d0f84d7b0cbd72da1b5535d0b0a3b59f9fe9303a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 10:52:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2a36710a19c24a5bbf11c4d30bf064fc : SUCCESS in 40m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/93a7701f77184faf934348657eb02fd0 : SUCCESS in 1h 10m 13s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b0fbd0cdd78440b99bd2564930116e56 : SUCCESS in 1h 18m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cda6ecb5bdd14d5ebbab9a816b2a7810 : SUCCESS in 36m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ef35b0e9e7ec47ef9906c54650cdbe17 : SUCCESS in 58m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3e5d6bcacdbe4a6f856d857bdc627bab : SUCCESS in 1h 01m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1f56187d544e7885230a2ce4a81d53 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d66b39e48864bd0bef86fe1c3d92226 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/585f6e0360f9464fb957e11f64518848 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7939b4aaf7374f32b47aae14128b23dc : SUCCESS in 8m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/82b66a1f0aed45ab9aeeff43c2ad6eb1 : SUCCESS in 41m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5f21f8c7696245b282132eef6eb3c772 : SUCCESS in 59m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ba5e2528748c42d0a9c43759881a9b57 : SUCCESS in 1h 08m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3e4f3c9b250553ca5778c0a56e2b7027bd6f5b70","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:11:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9068a04376ed4db5bf562cb85147c285 : SUCCESS in 1h 56m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/243d6dab435346d58d6781f1a8689e62 : SUCCESS in 1h 21m 24s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/41fab4dadfa84dbaa36e390d4969813b : SUCCESS in 2h 07m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/287fa7f03c324172ac9d3f8d2a4c7a15 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d29be6960569450b81f43821b5e1c938 : SUCCESS in 5m 51s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b56f3772ff684262889626610f499dcc : SUCCESS in 1h 21m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ea9bb4fa9e48891e84b6b949beb1304cea285de","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-07 13:30:22.000000000","message":"Removed Workflow-1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz+kolla@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"7b02e82d4c3b92ff7e3e2a36ae9e8470584add7c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-04-07 17:46:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0573261d2c5b6b827deb114ef338fc4c29b90bf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 06:58:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de95d9c6352fa3af69ddd05815a8608b70f8a632","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 07:05:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cefb421e020c6493b4673f0422943c15efa04469","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 07:06:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"79656e871893afa46586559d4227a6172b13b3a2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 07:07:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a265e3cb11d3c60d356aac9eb3adf75e7aa0381d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-04-08 07:22:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"77ff8fb689e793010da5c680a23812dd6c4329e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 08:38:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f194e19ef6364229991a1d9516d6c494 : SUCCESS in 50m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0f9d0e4331684aae893b4ccd81cddd61 : SUCCESS in 1h 02m 38s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0796524223f74293bc71120e38e05b76 : SUCCESS in 1h 27m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8b763413532247c7b9958cfff8894a67 : SUCCESS in 40m 58s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d75fc792f03d4e64bd3cf5fd731b870f : SUCCESS in 1h 02m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6beaac1103224938a6cd3d144ee0daa7 : SUCCESS in 1h 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/812fa806441c4ba4a8a944a2cd06f6d7 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dce0e509c4f948f9a1080c668e9e09ba : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66922f3b37cd4b98b5302ac2a7811ece : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf4c7d376f34417180251241d5c6a059 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a2b4ae97314089be7861e8c33fa3af : SUCCESS in 5m 24s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e2e9916d5504ca6aaa89021d6773377 : SUCCESS in 28m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b5c73e83b60457aa262758446ed192a : SUCCESS in 55m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c18c87a709304fb69aa35483d25ed299 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":3},{"id":"364a25d9fbb9e84052d25918af757f2b8b4d1ac9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 08:43:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"63b16224981d275a76b22bde52befbd87321156a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 09:28:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3fe99759a2dd459797ba7bd7ab4bb4ce : SUCCESS in 2h 00m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/593e25833f0f45b0a1f709a5a5bab135 : SUCCESS in 1h 30m 11s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2a5b7205d2bf421b93cb8b0ffeea63fa : SUCCESS in 2h 11m 51s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8398771d08b943df98d8daf4285a335f : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7de6ab7723034764a2e52c442cf0e7e0 : SUCCESS in 5m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/19d747aab7524091b829682536b8e9fd : SUCCESS in 1h 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5e9cd6ee40999ea1adc3d76f6c74dff6270ff66e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:45:55.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/kolla/+/836574","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c0806e43424d52ef7859368361fbf83831fabb93","revisions":{"e109e54fb0469a040d4883afc5dff135727697e4":{"kind":"REWORK","_number":1,"created":"2022-04-05 09:20:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/73/836573/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/836573/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/836573/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/836573/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/836573/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/836573/1"}}},"commit":{"parents":[{"commit":"12dd8871d10e7ce563dcdc64acea1cb1a76ea954","subject":"CI: remove binary jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12dd8871d10e7ce563dcdc64acea1cb1a76ea954"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 10:36:00.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-05 09:17:35.000000000","tz":120},"subject":"tests: drop testing for binary install type","message":"tests: drop testing for binary install type\n\nDuring Yoga PTG we agreed about dropping support for binary images in\nZed cycle.\n\nOn start of build user was greeted with \"binary images are deprecated\"\nmessage when tried to build them.\n\nPart 2: dropping all binary images tests\n\nChange-Id: I98b8edb575c69e59a1f342237a249bad049c8bcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e109e54fb0469a040d4883afc5dff135727697e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e109e54fb0469a040d4883afc5dff135727697e4"}]},"branch":"refs/heads/master"},"ee26af82af3ece3104e91a732410fc9e1dffb79e":{"kind":"REWORK","_number":2,"created":"2022-04-08 07:05:56.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/73/836573/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/836573/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/836573/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/836573/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/836573/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/836573/2"}}},"commit":{"parents":[{"commit":"12dd8871d10e7ce563dcdc64acea1cb1a76ea954","subject":"CI: remove binary jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12dd8871d10e7ce563dcdc64acea1cb1a76ea954"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 10:36:00.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-08 07:05:56.000000000","tz":0},"subject":"tests: drop testing for binary install type","message":"tests: drop testing for binary install type\n\nDuring Yoga PTG we agreed about dropping support for binary images in\nZed cycle.\n\nOn start of build user was greeted with \"binary images are deprecated\"\nmessage when tried to build them.\n\nPart 2: dropping all binary images tests\n\nChange-Id: I98b8edb575c69e59a1f342237a249bad049c8bcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee26af82af3ece3104e91a732410fc9e1dffb79e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee26af82af3ece3104e91a732410fc9e1dffb79e"}]},"branch":"refs/heads/master"},"c0806e43424d52ef7859368361fbf83831fabb93":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-08 07:06:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/73/836573/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/836573/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/836573/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/836573/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/836573/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/836573/3"}}},"commit":{"parents":[{"commit":"7095af9fd0eddcba574bfbc1d203681ecda8a643","subject":"kolla: drop \u0027binary\u0027 install type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7095af9fd0eddcba574bfbc1d203681ecda8a643"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 10:36:00.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-08 07:06:48.000000000","tz":0},"subject":"tests: drop testing for binary install type","message":"tests: drop testing for binary install type\n\nDuring Yoga PTG we agreed about dropping support for binary images in\nZed cycle.\n\nOn start of build user was greeted with \"binary images are deprecated\"\nmessage when tried to build them.\n\nPart 2: dropping all binary images tests\n\nChange-Id: I98b8edb575c69e59a1f342237a249bad049c8bcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0806e43424d52ef7859368361fbf83831fabb93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0806e43424d52ef7859368361fbf83831fabb93"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
