)]}'
{"id":"openstack%2Fkolla~836574","triplet_id":"openstack%2Fkolla~master~If0e881d253d6a4c89ffe27831a661df321b6d5b6","project":"openstack/kolla","branch":"master","topic":"zed-remove-support-for-binary-images","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-08 13:09:37.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-08 11:42:22.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If0e881d253d6a4c89ffe27831a661df321b6d5b6","subject":"kolla: drop \u0027binary\u0027 install type","status":"MERGED","created":"2022-04-05 09:20:48.000000000","updated":"2022-04-08 13:11:02.000000000","submitted":"2022-04-08 13:09:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836574-zed-remove-support-for-binary-images","meta_rev_id":"a80945e41632475efc769256f6c7e51e76d405a3","_number":836574,"virtual_id_number":836574,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-08 13:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-04-08 09:52:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-08 09:45:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-04-08 11:42:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 11:11:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 07:22:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-04-08 08:42:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-04-08 08:43:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"d50e402b3cc53b856ec7c33fef03640193697c09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f3d2493b7ece53949a5070fb80036e9f0f19c95","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:21:36.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ff2fb5f255a38bd1fe699a68fa683665bb3c29dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 11:11:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ee8d2d60aea44fb4b12805c6a6236c09 : SUCCESS in 48m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e56d21e8de144ffb9daf4c8b63119340 : SUCCESS in 1h 36m 27s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b77c11729995453dabe6fa7e6d9a915f : SUCCESS in 1h 14m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/efec9aa6475c4980a231811496264cc0 : SUCCESS in 38m 05s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d324e0348d6e4e5a8bb31dd1532557bd : SUCCESS in 1h 04m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/93aef4e47d3e43698198dfe7e8d96cc2 : SUCCESS in 1h 17m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4fd644be3944f50aede2752a84fb0bb : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b5c54ca480c464ba2f0a7b2e8824cfa : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82f198af621845e4bc8e2416ba865c2d : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e1f66f4f9774e858efea53d4e157b93 : SUCCESS in 7m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca0d1bbf168c4fb6b3ffac1df5efab18 : SUCCESS in 28m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d54d4211e186490f8f7baaa0cf1c9850 : SUCCESS in 59m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/060c07d171dd47868db05b05399631c7 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":1},{"id":"aaa80d301871e61866c91a0e805eef1d90f1caf8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:25:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7df271625da74e778d0243ec7c375021 : SUCCESS in 2h 04m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8dfd4d7e9fee48e8ab71d2cd7f3fd90a : SUCCESS in 1h 48m 46s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d1f4f53798ec49f3a9635e3a9ef6e820 : SUCCESS in 2h 06m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70960bf217ff4c789ea7e4159ae9be02 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b0d5f1cd3224fcd89498ee99bb5d0e2 : SUCCESS in 7m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d54fa0c7020d4418be871d6c018eb72a : SUCCESS in 1h 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60daf46518540a450c4ab15179aa46680f305943","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-07 13:30:27.000000000","message":"Removed Workflow-1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz+kolla@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"d8cd9013d8ee61ce31a577b1044ae21e6cad5236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 07:06:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9db6157af07b1dfafca95b88c5dc1c31d648f158","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-04-08 07:22:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"790c45730253ebb0e9bab25bb321f2eada5c84e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 08:20:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b212460a5c7546ffb96b574f759f2851 : SUCCESS in 27m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/824baa4e5e3042968b48c1bf55bd5ffc : SUCCESS in 1h 08m 39s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/536bce0e9e6f41bab2269772796f8552 : SUCCESS in 1h 12m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8b39806fe1474ac3ba0fec17911bac1e : SUCCESS in 30m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d57cf1de0d994bae93db65ebf5d05bf8 : SUCCESS in 51m 13s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1766064d5afb40b28baa6f1b3b8255e1 : SUCCESS in 59m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/433b2ed425d5438d87b7f0d4342734c1 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89e4a552842846569afe060c15c67f4b : FAILURE in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ae77f1153cb4914b5cd0d87f30be62e : FAILURE in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc3bc6dffc8349c4b86a20c07a65aa82 : FAILURE in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe95450a6b0f41588d8b966ce55bb262 : SUCCESS in 5m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3693739ba97c4af0b6456329024838b0 : SUCCESS in 38m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/98f97193459b4f8c9cd8ebc2616b2452 : SUCCESS in 48m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bf34343ec117471da7d2d56782024cbf : SUCCESS in 1h 05m 10s","accounts_in_message":[],"_revision_number":2},{"id":"051a70355bca731c1b4eb16f2cd224a0c86c099a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 08:42:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cdf310169160559a0468f6dd99632b062fd7bd63","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 08:43:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b0f3d053d79f6ab3ddac674919edf8640ccd1c90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 09:25:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8beeae6cf824226abf76b73541a594c : SUCCESS in 2h 00m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e48f267892de4e81849a7652cb089663 : SUCCESS in 1h 53m 55s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dde508d8e67f4f588cc1637fab9e8e63 : SUCCESS in 2h 17m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6fab2ba821b4d6387326639a93dbfb7 : FAILURE in 6m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d3d9f2da283c4fae89038b4addab9687 : FAILURE in 6m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ebc378eaf19e470e90aec77e84b02ee6 : SUCCESS in 1h 25m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d721b1de35225bb70b0fdd9dff68a72f73aeeece","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:45:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"84c970d1a6cd92b055d9d269ba6f7602bb54549e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:45:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d6c3e5d1a23d10afb801a3e4d875733f48cd5e18","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-04-08 09:52:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d847570df04097a028d46751fd83e44a588f2083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 11:18:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5f775217dd9240458ad70af2d98e328a : SUCCESS in 40m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/86aa6b6337104e3bbb505aa5b76b9612 : SUCCESS in 1h 08m 28s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a175b917739246de90120abe59e10678 : SUCCESS in 1h 26m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8a9ef1c458b94ed8ab5dcfcb50626234 : SUCCESS in 30m 52s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4459a07238f842fb9dbb8e1ebe5c14d7 : SUCCESS in 1h 11m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0bad976d69fd49a9bfb35e3bd38e4df3 : SUCCESS in 1h 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad4e84114743495b872c883694d9d6ca : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20cf08c881b94359951793d525f5f73b : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/287340c2e671483fa54552e5a7b7d681 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6518b0c6be4b47d9b1ebce30c26ec37c : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09ad1ea22854b6799bfb5fc432deba6 : SUCCESS in 5m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a008b2f66c64306afaec62e162a331a : SUCCESS in 36m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/27e83d67d0e348c58671e720b232dbb2 : SUCCESS in 50m 11s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e6aff0e2928940cbbe328fec46e7b275 : SUCCESS in 1h 10m 08s","accounts_in_message":[],"_revision_number":3},{"id":"46f2b97ed5a7d0b5b6c217f13b2c38ae8f360aa0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 11:42:22.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"95c8513b8abfe7b72fbc973c3bd6074d1f38c7fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 11:42:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ae574e11a9f6a49fdec12c1a6e9991f041439bcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 12:01:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f4659866d9134a60ac18d8da531136b0 : SUCCESS in 2h 11m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/757e9fa0b6484011853e46aa3a70a93f : SUCCESS in 1h 44m 28s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2d215eed5a3d4138b29ff494826e6732 : SUCCESS in 2h 08m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2bd3cfff19794528a45304b91737a3e2 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/838a5e786d584ff184237826b9f05a39 : SUCCESS in 8m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4c393bccdd141399f6937d380d98bb3 : SUCCESS in 1h 49m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2f272d2cf8c48883837eae3a4054acdb519d4ffc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:09:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/12e6fc91bd254592bc0e344c63b5926d : SUCCESS in 37m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1357ff37b7e941b6b6348bb57317eb84 : SUCCESS in 1h 04m 01s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8897609d946f4819b42b3bef98d95022 : SUCCESS in 1h 26m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ab4d7a71e8b246048a6241a09b73341d : SUCCESS in 30m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/12233b8f5f17449c8868db4a2c4e069a : SUCCESS in 1h 00m 18s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2bdaf710f0214f0fa6c0b7ed7bbd1583 : SUCCESS in 1h 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/287a65c94fa64885aa21a20f36e4703e : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfb0692b4efc4a84847af05c7638f3a3 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e064340f152e409b9db2f7126d7c7b26 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12ae981fe9dc4eef904bcc05b84ebc13 : SUCCESS in 6m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b8487cbf0a941f297430c94405fe4be : SUCCESS in 34m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e5735c2419744d9dbf5c86c6ce34f68e : SUCCESS in 54m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/debcc63367504336b30940aa021aa1e8 : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":3},{"id":"f6648d42e9887010af446304942dc7cd5ee940cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:09:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a80945e41632475efc769256f6c7e51e76d405a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:11:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4514ca7f144a44d0b9fd17016998eb90 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9b8b22b8d4d5e89820e54e819ef7f7d310f0d040","revisions":{"f90d466d0e094d602ed9483db337c6548a326347":{"kind":"REWORK","_number":1,"created":"2022-04-05 09:20:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/74/836574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/836574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/836574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/836574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/836574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/836574/1"}}},"commit":{"parents":[{"commit":"e109e54fb0469a040d4883afc5dff135727697e4","subject":"tests: drop testing for binary install type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e109e54fb0469a040d4883afc5dff135727697e4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 10:46:23.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-05 09:17:53.000000000","tz":120},"subject":"kolla: drop \u0027binary\u0027 install type","message":"kolla: drop \u0027binary\u0027 install type\n\nWe no longer support binary images. This patch removes a way to build\nthem.\n\n\"--install-type\" argument has one available option now. Later in cycle\nit will be ignored.\n\nChange-Id: If0e881d253d6a4c89ffe27831a661df321b6d5b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f90d466d0e094d602ed9483db337c6548a326347"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f90d466d0e094d602ed9483db337c6548a326347"}]},"branch":"refs/heads/master"},"7095af9fd0eddcba574bfbc1d203681ecda8a643":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-08 07:06:33.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/74/836574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/836574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/836574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/836574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/836574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/836574/2"}}},"commit":{"parents":[{"commit":"d8a7119009f50631045565940921b785e71c0c21","subject":"Merge \"CI: remove binary jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8a7119009f50631045565940921b785e71c0c21"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 10:46:23.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-08 07:06:33.000000000","tz":0},"subject":"kolla: drop \u0027binary\u0027 install type","message":"kolla: drop \u0027binary\u0027 install type\n\nWe no longer support binary images. This patch removes a way to build\nthem.\n\n\"--install-type\" argument has one available option now. Later in cycle\nit will be ignored.\n\nChange-Id: If0e881d253d6a4c89ffe27831a661df321b6d5b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7095af9fd0eddcba574bfbc1d203681ecda8a643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7095af9fd0eddcba574bfbc1d203681ecda8a643"}]},"branch":"refs/heads/master","description":"Rebase"},"9b8b22b8d4d5e89820e54e819ef7f7d310f0d040":{"kind":"REWORK","_number":3,"created":"2022-04-08 09:45:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/74/836574/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/74/836574/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/836574/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/836574/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/74/836574/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/74/836574/3"}}},"commit":{"parents":[{"commit":"d8a7119009f50631045565940921b785e71c0c21","subject":"Merge \"CI: remove binary jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8a7119009f50631045565940921b785e71c0c21"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 10:46:23.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-08 09:45:04.000000000","tz":120},"subject":"kolla: drop \u0027binary\u0027 install type","message":"kolla: drop \u0027binary\u0027 install type\n\nWe no longer support binary images. This patch removes a way to build\nthem.\n\n\"--install-type\" argument has one available option now. Later in cycle\nit will be ignored.\n\nChange-Id: If0e881d253d6a4c89ffe27831a661df321b6d5b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b8b22b8d4d5e89820e54e819ef7f7d310f0d040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b8b22b8d4d5e89820e54e819ef7f7d310f0d040"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
