)]}'
{"id":"openstack%2Fkolla~836575","triplet_id":"openstack%2Fkolla~master~I7bdd29eb48a23faa04a98e785129d97258f05e64","project":"openstack/kolla","branch":"master","topic":"zed-remove-support-for-binary-images","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-11 14:50:55.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-11 14:50:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7bdd29eb48a23faa04a98e785129d97258f05e64","subject":"drop add_binary_source_envs() macro","status":"MERGED","created":"2022-04-05 09:20:48.000000000","updated":"2022-04-11 14:50:55.000000000","submitted":"2022-04-11 14:50:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836575-zed-remove-support-for-binary-images","meta_rev_id":"9462a08f57b534d2f1c9dd0a6aafc1fe23e39156","_number":836575,"virtual_id_number":836575,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 14:50:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-11 14:50:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-11 10:49:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-04-11 14:50:55.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-04-11 10:49:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 11:07:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 09:52:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-08 14:27:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-11 10:49:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"e73ff600dd42709a4395e14a1cbb3da81a7c413a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50c1a6c20d60c0d0a13e770df1ac74b50e82faf5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:21:28.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d0f7be4f10a8197dbe7a214772cbd502c0249459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 11:07:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fa30aae99f084d2ebeb8e7aedab7ea63 : SUCCESS in 52m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8d9b0e47872b4757bdeb0a831450e126 : FAILURE in 57m 16s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/638a08c3fe0a455c906cf67cc51d4834 : FAILURE in 1h 34m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/376130308f4442549f54ad335005e39e : SUCCESS in 39m 04s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/30f10a5531e14b2b9e979375263e9a6b : FAILURE in 37m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/76e9406a4f2a4935ae3d0c4e141abad1 : FAILURE in 1h 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154a1f14f1e044ffbbb65a640388f5cd : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f61b82334a0e4ea1bd203640798c3ed9 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7ecc6f419784e6ab3e04a30817f3965 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3be16c140b4a46c5817aba8343648a51 : SUCCESS in 6m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/23f3f78d31b54a5289b3f0e5517c7ec5 : SUCCESS in 28m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/be3d8f3771794fdf9e3ce568596bf151 : FAILURE in 39m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5d0b2b4c0b3e4abfbca030b1c1c53425 : FAILURE in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a1df9e58c527f50a6534238790bbc47516a3ab6f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 12:58:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af2303328fd520ed8314122f264af9b13e786ef3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 14:53:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/33f28cf0130c4dd4a3c08f7391b25470 : SUCCESS in 1h 53m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b04cd1b5ffc94d40b6d3c12977acfa34 : SUCCESS in 1h 27m 59s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a60909105e34d718bd58354ca850499 : FAILURE in 1h 31m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6e9a369dbe94e549f58546dce5ce971 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/296f01dcb19a464d9b103eb11aea7e45 : SUCCESS in 6m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7eeeef2665a74b01a9b7e39423e2c318 : SUCCESS in 1h 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02eacc8a50b52276a69576baa9373b5438a7c697","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-07 13:30:34.000000000","message":"Removed Workflow-1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz+kolla@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"fb63a9635ce119b300cb77b81eed7cd4bee855ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:47:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"63c1b9534680f631ce5c43e0046eb92f132024e8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:52:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"232d9cd14fb38dd242b9180d49509e5e6f2842ea","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:53:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a2517d7de23dad371a4aa01b5a73e34a9d3ffad","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:54:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a99775c25b27ae9baef31d13d1787fbb6ba7d220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:55:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"53633b61bfa964927cfc450b16e3f46450e38b2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:58:44.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"7aba471f21b4494d46fa35d29cc9cbedc6648843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 11:16:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/88690e9183c14f10926334c07df687f9 : SUCCESS in 59m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a6e15fcbd62b454e88ea776f8fc64ed1 : SUCCESS in 1h 07m 38s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d6e09319b67f4f6ca3451176fe346ac8 : SUCCESS in 1h 15m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/511dd709f2604f929ae6011de9ef5ec9 : SUCCESS in 37m 53s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fedb0e12277d4311986a153c917364dc : SUCCESS in 1h 02m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b18c38b15c36472887fcc4f2f60a7144 : SUCCESS in 1h 00m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a9cbc2b18e460eb55e9ea26ee4f19d : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3ac04bc4cd04dfbac4bea0125489f94 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02b17812ea054aff9e2d27559c163497 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f5c96de97c3451f96c13b8d4ed9f416 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1844b2c317a64ab486189034aac27a34 : SUCCESS in 7m 17s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/733a7b8a45484435951df5687240ddf1 : SUCCESS in 32m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/190976e854b049c99914ca92f2d09a71 : SUCCESS in 53m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/75c9f04429ff40f0b297735d03ac7e92 : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":4},{"id":"c40940d3f4735c6efc823f11b8d1987f7663452e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:58:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/614679103d2e42b6880f3c2dd8575f4d : SUCCESS in 2h 15m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/571ccc552dba48019b27e8126dc1ea4f : SUCCESS in 1h 41m 28s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c0cbf2f3d6e44078b70fd3fac874084 : SUCCESS in 2h 19m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e6e5ba5ff1f4b6fa8a7ebd28245d4cd : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9ca6206d6c7492ba648d69bd768823c : SUCCESS in 7m 01s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b7b568ec2224c3abced10b577ad0e38 : SUCCESS in 1h 51m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc19715a1d380b41615b496eda0c7e823ff7e6eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-08 14:27:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"70d987839c88293c4712a34a926593083cdf5850","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 10:09:39.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"3687f29a3bf604356ea67573384819eba862558e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-11 10:49:46.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bff4065fa29b595583c756c46cae2c01b1e9c166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 11:23:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/66eff90a6211476fb838377b12233fff : SUCCESS in 30m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c9895fd46a8a42bf9029a4a614ef4de6 : SUCCESS in 1h 01m 59s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e2980a43f0114b088d8c463d4f801a01 : SUCCESS in 1h 10m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4303f4b32d604fb5b4d4179aa68827b6 : SUCCESS in 27m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cf0b88d2c7eb4e4598976bd81a5122ef : SUCCESS in 50m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/759546008ff1418592b2daee354e00b0 : SUCCESS in 58m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a68070130e74161b364462835f87c94 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d30f0e76edc1465bb4e72c1adf205d4e : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b9695d88e3d4f8a901382349eadd227 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4acece18d86f4f86bc28a1a36c78ba90 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a6b3b95d55448a2bbce36e5b3acb12d : SUCCESS in 5m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1833e89652eb47c893d8c164426c592e : SUCCESS in 28m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/184981985b3d4f758d65ec3d0a0dd48f : SUCCESS in 54m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fb5199e4ba814df98b4c73e0011755e6 : SUCCESS in 1h 08m 01s","accounts_in_message":[],"_revision_number":5},{"id":"8a904f2226e902e79158a60f35463727d11d96f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 11:23:43.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"50cc287621e9dd038ebd1e0471d4d02f16eb26c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:22:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d6b6d4a233874763b4c0f5d71e72df41 : SUCCESS in 2h 07m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f91e969b9b841d090e7930c3886cd48 : SUCCESS in 1h 58m 39s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1bbfe7651f3b467f91a4159d545c3527 : SUCCESS in 2h 10m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/70d160951cf44ef0ad71c6ff6c4d1bd2 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e81961a4c66d4055ae70f5828bbcef65 : SUCCESS in 6m 24s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9af4531bd669460883dbdaf67bba380b : SUCCESS in 1h 57m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a93e20c62b0aa17a442387ab7514557622aeb6b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:50:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e4ed83d9f4ca469dbd35a752efb60c82 : SUCCESS in 46m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/21616b5195b14dc5bdd28633785e71dd : SUCCESS in 1h 19m 43s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e8d60bfef4f84dcea385ecc87a7cb966 : SUCCESS in 1h 25m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dd9253d3d7814084b825eff88d8e89b5 : SUCCESS in 33m 03s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8b7073ec4a7b4da18bd7e3fb8506f264 : SUCCESS in 1h 04m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/59bf5a511fc047cbbd1c7e695ccae727 : SUCCESS in 1h 19m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889f72f19b034b88a37a5986bd97d11d : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b6e9d62492c4878b3c7223c63f75bb9 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7cfdc7aa0fc4297aecd063ab52030d3 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa1bd3ddd08e428d853ac3322c7262d3 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/df6c2b34177e4f27970ac2c821eae6e2 : SUCCESS in 39m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe6af0c9a67647d0adfdf56cf9dff560 : SUCCESS in 58m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/48db5d7daed6458192dbf8c22f478a65 : SUCCESS in 1h 06m 18s","accounts_in_message":[],"_revision_number":5},{"id":"c56ffc61a3e048da754f9dc0cb23befde3ed9d11","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:50:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3915785884bd9a1746f635196a0e361cc70fbe2d","revisions":{"149063d2731e6c4ac36737f3493b088a695de1c2":{"kind":"REWORK","_number":1,"created":"2022-04-05 09:20:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/75/836575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/836575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/836575/1"}}},"commit":{"parents":[{"commit":"f90d466d0e094d602ed9483db337c6548a326347","subject":"kolla: drop \u0027binary\u0027 install type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f90d466d0e094d602ed9483db337c6548a326347"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 13:27:43.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-05 09:20:32.000000000","tz":120},"subject":"drop add_binary_source_envs() macro","message":"drop add_binary_source_envs() macro\n\nWith binary install type dropped there is no need for this macro.\n\nChange-Id: I7bdd29eb48a23faa04a98e785129d97258f05e64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/149063d2731e6c4ac36737f3493b088a695de1c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/149063d2731e6c4ac36737f3493b088a695de1c2"}]},"branch":"refs/heads/master"},"ede8f995fdc2dfe401722e6a68e59d0ec6392299":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-08 09:47:19.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/75/836575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/836575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/836575/2"}}},"commit":{"parents":[{"commit":"9b8b22b8d4d5e89820e54e819ef7f7d310f0d040","subject":"kolla: drop \u0027binary\u0027 install type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b8b22b8d4d5e89820e54e819ef7f7d310f0d040"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 13:27:43.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-08 09:47:19.000000000","tz":0},"subject":"drop add_binary_source_envs() macro","message":"drop add_binary_source_envs() macro\n\nWith binary install type dropped there is no need for this macro.\n\nChange-Id: I7bdd29eb48a23faa04a98e785129d97258f05e64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ede8f995fdc2dfe401722e6a68e59d0ec6392299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ede8f995fdc2dfe401722e6a68e59d0ec6392299"}]},"branch":"refs/heads/master","description":"Rebase"},"5cd955ecd6a43eb428435fde7b06e8c395dddda4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-08 09:55:27.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/75/836575/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/836575/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/836575/3"}}},"commit":{"parents":[{"commit":"84dee629ce0909248c7a688a3019e0406abce4fb","subject":"docker: drop binary parts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84dee629ce0909248c7a688a3019e0406abce4fb"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 13:27:43.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-08 09:55:27.000000000","tz":0},"subject":"drop add_binary_source_envs() macro","message":"drop add_binary_source_envs() macro\n\nWith binary install type dropped there is no need for this macro.\n\nChange-Id: I7bdd29eb48a23faa04a98e785129d97258f05e64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cd955ecd6a43eb428435fde7b06e8c395dddda4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cd955ecd6a43eb428435fde7b06e8c395dddda4"}]},"branch":"refs/heads/master","description":"Rebase"},"084f5505988071b3e8ee7050fbf7a8c9cc9c7170":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-04-08 09:58:44.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/75/836575/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/836575/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/836575/4"}}},"commit":{"parents":[{"commit":"a59d191c0fde3cbc7c1f0d78cecc810012c249d5","subject":"docker: drop binary parts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a59d191c0fde3cbc7c1f0d78cecc810012c249d5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 13:27:43.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-08 09:58:44.000000000","tz":0},"subject":"drop add_binary_source_envs() macro","message":"drop add_binary_source_envs() macro\n\nWith binary install type dropped there is no need for this macro.\n\nChange-Id: I7bdd29eb48a23faa04a98e785129d97258f05e64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/084f5505988071b3e8ee7050fbf7a8c9cc9c7170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/084f5505988071b3e8ee7050fbf7a8c9cc9c7170"}]},"branch":"refs/heads/master","description":"Rebase"},"3915785884bd9a1746f635196a0e361cc70fbe2d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-04-11 10:09:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/75/836575/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/836575/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/836575/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/836575/5"}}},"commit":{"parents":[{"commit":"657820362dd0b52f8462054e043fd892d14fbe76","subject":"Merge \"docker: drop binary parts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657820362dd0b52f8462054e043fd892d14fbe76"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 13:27:43.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:09:29.000000000","tz":120},"subject":"drop add_binary_source_envs() macro","message":"drop add_binary_source_envs() macro\n\nWith binary install type dropped there is no need for this macro.\n\nChange-Id: I7bdd29eb48a23faa04a98e785129d97258f05e64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3915785884bd9a1746f635196a0e361cc70fbe2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3915785884bd9a1746f635196a0e361cc70fbe2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
