)]}'
{"id":"openstack%2Fkolla~836576","triplet_id":"openstack%2Fkolla~master~I28b7563e5ce1efac0298a62fee96a9f946f3341b","project":"openstack/kolla","branch":"master","topic":"zed-remove-support-for-binary-images","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-11 07:37:22.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-08 15:57:33.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I28b7563e5ce1efac0298a62fee96a9f946f3341b","subject":"drop infra-rename support","status":"MERGED","created":"2022-04-05 09:20:48.000000000","updated":"2022-04-11 07:38:46.000000000","submitted":"2022-04-11 07:37:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836576-zed-remove-support-for-binary-images","meta_rev_id":"f9a66786bae3fb3762617eacb90dd3cb91399589","_number":836576,"virtual_id_number":836576,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 07:37:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-04-11 06:22:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-08 15:57:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-04-11 06:22:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 11:07:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 09:58:17.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-11 06:22:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"224a43ab167be75322d5e873169679c7c1596543","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebb1f0b977522ab00ab44bb6de5c4bcf7d8f31b5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:21:15.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"930847d32f7a52a6bcd26b81f88b9a5a67a6cfef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 11:07:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1533417a27bf4b928abe438c6769ae95 : SUCCESS in 45m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d2996ea8ec8044df92eff982a6ffe1e8 : FAILURE in 1h 06m 20s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/581baf2d226040dd959204983c176b70 : FAILURE in 1h 32m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2ef4c5987e1842d197b223dd416819c9 : SUCCESS in 31m 16s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d1ea136662394c0fa8a34f04b22cf9dd : FAILURE in 52m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9a783b34117443d29628385b25fb1d6a : FAILURE in 1h 16m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e413eaccb4847b1b5d935c8b3404952 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59e2ec1658d14c91b8a40718a3dd87b7 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fa2a0b7c76e465692164f75b76abcf7 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d885aa219624f7ea3f74a039660c21f : SUCCESS in 8m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f41fc86f9a5c4c36897a4594a401bea2 : SUCCESS in 36m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/60ef393f74664c24955f81b279047a7b : FAILURE in 43m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c424ada210954f50978bb08842fdaf9f : FAILURE in 1h 06m 21s","accounts_in_message":[],"_revision_number":1},{"id":"71f31c868f7b648494d90f04c08b5398f84fb8bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:50:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/0ac0c36fa320498d831fd62b6586099c : FAILURE in 2h 15m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c4d1ad8a40114ab180c9e3d5fc8ff255 : FAILURE in 2h 13m 07s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/19941fdf700e428f8267521b81e29251 : FAILURE in 1h 41m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3731a89069074b2594968ae55cf52c5b : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ccdfc2fb67a4cf5a050868348af4c08 : SUCCESS in 7m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/8b724d4e9c81491ab81fc3d83b9999eb : FAILURE in 1h 33m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a536ad59c936bacae218a854a3e9d59b9d8abcce","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-07 13:30:09.000000000","message":"Removed Workflow-1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz+kolla@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"6af05bd4ff61ecebe6ceccb84bc8668dc9f38aff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:47:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a3efd29531513b083b395b04ab93d5832eaefc56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:54:49.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ec789c50a796b9d1df71eebf4e8c0c7ab1e025cb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:58:17.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9b4729c77cbf7ade092bacba896a1a1a39f30804","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 10:20:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fc4531e61db4e359607decd699fefeb478c02e28","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 11:37:22.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"70a5793bc31d5418c0e45284babc896ca442b263","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 11:38:04.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0132f5d86985d13e9de8c88e94139553bf5d0207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 11:40:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8b48de62a3ef4137838b2a9678546f55 : SUCCESS in 30m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fae8a30ac8ee4bd9ad24dbb1c97237d9 : SUCCESS in 1h 24m 04s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e7a4133a1cd54852b737ccff6afffb75 : SUCCESS in 1h 38m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/324dcb03efaf4745835f810133e0f310 : SUCCESS in 38m 20s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c41045e810224f31826f58d0a5b34435 : SUCCESS in 1h 02m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/46661aa99aec42858809ae6a39fdcb66 : SUCCESS in 1h 06m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10bf6daad1b747d0af6431f47fe262c7 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cea6408abd6242dca0082fa67e328650 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a92598e452c348179aa4b7303b4f5018 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34410e9969f749bd89c88c11592dee2c : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cd44e31986148d1ad23299a1526470d : SUCCESS in 6m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4b7d121ff344a918d348a63e659d304 : SUCCESS in 32m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/66b6bcd074804bbbbce087b2fdd04d99 : SUCCESS in 53m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/52de13928f2a45d1836d46724c1ef39c : SUCCESS in 1h 15m 27s","accounts_in_message":[],"_revision_number":3},{"id":"2af1aa47a185d6c8d7292efade98b803a99a3bc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 12:42:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ef3e7ee02614dab8c2caa9fd8ef0c4f : SUCCESS in 2h 18m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fc5ef2fbf7cc4f5fa2819edface56d35 : SUCCESS in 1h 24m 13s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3e972bbb63f04eaa895f54767eccda70 : SUCCESS in 1h 57m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a4417c45f14457a91a2f2be832f93b5 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e15005f6ee2c4c0089a2c6a8487fd92f : SUCCESS in 6m 27s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5dd03b7047c14d41a0466e881a8617e9 : SUCCESS in 1h 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6f87b29ed67e5361d7ed9623c12cdea464cf8045","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 15:55:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0a4da580410512d0364f901cce187fcc3b26621f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 15:56:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c1524f27fc402bf35cf432671ad4e08f8e56fa98","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 15:57:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9bb5bb2f176a97589ab604653c028f353c7e0503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 17:27:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8296c615d21142cdbb1251907dcb4fa3 : SUCCESS in 49m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dfc9839c034b4bbabe61232df619bbb9 : SUCCESS in 1h 03m 46s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ba8c8846dce54970a71fa9e968674e54 : SUCCESS in 1h 28m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/419b538f239847728ec8fdc9976b3943 : SUCCESS in 29m 48s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6315c4038e984f048816922cb8e89b83 : SUCCESS in 56m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/002e26d43da34215bf17159f9c339b32 : SUCCESS in 1h 09m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6774a5b5f6624c9ba621aadb07bc4645 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59dfcea319174c8383b24a50d428b81f : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/102b9c20f2b8491896c6f6147111f7f0 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a19cd7da348b483fbdd2f844604b060d : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c296bcb0b164ca0b1230e55e2ba8cd0 : SUCCESS in 7m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4039bbfb3c754a5b8c8c4b716b0ed8b8 : SUCCESS in 36m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec9f1fa7a70f4e0ab501ac57c4852a1b : SUCCESS in 55m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/52e00b4251a843bf9fe402afaace132c : SUCCESS in 1h 06m 41s","accounts_in_message":[],"_revision_number":4},{"id":"65746cdedb27bebac7e3b3de131f2dac70c9870c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 19:36:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5bb43cebe9c84dc9ae111643ea56503d : SUCCESS in 2h 09m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a53066d06bc41cebfa52a1ae6b0876c : SUCCESS in 1h 51m 39s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80edab3fa2df45f1969872c78e7d6ab1 : SUCCESS in 2h 06m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c44927a3ee6a442d98e9a2bd9b5c3ce1 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe91aa26bf5442fab06478b5bc469543 : SUCCESS in 6m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ece1adc7f434b53a0c78f55446e24eb : SUCCESS in 1h 44m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"07dc2f995c4d5aae9f4ae992a46c9713f617b9df","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-04-11 06:22:05.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"17bb8406fd83da998e9d3d7bebe8ea4fdbe9e5f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 06:22:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ce70e40de0a06110b727fe951e385f520a6eb244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 07:37:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3249e3ca7992464c9a87d2d6283323a5 : SUCCESS in 33m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ca4cfd4fc3cd476591e520cc2c1e8ca0 : SUCCESS in 1h 14m 06s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/25d1833e447545adab620c9916c42e8f : SUCCESS in 1h 12m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2a52db8a075245b094c44e665b93394e : SUCCESS in 24m 07s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/80d50196d1cb439da6e82f5461d282ad : SUCCESS in 58m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c04dda749f3542db95d27d0935488839 : SUCCESS in 1h 03m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b28c775cf647759cdc759ac6511944 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/282e1894475340fdab91e474ebedf1a9 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9f73fe8f073462688908bee87358405 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2f5d93f2e384e3fae682dee2af5aeb9 : SUCCESS in 6m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f9e8a6248d6478484ec3e383d4cfde7 : SUCCESS in 31m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/896b1ceae3434cb19ced3ea485f3a4b3 : SUCCESS in 54m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee2f4506e6994ee9bd3a3cbf932a2cc3 : SUCCESS in 55m 54s","accounts_in_message":[],"_revision_number":4},{"id":"a60fe124992536ed717388de1347d0007056c4b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 07:37:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f9a66786bae3fb3762617eacb90dd3cb91399589","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 07:38:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa42fa10f5f445cbe52d191742364d1 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ad53e1fe22fb04623718635faced5d14a4d17abe","revisions":{"263c634841f86d68769c54d370139fe610d9b09a":{"kind":"REWORK","_number":1,"created":"2022-04-05 09:20:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/76/836576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/836576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/836576/1"}}},"commit":{"parents":[{"commit":"149063d2731e6c4ac36737f3493b088a695de1c2","subject":"drop add_binary_source_envs() macro","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/149063d2731e6c4ac36737f3493b088a695de1c2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 13:31:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-05 09:20:33.000000000","tz":120},"subject":"drop infra-rename support","message":"drop infra-rename support\n\nWith one type of install type there is no need for renaming infra\nimages.\n\nChange-Id: I28b7563e5ce1efac0298a62fee96a9f946f3341b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/263c634841f86d68769c54d370139fe610d9b09a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/263c634841f86d68769c54d370139fe610d9b09a"}]},"branch":"refs/heads/master"},"e770353c35dce0e993f4afd9979d9d29a2aeafcf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-08 09:47:36.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/76/836576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/836576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/836576/2"}}},"commit":{"parents":[{"commit":"ede8f995fdc2dfe401722e6a68e59d0ec6392299","subject":"drop add_binary_source_envs() macro","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ede8f995fdc2dfe401722e6a68e59d0ec6392299"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 13:31:01.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-08 09:47:36.000000000","tz":0},"subject":"drop infra-rename support","message":"drop infra-rename support\n\nWith one type of install type there is no need for renaming infra\nimages.\n\nChange-Id: I28b7563e5ce1efac0298a62fee96a9f946f3341b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e770353c35dce0e993f4afd9979d9d29a2aeafcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e770353c35dce0e993f4afd9979d9d29a2aeafcf"}]},"branch":"refs/heads/master","description":"Rebase"},"502fde1fb76ea4cabdedc87119bda43227d1ebec":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-08 09:54:49.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/76/836576/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/836576/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/836576/3"}}},"commit":{"parents":[{"commit":"9b8b22b8d4d5e89820e54e819ef7f7d310f0d040","subject":"kolla: drop \u0027binary\u0027 install type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b8b22b8d4d5e89820e54e819ef7f7d310f0d040"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 13:31:01.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-08 09:54:49.000000000","tz":0},"subject":"drop infra-rename support","message":"drop infra-rename support\n\nWith one type of install type there is no need for renaming infra\nimages.\n\nChange-Id: I28b7563e5ce1efac0298a62fee96a9f946f3341b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/502fde1fb76ea4cabdedc87119bda43227d1ebec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/502fde1fb76ea4cabdedc87119bda43227d1ebec"}]},"branch":"refs/heads/master","description":"Rebase"},"ad53e1fe22fb04623718635faced5d14a4d17abe":{"kind":"REWORK","_number":4,"created":"2022-04-08 15:55:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/76/836576/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/76/836576/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/76/836576/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/76/836576/4"}}},"commit":{"parents":[{"commit":"46d7f1dc10bf67f6dde3beaca59d1ef8b07acef9","subject":"Merge \"add releasenote about dropping binary images support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46d7f1dc10bf67f6dde3beaca59d1ef8b07acef9"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 13:31:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-08 15:55:26.000000000","tz":120},"subject":"drop infra-rename support","message":"drop infra-rename support\n\nWith one type of install type there is no need for renaming infra\nimages.\n\nChange-Id: I28b7563e5ce1efac0298a62fee96a9f946f3341b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad53e1fe22fb04623718635faced5d14a4d17abe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad53e1fe22fb04623718635faced5d14a4d17abe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
