)]}'
{"id":"openstack%2Fkolla~836577","triplet_id":"openstack%2Fkolla~master~I5d5a46c6ce7734ceb8b844e17b43e359d7cac6e3","project":"openstack/kolla","branch":"master","topic":"zed-remove-support-for-binary-images","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-04-08 14:31:54.000000000","reason":"Michal Nasiadka replied on the change"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-11 09:39:28.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-11 09:39:28.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-09 19:25:29.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I5d5a46c6ce7734ceb8b844e17b43e359d7cac6e3","subject":"docker: drop binary parts","status":"MERGED","created":"2022-04-05 09:20:48.000000000","updated":"2022-04-11 09:48:59.000000000","submitted":"2022-04-11 09:39:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":65,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836577-zed-remove-support-for-binary-images","meta_rev_id":"dac7f6d7f58289adeedc4368e7ae9482b68d5a15","_number":836577,"virtual_id_number":836577,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 09:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-11 08:12:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-04-09 19:25:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-04-11 08:12:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 11:06:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 14:17:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-04-08 16:34:12.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-11 08:12:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"1b707436e4feec9f8c13d61dd72bd838362e2eb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"157fa48d899ee5bec52b465526b6d314a271c1c6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:21:20.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"58c37a804e36b98a5692803539655a0813869a7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 11:06:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/58db39e222c84c2d9ccfe875476e3e19 : SUCCESS in 38m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/832f81bbbfb346d1a0064df60c4e097f : SUCCESS in 1h 31m 45s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2ded77b7a6734b1faf22ebecc968bf4b : SUCCESS in 1h 16m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/260928921e3e4c2da0b1ef4b7d143aed : SUCCESS in 37m 07s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b9034c71507d4b0fb3de96b872e173cd : SUCCESS in 59m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5eecd49b93044684b813e167dda1ec59 : SUCCESS in 1h 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2178888db0d746e1addaa5209e42e6db : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5afaea280aaf453188b757b6192153c5 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e05cb2442ca04eb0a816954e75cbc785 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4f8061df144fcf8a12c844f4e8217f : SUCCESS in 6m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c60250ef928442a9acc265a1ede3dff0 : SUCCESS in 39m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/208068fe3397472daf8016d12a815b94 : SUCCESS in 51m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e2fdfaf4d1e045bfbb60574ef273e489 : SUCCESS in 58m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3df2da6c8d99ff1f95adf8e1c492295620f50f2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 17:24:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4011b71fb86640ff84ab89203bfe7791 : SUCCESS in 2h 22m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f6b0a850e4b412b908891f39f362d65 : FAILURE in 42m 32s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6f7735a3e427415088fa84f17fdb9380 : FAILURE in 21m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef39faa353124934970346340f536d24 : FAILURE in 5m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cbff5eab08784cecad061b03b00ae57d : FAILURE in 5m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e4d5277ff10f48b4a8ef46edd95848f8 : SUCCESS in 2h 09m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb427975e3b67eaa3424c144e57d85446f937577","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-07 13:30:38.000000000","message":"Removed Workflow-1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz+kolla@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"849d10ad4269df3d9f0ac6e496fb07acaf9c6df9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:47:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5836efd0d4e9b58c145d3052d7e7c9dfc9976d41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:55:03.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"48bc05a499b14542ef426f4515388a2ee198c604","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 09:58:33.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"eb09f4d538519f189c9c5a74094965f3e5630e9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 11:47:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4c51e7c831e743a89131684ea99c93a9 : SUCCESS in 40m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/08af70ec64ec4c958256981db09b956f : SUCCESS in 1h 08m 09s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ef302763db90412fa88fa0e22b996113 : SUCCESS in 1h 43m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8a18c30d000744bc89a9713074d3c6e4 : SUCCESS in 32m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/18437afb16c14043bf040b11053b6b8b : SUCCESS in 51m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3c7c730cb8c641d5bf5bd8009becf08e : SUCCESS in 1h 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/157fe5056df04070946fbee1e2c5e9b0 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f842007a1cf34ae8b68e5c3f8d8aa8d3 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec72258a2a1d47cc9d1900c6e2444aaf : SUCCESS in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2d62d89b74c416d9ca307d98eb1891c : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ee3a8dc43c49438aa89c5234bea7f5 : SUCCESS in 5m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/94e72f090de543de82dda482488f013a : SUCCESS in 36m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aba8d7506a284f96be9108d5c207e3f7 : SUCCESS in 52m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1869f4c4cbb4428988f46eed5d4388f7 : SUCCESS in 57m 36s","accounts_in_message":[],"_revision_number":4},{"id":"b14bc56570e725af0c0692dbf22c2fecca7c31c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 13:42:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/505168c28e0d4e718824eec6b52149cb : SUCCESS in 2h 29m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/71590cad58ce4ce98c91d950c965878d : SUCCESS in 1h 46m 39s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3284caf2597147e5b682443672e1a242 : SUCCESS in 2h 09m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/759ca9debded49378dc0bc2b19a99ac6 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cee69ec7d6ee440e86c84c4984640691 : SUCCESS in 7m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9aae02a1b26f4f3a8bd35f2284657eeb : SUCCESS in 1h 47m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"131505b5ba24929aceb96397e752cb0f7da6eb00","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-08 14:17:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"74c9422e61ddfed1d5bee1c4322e002062f5a8d2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 14:24:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8a4cfb10d52eb8681b9a2e1fb2c9d837a9f7fbea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-08 14:31:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"eaff04c42db3c382d1cd69949ca41a71407fac15","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 16:34:12.000000000","message":"Patch Set 4: Code-Review-1\n\n(30 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3ff911a95e7c3c2c036b5dc6cd01706899f267c5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 16:34:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"46583934ba04534b09d771f526284189fd921c52","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 18:13:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/f751334587bc4d6ab3d22061927b575d : SUCCESS in 37m 51s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/0c999e2407974031aa1151cdc23aee95 : SUCCESS in 37m 36s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/113bdfda1e354c57abc052e32d580e83 : SUCCESS in 1h 19m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0d0671e85b92497bb81c8a93dbbaf55f : SUCCESS in 1h 33m 25s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/544936ea216544799c49513777f14d66 : SUCCESS in 1h 23m 38s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/078ccaba2f2b4fe19363c076f560dcd2 : SUCCESS in 1h 24m 59s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/cd1e3a43823b41d7910d3e6237292157 : SUCCESS in 1h 15m 11s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/a3cf1c36f34c4014a2f558fc3f8c77ae : SUCCESS in 1h 15m 34s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/c9a81d63de344b3abcc22a9a19975697 : FAILURE in 45m 23s (non-voting)\n- kolla-ansible-centos8s-source-octavia https://zuul.opendev.org/t/openstack/build/61969c01abd44d07990375bb448115f0 : SUCCESS in 1h 17m 42s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/fe0d57637303450f939fe462f0852fa6 : SUCCESS in 1h 32m 56s (non-voting)\n- kolla-ansible-centos8s-source-kvm https://zuul.opendev.org/t/openstack/build/6207cba1d7c44a7493842a2353a9ab43 : SUCCESS in 53m 56s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/17b64f0df39d4d3eb59bd401978b2696 : SUCCESS in 39m 30s\n- kolla-ansible-debian-source-ironic https://zuul.opendev.org/t/openstack/build/fdc1d495e3304fbaab39e805032edccc : SUCCESS in 1h 14m 34s (non-voting)\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ad8967c13fb49f5b07740f02c0047a9 : SUCCESS in 37m 14s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/cfa658a966f649bebb513a7147185527 : SUCCESS in 1h 23m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ad392d2c5d31453487d4b7e66bcda8d2 : SUCCESS in 1h 34m 17s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/ab164293b1d444e18231ff29dac929bd : SUCCESS in 1h 20m 43s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/e5664d45fbe14efca9777ca4034b2dd9 : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/3af8988ebd7944cf9036ea3e617d2887 : SUCCESS in 1h 09m 31s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/62858a91cba44c5e9fbf720f2823b0ef : FAILURE in 47m 59s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/00bc181fafc245c6adccb982a0c72162 : SUCCESS in 44m 09s (non-voting)\n- kolla-ansible-ubuntu-source-octavia https://zuul.opendev.org/t/openstack/build/1a4e1f6961314974a03c42dec7d5cdfa : SUCCESS in 55m 15s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/67cbd7bd0ce34c988ae476ea6c12c2ec : FAILURE in 54m 27s (non-voting)\n- kolla-ansible-ubuntu-source-kvm https://zuul.opendev.org/t/openstack/build/05145a54095346b39131a9a497939f7c : SUCCESS in 38m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9af6069995eb40aa0cd1cc16f4dbf27a3f0ffdde","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-09 15:43:46.000000000","message":"Patch Set 4:\n\n(27 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ca01da9f7f77b6ceb7f5e86372049d2ca2947afc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-09 15:44:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"32cfc45a08bd87c9784c20893dbfd02c2ab80e31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-09 17:13:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/223bbf1ddb974322af987487c482485a : SUCCESS in 37m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3d91f14c380f4558ba183f8bfdc01009 : SUCCESS in 1h 11m 21s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/def06a954efe439db602bef8553dd424 : SUCCESS in 1h 28m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f0990998038a4418a3f2b44dd61653e1 : SUCCESS in 35m 55s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a4a78e0042954068bc27c425fd7230f5 : SUCCESS in 1h 05m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c63f2ef983b94302b4c6d4153eae4024 : SUCCESS in 1h 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533e83759c3b40f69f145ad931815aab : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e7e53745e2241e38724e4805d506c71 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50ec7fa6ad8f4234a279a904ad610832 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8200df4e90ef47659be7ab9c9df79a06 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9008eec97a34a7b83d519268779f388 : SUCCESS in 6m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d772c5a2cb441ad8e27760b7e9ad29f : SUCCESS in 29m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/151c0537faf84b829a955535af84fb01 : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/172c567185164ff09b44fb513c162ee4 : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":5},{"id":"0f6aae33f02036ba05af3485857a9b320550077b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-09 17:52:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/7937323e4acc4c29abae068aaf1d5db2 : SUCCESS in 1h 50m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4acf32dd609640cab4a8178dd3257027 : SUCCESS in 1h 22m 27s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8a85fe372c8b4a8fb4150c3aae1c4b27 : SUCCESS in 2h 06m 01s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43d7810fc9eb4264aecf154ab80d9437 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25eb03e87dc64c819f1f15e9baf15eac : SUCCESS in 5m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e8ca3eb805d46e9b2f50b61558f0e79 : SUCCESS in 1h 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"58493a0b1f3f1272a66b7551ab3f2ac56dbebb87","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-09 19:25:29.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"97533b74b2d371750f4334c1fc55d2953ce105e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-11 08:12:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2bbcd584d0cf45b83781269ee94b124d98c10ae6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 08:12:31.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"796491b0b43dbaa13e924965a97f416854edb90e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 08:15:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7faaa1d97342778cc7172d2c94985b211faf125f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 09:39:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/44835295360f43c2a48ab2055a64b1f1 : SUCCESS in 44m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/205ef861811e4c5bb3a01b5288a2c545 : SUCCESS in 1h 10m 07s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/726c7a0dbd9d407fa6edfccb16f08b91 : SUCCESS in 1h 22m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/31e2f81ef5ca409e984436882efd61b0 : SUCCESS in 32m 40s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/91d57916af7247999c274546fcdfe499 : SUCCESS in 1h 21m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/12da1aa7159248bdb09530fae4957e95 : SUCCESS in 1h 20m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49a840c0d1fb44b09c049cfb83bbcd26 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b74897e301574f9dae13251700adc45e : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fc8750aff134f5dbea781f7554ce3c9 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/785133ccf8d2469590c7ec97ee9cb83f : SUCCESS in 5m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/84068a39d74943729d93d7105590c572 : SUCCESS in 23m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e1ec7f29b47649329de8ae5cb1e0073a : SUCCESS in 42m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/83baa8e03f464ba4873fa2e26ca68084 : SUCCESS in 49m 55s","accounts_in_message":[],"_revision_number":5},{"id":"8fe915516b9779ae78f3cda10368ff11f9481434","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 09:39:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"0a4225fedd0e95ba6a3d870a9ab5d1eec6a59fbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 09:41:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/924f217006b9492988d0063f30eb11a8 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":5},{"id":"dac7f6d7f58289adeedc4368e7ae9482b68d5a15","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 09:48:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1749da2fbf6b395b1a2cd77f349e0bea5d2f4a5e","revisions":{"72daa35221587a29db59328a38ece072ae9b3430":{"kind":"REWORK","_number":1,"created":"2022-04-05 09:20:48.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/77/836577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/836577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/836577/1"}}},"commit":{"parents":[{"commit":"263c634841f86d68769c54d370139fe610d9b09a","subject":"drop infra-rename support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/263c634841f86d68769c54d370139fe610d9b09a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 11:41:14.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-05 09:20:33.000000000","tz":120},"subject":"docker: drop binary parts","message":"docker: drop binary parts\n\nBig patch drops all mentions of binary images support. Suggestions are\nwelcome how to split it into parts or handle better.\n\nChange-Id: I5d5a46c6ce7734ceb8b844e17b43e359d7cac6e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72daa35221587a29db59328a38ece072ae9b3430"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72daa35221587a29db59328a38ece072ae9b3430"}]},"branch":"refs/heads/master"},"0e3e082901ad67e0e9a79ab0020efeba070ea086":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-08 09:47:47.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/77/836577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/836577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/836577/2"}}},"commit":{"parents":[{"commit":"e770353c35dce0e993f4afd9979d9d29a2aeafcf","subject":"drop infra-rename support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e770353c35dce0e993f4afd9979d9d29a2aeafcf"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 11:41:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-08 09:47:47.000000000","tz":0},"subject":"docker: drop binary parts","message":"docker: drop binary parts\n\nBig patch drops all mentions of binary images support. Suggestions are\nwelcome how to split it into parts or handle better.\n\nChange-Id: I5d5a46c6ce7734ceb8b844e17b43e359d7cac6e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e3e082901ad67e0e9a79ab0020efeba070ea086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e3e082901ad67e0e9a79ab0020efeba070ea086"}]},"branch":"refs/heads/master","description":"Rebase"},"84dee629ce0909248c7a688a3019e0406abce4fb":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-04-08 09:55:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/77/836577/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/836577/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/836577/3"}}},"commit":{"parents":[{"commit":"502fde1fb76ea4cabdedc87119bda43227d1ebec","subject":"drop infra-rename support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/502fde1fb76ea4cabdedc87119bda43227d1ebec"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 11:41:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-08 09:55:03.000000000","tz":0},"subject":"docker: drop binary parts","message":"docker: drop binary parts\n\nBig patch drops all mentions of binary images support. Suggestions are\nwelcome how to split it into parts or handle better.\n\nChange-Id: I5d5a46c6ce7734ceb8b844e17b43e359d7cac6e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84dee629ce0909248c7a688a3019e0406abce4fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84dee629ce0909248c7a688a3019e0406abce4fb"}]},"branch":"refs/heads/master","description":"Rebase"},"a59d191c0fde3cbc7c1f0d78cecc810012c249d5":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-04-08 09:58:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/77/836577/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/836577/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/836577/4"}}},"commit":{"parents":[{"commit":"9b8b22b8d4d5e89820e54e819ef7f7d310f0d040","subject":"kolla: drop \u0027binary\u0027 install type","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b8b22b8d4d5e89820e54e819ef7f7d310f0d040"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 11:41:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-08 09:58:33.000000000","tz":0},"subject":"docker: drop binary parts","message":"docker: drop binary parts\n\nBig patch drops all mentions of binary images support. Suggestions are\nwelcome how to split it into parts or handle better.\n\nChange-Id: I5d5a46c6ce7734ceb8b844e17b43e359d7cac6e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a59d191c0fde3cbc7c1f0d78cecc810012c249d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a59d191c0fde3cbc7c1f0d78cecc810012c249d5"}]},"branch":"refs/heads/master","description":"Rebase"},"1749da2fbf6b395b1a2cd77f349e0bea5d2f4a5e":{"kind":"REWORK","_number":5,"created":"2022-04-09 15:44:32.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/77/836577/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/836577/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/836577/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/836577/5"}}},"commit":{"parents":[{"commit":"cabfd0f3eff38e2c10cdc7cf4e9271d72eeca91a","subject":"Merge \"build watcher images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cabfd0f3eff38e2c10cdc7cf4e9271d72eeca91a"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-01 11:41:14.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-09 15:44:26.000000000","tz":120},"subject":"docker: drop binary parts","message":"docker: drop binary parts\n\nBig patch drops all mentions of binary images support. Suggestions are\nwelcome how to split it into parts or handle better.\n\nChange-Id: I5d5a46c6ce7734ceb8b844e17b43e359d7cac6e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1749da2fbf6b395b1a2cd77f349e0bea5d2f4a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1749da2fbf6b395b1a2cd77f349e0bea5d2f4a5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
