)]}'
{"id":"openstack%2Fkolla~836590","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I72d7920acd8d15941c8c57a4186186212b273a38","project":"openstack/kolla","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-06 10:47:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72d7920acd8d15941c8c57a4186186212b273a38","subject":"elasticsearch: install Java first on CentOS too","status":"MERGED","created":"2022-04-05 09:39:27.000000000","updated":"2022-04-06 10:48:27.000000000","submitted":"2022-04-06 10:47:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836590","cherry_pick_of_change":836449,"cherry_pick_of_patch_set":1,"meta_rev_id":"dae1ef32bd691d0207a3380ea08e6bc6a44f5377","_number":836590,"virtual_id_number":836590,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-06 10:46:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-06 09:08:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-04-05 16:28:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2022-04-06 09:08:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 11:02:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 16:28:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-06 09:08:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"e59cb8096850588e018eed4e41edf5d994e3bac9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:39:27.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"7932ea6d1f535e9889ce23ad378e232f4cb35616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 11:02:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/693f1a2a7a3447aca9f4d7217b36c5ce : SUCCESS in 29m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e98580e855d54a7f9d32fe900f84287b : SUCCESS in 51m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d9d8a48986e243e1a57ccbd350968836 : SUCCESS in 1h 02m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a237028d777f48de8f2860aae3fb9812 : SUCCESS in 1h 01m 19s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6e34f169b2714041a8139538e82463f7 : SUCCESS in 1h 14m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/54233ced50e548fdba8d1144771f2877 : SUCCESS in 40m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d018e5035a9e401eb1ac37797ebd4e8c : SUCCESS in 22m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/404063f9637d4a559cb749e86c6407f7 : SUCCESS in 50m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3346c06cde648ba9a762848ec77a10b : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47c1a80bee524bb5ab3dbb99a3f42b92 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c1601f6d03d427584b200067f143c5d : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61828cceb8174c278032886af4c171f1 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee59df83be3e49839855c80c6b2c228a : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8a18ebe0037949199825554ea545bbad : SUCCESS in 20m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cb882c58316b4739b8837b19d83008b7 : SUCCESS in 37m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a81d84789284da0a158566e613e2347 : SUCCESS in 55m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bca4b697561d4bb9823125b1ed47b128 : SUCCESS in 55m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eea1a5719c5a4f25916e0a9b28889968 : SUCCESS in 55m 19s","accounts_in_message":[],"_revision_number":1},{"id":"db46cc5ac18d212ca3b6eea7b690373d2ae0cd47","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-05 16:28:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5e9dd7487268d427510c07cf3b03945102c1dad5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 19:12:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8de31dc1b51a4bfea0ccff883b7ae9d9 : FAILURE in 1h 23m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/32cae0ca93074a68b5199754bd06c66e : SUCCESS in 2h 12m 20s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/05cc63e45ae74d28a5b77df9dffb3e2c : SUCCESS in 2h 52m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/59167eb8766b49fdb7d4eea9ba5477bb : SUCCESS in 2h 37m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91a8d64a0ca3a92bcc3c8541343e728a29e54fcd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-06 09:08:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"633c180dd55bbcdff601d03da315ffb6f77c06c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 09:09:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8f5180d7a02e8f466004c9d5b12fcd9c3e47f50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 10:46:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/64036950c9634f14bffef2d6d4bbbd37 : SUCCESS in 46m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4ccb418e73a647dda93e73624119c2da : SUCCESS in 1h 07m 18s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/31915f7267f74235a787a9ba0a378c42 : SUCCESS in 1h 23m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6135bd5892d1405f83318e3a277faf09 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e54e1dbe2a6342bc981f7829bd776849 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44e830d6c10d47bd9dc197621b8be02a : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ff2ffa248b044cfa526bce4f5416e24 : SUCCESS in 6m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/89fbc798c7ca46519d50552b020c9acd : SUCCESS in 33m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e057ab9b34644ccae46f480063372ea : SUCCESS in 52m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2e51d80a1e574bc598ca390669fb1072 : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":1},{"id":"83e5f0140c566fbda22f9c045d3a725bc5d6f970","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 10:47:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dae1ef32bd691d0207a3380ea08e6bc6a44f5377","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 10:48:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/387a47f6490c4865b1b6f603186f22fd : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a39d8e3d87e9bc0f658fe4472b88bcceb074858","revisions":{"6a39d8e3d87e9bc0f658fe4472b88bcceb074858":{"kind":"REWORK","_number":1,"created":"2022-04-05 09:39:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/90/836590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/836590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/836590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/836590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/836590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/836590/1"}}},"commit":{"parents":[{"commit":"cba638f793299e008a0ee5ab8d24eed566df0d89","subject":"Merge \"Fix disabling of curlrc in healthcheck_curl\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cba638f793299e008a0ee5ab8d24eed566df0d89"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-04 11:35:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-05 09:39:27.000000000","tz":0},"subject":"elasticsearch: install Java first on CentOS too","message":"elasticsearch: install Java first on CentOS too\n\nelasticsearch-oss package fails to install if we install it with Java:\n\nINFO:kolla.common.utils.elasticsearch:could not find java; set JAVA_HOME\nINFO:kolla.common.utils.elasticsearch:error: %prein(elasticsearch-oss-0:6.8.23-1.noarch) scriptlet failed, exit status 1\nINFO:kolla.common.utils.elasticsearch:\u001b[91mError in PREIN scriptlet in rpm package elasticsearch-oss\n\nBackport down to ussuri needed.\n\nChange-Id: I72d7920acd8d15941c8c57a4186186212b273a38\n(cherry picked from commit 976465c448c4de2bd323021b946770b26ab1e912)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a39d8e3d87e9bc0f658fe4472b88bcceb074858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a39d8e3d87e9bc0f658fe4472b88bcceb074858"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
