)]}'
{"id":"openstack%2Fkolla~836591","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I72d7920acd8d15941c8c57a4186186212b273a38","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-04-19 10:34:52.000000000","reason":"Michal Nasiadka replied on the change"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-19 12:14:48.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-19 12:14:48.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-04-19 12:14:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I72d7920acd8d15941c8c57a4186186212b273a38","subject":"elasticsearch: install Java first on CentOS too","status":"MERGED","created":"2022-04-05 09:39:34.000000000","updated":"2022-04-19 15:48:32.000000000","submitted":"2022-04-19 12:14:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836591","cherry_pick_of_change":836590,"cherry_pick_of_patch_set":1,"meta_rev_id":"9fa45a2f53d0091b648c28452b595b6793ba27c5","_number":836591,"virtual_id_number":836591,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-19 12:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-19 09:31:08.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-19 09:17:03.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-19 10:34:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 11:09:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 16:28:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-19 09:12:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-19 09:17:03.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"6b715fde8819985f8cc4b3394358153eb26c5831","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-05 09:39:34.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"4dd0d14928c0e2094e9aa180eb29cd91f1077a0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 11:09:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b822874d366344708c322dfb87368d9f : SUCCESS in 46m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f5b70d3a0a6a4d929f9e494cde4b40c5 : FAILURE in 48m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4167fa9a4a7c46dbbc2883317613975b : SUCCESS in 1h 22m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/de0c16c8dcc54bc29e814366796f1051 : SUCCESS in 1h 22m 35s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/97badce128d846ada952bf780994248a : SUCCESS in 1h 03m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a1bced6c6d1f407a98a651ce76865b85 : FAILURE in 3m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a20939b7f27b4f178a4d41013f836cda : FAILURE in 4m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e2c2bab84b444619a82f9ce085b98e10 : FAILURE in 7m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee4117ffb4aa4800a31a8ab25aa6e23b : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb4230528084403db09c8f0a00d7527e : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3dc989b5a46412cb353d55438c2d173 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc2080a5131745f589db892841ecec33 : SUCCESS in 6m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/df4e2952de8b4e66825988c0993dd587 : FAILURE in 5m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7cf51ece4c654056a7fbd7b5e8449445 : FAILURE in 4m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f83fbe88c6eb46d191b0294f0a26e376 : FAILURE in 8m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/833ee4ef54824e15aff8d0661b6c1ca9 : FAILURE in 8m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8459b48a78714095bbffb052e94ea04c : SUCCESS in 1h 05m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5a4f4a886b42c131aa5272e917ac3b3ce75d8144","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-05 16:28:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8a24446c68ab0d5ade7571d69f065f8b3c6aad1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 17:35:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/8e9fcd3dbd304250859688b38779c582 : FAILURE in 1h 11m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f09f6eefec684853ac24f1b7be1b17cb : FAILURE in 7m 13s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7515fdb5de6242b1988428afba99d64f : FAILURE in 14m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2a2ecfab7583462a977f27342def68fc : FAILURE in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"468e1656baa0202cbe0578a75cd6f472293403ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-19 06:58:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6ce257139b3bef6ab14a41c3df549483c3a5bcf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-19 07:25:15.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"49b942f830723126bb548e2ce6b14ec3c9d5d386","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-19 07:27:16.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"032a6e1d631b3dbc7b9caa903698e85f48a91f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 08:57:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9d1706f0338f404a8e05e775c3185fbb : SUCCESS in 52m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f9491d4a601e4e54a5270751195317fd : SUCCESS in 29m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8b3e3273f2d94266aa5b65d99e822141 : SUCCESS in 53m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e32609e4ab014d489721bc3c92d46f12 : SUCCESS in 1h 24m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/852ab86e715f40df87936b9dd50a586d : SUCCESS in 1h 08m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3e4dfde0c9304be58f4d117d5924b79d : FAILURE in 18m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0539e0e2ad25408ca1736b01c6cdeee0 : SUCCESS in 16m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/140fe481ef384602a474c2ece548b231 : SUCCESS in 55m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/474a4cfef5764dccb74edbe865f2502a : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13de5a9fa6404bcba3edc67cc65053c7 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07688a554d1240ffae04f88c7cd0c97c : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a08bbcd008a7426f94860512d3863b78 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a00dd1fb0e2b4effb1572bb93422d60e : SUCCESS in 4m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/708b858f24f847ef90a6c73539d1020e : SUCCESS in 38m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b139832bb58d4347840d79f3fde7c9ce : SUCCESS in 38m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ca3f727c3e241ca8bc82fea00513850 : SUCCESS in 1h 07m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3f4746ae64614408853d0fa960916ff9 : SUCCESS in 1h 06m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/43aef5f59c824c4492f790ac9ff60807 : SUCCESS in 51m 44s","accounts_in_message":[],"_revision_number":4},{"id":"388b0ca1d486cdeb8cf275e82033a60e0a994b2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 09:11:39.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"fad0ed05eb77e07853a440db9da52575ebd19054","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 09:12:09.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0faac668a295d2dc09db6d3a0e1343708ffaa3f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 09:16:49.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"8b064fc7abdd1335c4e65e32bd785da3e146c3a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 09:17:03.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"3f45a961921139dd3028b83b3209b6c9bd9fb09f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 09:31:08.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"71e5d2c192258152fc5859c26293fd154d82fd9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 10:28:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/776cafb1f66d41a3b7020d899608a71e : SUCCESS in 35m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d8e75d92628d4916896077a2f3d9d9f9 : SUCCESS in 27m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a98ce039a1944a3e86940cb5863e9164 : SUCCESS in 59m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/49924bfed569487f97baf733c3927c78 : SUCCESS in 59m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/94c72702248b48b699ae2fb7b8facb42 : SUCCESS in 1h 09m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/86a483d38bbf4f0e8f043dfaeb92ba73 : SUCCESS in 17m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/43d22f385c344ff398cf55344282ac26 : SUCCESS in 20m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7b12edcf915c4f0c96d0869feca8fc10 : SUCCESS in 51m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b87748db3cfe46de8e37c54655ac0f6b : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7f643ec037842a9a29dc97bb6fad6e6 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91e4bcd57ece41edab2aa8456f0489ff : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a37fd14624f7427e8b79a6f5ba7f33ca : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/034ffe5b1a424ee5bbc089719ee5426e : SUCCESS in 4m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6019b550a434e3490b963db8f117469 : SUCCESS in 20m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4844930947f4564b9df038e7c0b0329 : SUCCESS in 23m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/67d7ea2d4b8445c486cf745f5c4e08df : SUCCESS in 48m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/234335150b8449d8b5e5e3800dbc8d4d : SUCCESS in 42m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cfc90e16c556443282e15e518d91c9c6 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":6},{"id":"aac17bbdf093c5a119ca0b8dd2e91077d8b422ed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-19 10:34:52.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"13cd19a83ae79fda33742a17423875078d1670e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 10:35:47.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"faea9e0dfeec6a29f99d17aa847c086ce011282e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 12:14:47.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fde004d3837c49f2a91d7e89cdf447d8 : SUCCESS in 54m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/746956f184fe4a28affb162ce90c11d1 : SUCCESS in 1h 20m 07s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3116f5ac1ac941a5a88e3935ff1f3734 : SUCCESS in 1h 38m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/543f2b34c258498cb87e17632b37113d : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6295cce70fcb4d0dbc6a9d6c98ae5404 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6bf98d809574d93919317d57096297b : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787b980e312c4fff9fd3d81fc07ff474 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc8ac331d9194692b32262f4e41b46b9 : SUCCESS in 4m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e28eb4deff2946dbaefe0b36fc4d9300 : SUCCESS in 21m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/394aef3e340941cab953dfb289a7489a : SUCCESS in 51m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5265b4a32b1f4482919272091cf43bee : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":6},{"id":"be7a0d8c7e4b107e63ef62003d5fff18475c4c4c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 12:14:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"22c0cd29afe629fa759eba0779948154ba970e39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 12:16:20.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a1a782f4a3547b18bc11cbbaea2a443 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b56d1c4f4844aa4abd2a87f1f9c7668 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":6},{"id":"9fa45a2f53d0091b648c28452b595b6793ba27c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 15:48:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ad78caf7485c4cd480493091c468ef2f : FAILURE in 53m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5dd31ca6711a4a84a85f7983bb3c6a13 : SUCCESS in 1h 29m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f7075d86733b48098b221e8fea1401f1 : TIMED_OUT in 3h 03m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f1cd10153c44d2e8d8c28dabefa96f9 : SUCCESS in 1h 43m 40s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"a7ff9f6c0d2d5af14a0c9bebd0ceb932bbfeffb5","revisions":{"a08f6da4dbefddb91ceebba4e6988c764f11ef16":{"kind":"REWORK","_number":1,"created":"2022-04-05 09:39:34.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/91/836591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/836591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/836591/1"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-04 11:35:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-05 09:39:34.000000000","tz":0},"subject":"elasticsearch: install Java first on CentOS too","message":"elasticsearch: install Java first on CentOS too\n\nelasticsearch-oss package fails to install if we install it with Java:\n\nINFO:kolla.common.utils.elasticsearch:could not find java; set JAVA_HOME\nINFO:kolla.common.utils.elasticsearch:error: %prein(elasticsearch-oss-0:6.8.23-1.noarch) scriptlet failed, exit status 1\nINFO:kolla.common.utils.elasticsearch:\u001b[91mError in PREIN scriptlet in rpm package elasticsearch-oss\n\nBackport down to ussuri needed.\n\nChange-Id: I72d7920acd8d15941c8c57a4186186212b273a38\n(cherry picked from commit 976465c448c4de2bd323021b946770b26ab1e912)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a08f6da4dbefddb91ceebba4e6988c764f11ef16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a08f6da4dbefddb91ceebba4e6988c764f11ef16"}]},"branch":"refs/heads/stable/victoria"},"0da43ac6cf3b542f0714efd59951e29cf9e001cb":{"kind":"REWORK","_number":2,"created":"2022-04-19 06:58:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/836591/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/836591/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/836591/2"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-04 11:35:01.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-19 06:58:10.000000000","tz":0},"subject":"elasticsearch: install Java first on CentOS too","message":"elasticsearch: install Java first on CentOS too\n\nelasticsearch-oss package fails to install if we install it with Java:\n\nINFO:kolla.common.utils.elasticsearch:could not find java; set JAVA_HOME\nINFO:kolla.common.utils.elasticsearch:error: %prein(elasticsearch-oss-0:6.8.23-1.noarch) scriptlet failed, exit status 1\nINFO:kolla.common.utils.elasticsearch:\u001b[91mError in PREIN scriptlet in rpm package elasticsearch-oss\n\nBackport down to ussuri needed.\n\nThis change also includes a backport of\nI7d4788f5b63fba24e3b2f9b15c16866ff811d83e:\n\nAlways use the distro-provided libvirt-python\n\nThis patch switches masakari-monitors image to follow nova-compute\nand ceilometer-compute images.\nThis will be used and required after [1] merges.\nUsage of libvirt-python from PyPI has already proven to be\nproblematic on CentOS Stream in our stable branches. [2]\nWith this patch we avoid those issues as well.\n\n[1] https://review.opendev.org/c/openstack/masakari-monitors/+/804913\n[2] https://review.opendev.org/c/openstack/kolla/+/797102\n\nDepends-On: https://review.opendev.org/c/openstack/masakari-monitors/+/804913\n\nNOTE(hrw): added removal of libvirt-python from upper-constraints as we have\nnever version of libvirt that Victoria used.\n\nChange-Id: I72d7920acd8d15941c8c57a4186186212b273a38\n(cherry picked from commit 976465c448c4de2bd323021b946770b26ab1e912)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0da43ac6cf3b542f0714efd59951e29cf9e001cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0da43ac6cf3b542f0714efd59951e29cf9e001cb"}]},"branch":"refs/heads/stable/victoria"},"312173d25f80ed09a5882e94b21bb482a189e6e4":{"kind":"REWORK","_number":3,"created":"2022-04-19 07:25:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/836591/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/836591/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/836591/3"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-04 11:35:01.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-19 07:25:14.000000000","tz":0},"subject":"elasticsearch: install Java first on CentOS too","message":"elasticsearch: install Java first on CentOS too\n\nelasticsearch-oss package fails to install if we install it with Java:\n\nINFO:kolla.common.utils.elasticsearch:could not find java; set JAVA_HOME\nINFO:kolla.common.utils.elasticsearch:error: %prein(elasticsearch-oss-0:6.8.23-1.noarch) scriptlet failed, exit status 1\nINFO:kolla.common.utils.elasticsearch:\u001b[91mError in PREIN scriptlet in rpm package elasticsearch-oss\n\nBackport down to ussuri needed.\n\nThis change also includes a backport of\nI7d4788f5b63fba24e3b2f9b15c16866ff811d83e:\n\nAlways use the distro-provided libvirt-python\n\nThis patch switches masakari-monitors image to follow nova-compute\nand ceilometer-compute images.\nThis will be used and required after [1] merges.\nUsage of libvirt-python from PyPI has already proven to be\nproblematic on CentOS Stream in our stable branches. [2]\nWith this patch we avoid those issues as well.\n\n[1] https://review.opendev.org/c/openstack/masakari-monitors/+/804913\n[2] https://review.opendev.org/c/openstack/kolla/+/797102\n\nDepends-On: https://review.opendev.org/c/openstack/masakari-monitors/+/804913\n\nNOTE(hrw): added removal of libvirt-python from upper-constraints as we have\nnever version of libvirt that Victoria used.\n\nChange-Id: I72d7920acd8d15941c8c57a4186186212b273a38\n(cherry picked from commit 976465c448c4de2bd323021b946770b26ab1e912)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/312173d25f80ed09a5882e94b21bb482a189e6e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/312173d25f80ed09a5882e94b21bb482a189e6e4"}]},"branch":"refs/heads/stable/victoria"},"7d1e07089323d2b2728439f36f11244fa3791374":{"kind":"REWORK","_number":4,"created":"2022-04-19 07:27:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/91/836591/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/836591/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/836591/4"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-04 11:35:01.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-19 07:27:16.000000000","tz":0},"subject":"elasticsearch: install Java first on CentOS too","message":"elasticsearch: install Java first on CentOS too\n\nelasticsearch-oss package fails to install if we install it with Java:\n\nINFO:kolla.common.utils.elasticsearch:could not find java; set JAVA_HOME\nINFO:kolla.common.utils.elasticsearch:error: %prein(elasticsearch-oss-0:6.8.23-1.noarch) scriptlet failed, exit status 1\nINFO:kolla.common.utils.elasticsearch:\u001b[91mError in PREIN scriptlet in rpm package elasticsearch-oss\n\nBackport down to ussuri needed.\n\nThis change also includes a backport of\nI7d4788f5b63fba24e3b2f9b15c16866ff811d83e:\n\nAlways use the distro-provided libvirt-python\n\nThis patch switches masakari-monitors image to follow nova-compute\nand ceilometer-compute images.\nThis will be used and required after [1] merges.\nUsage of libvirt-python from PyPI has already proven to be\nproblematic on CentOS Stream in our stable branches. [2]\nWith this patch we avoid those issues as well.\n\n[1] https://review.opendev.org/c/openstack/masakari-monitors/+/804913\n[2] https://review.opendev.org/c/openstack/kolla/+/797102\n\nDepends-On: https://review.opendev.org/c/openstack/masakari-monitors/+/804913\n\nNOTE(hrw): added removal of libvirt-python from upper-constraints as we have\nnever version of libvirt that Victoria used.\n\nThis change also includes a backport of\nI5efab66e487e06abd1a56af97d7e7caa1ebc880d:\n\nUse jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (filters.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change has been updated to also support Jinja2 2.x releases,\nsince the Wallaby upper constraints specify 2.11.3. In practice, most\nusers will not use UC to install kolla.\n\nCoAuthored-by: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: I72d7920acd8d15941c8c57a4186186212b273a38\n(cherry picked from commit 976465c448c4de2bd323021b946770b26ab1e912)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d1e07089323d2b2728439f36f11244fa3791374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d1e07089323d2b2728439f36f11244fa3791374"}]},"branch":"refs/heads/stable/victoria"},"15d3f1b3130e4e09cc470871162656359a9f1af6":{"kind":"REWORK","_number":5,"created":"2022-04-19 09:11:39.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/91/836591/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/836591/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/836591/5"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-04 11:35:01.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-19 09:11:39.000000000","tz":0},"subject":"elasticsearch: install Java first on CentOS too","message":"elasticsearch: install Java first on CentOS too\n\nelasticsearch-oss package fails to install if we install it with Java:\n\nINFO:kolla.common.utils.elasticsearch:could not find java; set JAVA_HOME\nINFO:kolla.common.utils.elasticsearch:error: %prein(elasticsearch-oss-0:6.8.23-1.noarch) scriptlet failed, exit status 1\nINFO:kolla.common.utils.elasticsearch:\u001b[91mError in PREIN scriptlet in rpm package elasticsearch-oss\n\nBackport down to ussuri needed.\n\nThis change also includes a backport of\nI7d4788f5b63fba24e3b2f9b15c16866ff811d83e:\n\nAlways use the distro-provided libvirt-python\n\nThis patch switches masakari-monitors image to follow nova-compute\nand ceilometer-compute images.\nThis will be used and required after [1] merges.\nUsage of libvirt-python from PyPI has already proven to be\nproblematic on CentOS Stream in our stable branches. [2]\nWith this patch we avoid those issues as well.\n\n[1] https://review.opendev.org/c/openstack/masakari-monitors/+/804913\n[2] https://review.opendev.org/c/openstack/kolla/+/797102\n\nDepends-On: https://review.opendev.org/c/openstack/masakari-monitors/+/804913\n\nNOTE(hrw): added removal of libvirt-python from upper-constraints as we have\nnever version of libvirt that Victoria used.\n\nThis change also includes a backport of\nI5efab66e487e06abd1a56af97d7e7caa1ebc880d:\n\nUse jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (filters.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change has been updated to also support Jinja2 2.x releases,\nsince the Wallaby upper constraints specify 2.11.3. In practice, most\nusers will not use UC to install kolla.\n\nCoAuthored-by: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nFinally, amended for Debian Buster to also skip the lower\nconstraint from masakari-monitors.\n\nChange-Id: I72d7920acd8d15941c8c57a4186186212b273a38\n(cherry picked from commit 976465c448c4de2bd323021b946770b26ab1e912)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15d3f1b3130e4e09cc470871162656359a9f1af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15d3f1b3130e4e09cc470871162656359a9f1af6"}]},"branch":"refs/heads/stable/victoria"},"a7ff9f6c0d2d5af14a0c9bebd0ceb932bbfeffb5":{"kind":"REWORK","_number":6,"created":"2022-04-19 09:16:49.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/91/836591/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/836591/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/836591/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/836591/6"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-04 11:35:01.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-19 09:16:49.000000000","tz":0},"subject":"elasticsearch: install Java first on CentOS too","message":"elasticsearch: install Java first on CentOS too\n\nelasticsearch-oss package fails to install if we install it with Java:\n\nINFO:kolla.common.utils.elasticsearch:could not find java; set JAVA_HOME\nINFO:kolla.common.utils.elasticsearch:error: %prein(elasticsearch-oss-0:6.8.23-1.noarch) scriptlet failed, exit status 1\nINFO:kolla.common.utils.elasticsearch:\u001b[91mError in PREIN scriptlet in rpm package elasticsearch-oss\n\nBackport down to ussuri needed.\n\nThis change also includes a backport of\nI7d4788f5b63fba24e3b2f9b15c16866ff811d83e:\n\nAlways use the distro-provided libvirt-python\n\nThis patch switches masakari-monitors image to follow nova-compute\nand ceilometer-compute images.\nThis will be used and required after [1] merges.\nUsage of libvirt-python from PyPI has already proven to be\nproblematic on CentOS Stream in our stable branches. [2]\nWith this patch we avoid those issues as well.\n\n[1] https://review.opendev.org/c/openstack/masakari-monitors/+/804913\n[2] https://review.opendev.org/c/openstack/kolla/+/797102\n\nDepends-On: https://review.opendev.org/c/openstack/masakari-monitors/+/804913\n\nNOTE(hrw): added removal of libvirt-python from upper-constraints as we have\nnever version of libvirt that Victoria used.\n\nThis change also includes a backport of\nI5efab66e487e06abd1a56af97d7e7caa1ebc880d:\n\nUse jinja2.pass_context instead of contextfilter\n\nThe contextfilter decorator was deprecated in jinja2 3.0.0, and has been\ndropped in 3.1.0. This results in the following warning, and failed\nattempts to use filters:\n\n    [WARNING]: Skipping plugin (filters.py) as it seems to be invalid:\n    module \u0027jinja2\u0027 has no attribute \u0027contextfilter\u0027\n\nThis change switches to use the pass_context decorator. The minimum\nversion of Jinja2 is raised to 3 to ensure pass_context is present.\n\nThis change has been updated to also support Jinja2 2.x releases,\nsince the Wallaby upper constraints specify 2.11.3. In practice, most\nusers will not use UC to install kolla.\n\nCoAuthored-by: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nFinally, amended for Debian Buster to also skip the lower\nconstraint from masakari-monitors.\n\nFinally, finally, includes also hotfix for contextfunction.\nIe4bccb9ed3f4d8782730c5929abbfa73da215a8c\n\nChange-Id: I72d7920acd8d15941c8c57a4186186212b273a38\n(cherry picked from commit 976465c448c4de2bd323021b946770b26ab1e912)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ff9f6c0d2d5af14a0c9bebd0ceb932bbfeffb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7ff9f6c0d2d5af14a0c9bebd0ceb932bbfeffb5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
