)]}'
{"id":"openstack%2Fkolla~837114","triplet_id":"openstack%2Fkolla~master~I60d50ba7699d46d001eaef365fb84b0ce8e9b262","project":"openstack/kolla","branch":"master","topic":"zed-remove-support-for-binary-images","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-11 09:21:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I60d50ba7699d46d001eaef365fb84b0ce8e9b262","subject":"drop infra_image_prefix","status":"MERGED","created":"2022-04-08 10:39:29.000000000","updated":"2022-04-11 09:23:46.000000000","submitted":"2022-04-11 09:21:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837114-zed-remove-support-for-binary-images","meta_rev_id":"47e97ab798dc7fcd11841ff95ee82b8dc5b812b0","_number":837114,"virtual_id_number":837114,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 09:21:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-04-11 06:24:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-04-11 09:23:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-11 09:23:46.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-04-11 06:24:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-08 11:39:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-08 12:08:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 14:27:23.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-11 06:24:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"260c5ba2bc68539fca0f27f0d20763e10be6e8d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 10:39:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b277e82638d81206ea0d3b6fd3783975203a57fc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 11:39:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"557414606ba878aaef31698c5f9f11279288cfb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 12:08:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a275672ed5184301b3db1571418a70e1 : SUCCESS in 48m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1e34230d87ae4f0aa56078dfd9e28cb7 : SUCCESS in 1h 01m 24s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1a6c8d784c484d56b48b0f3f3f66fa77 : SUCCESS in 1h 24m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/949468851141491896a0409bd0644d2e : SUCCESS in 33m 46s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7e912df8513547d1a5029713a9af1712 : SUCCESS in 1h 11m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/dc39c03361794e5a97fe901119f6b520 : SUCCESS in 1h 13m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2960ff85639439f8bcaa58f734c9614 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d3528def94342159b9675ac17206401 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6239cd13e9442f4bb94e056dcbf9dd2 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fd644fdfe7e498e9f8418aaf05b4bad : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e752ad0d3ac4f33ba202b93bcf07d2f : SUCCESS in 5m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0ea77bf3536474fa85339921ced58a2 : SUCCESS in 32m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca74e6924f9941829540486c12993a0c : SUCCESS in 48m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a6995ba629334b818a8f19001ae100fd : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":1},{"id":"17cb0543775620a4b854a85296a7f15bf6b29a51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 14:10:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/03c19f4dd5174a3daf80e874a87b631e : SUCCESS in 2h 05m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6bdbf337f1214b9cab795ece4254d119 : SUCCESS in 1h 51m 56s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd4572f06f55457291a21eaef1a92feb : SUCCESS in 2h 09m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f85b8d3178f6467b82bf7961d69ecd84 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eecf7abc190b4bf4af80e85bcfe1594a : SUCCESS in 6m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bdbca5b1e6a241c793143aa542d8b5f3 : SUCCESS in 1h 46m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58ce24415ec4ed76cb4b49a581156c6b942acef5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-08 14:27:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3c456d5da50639e10bfdb0afece16a8a514f9ae2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-04-11 06:22:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9f0d9a887f9b1662bb820c6634176695ca6621f4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-04-11 06:24:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"68c83f18d6e730b30ed2bed9af0760ebba116929","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 07:45:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2c19658ef1284059a55e063d0efb2a25 : SUCCESS in 40m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2f5f110a9c2441d481ece3ba4bdafda2 : SUCCESS in 1h 07m 20s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3d0b345334c84ac9bd4f7b83b60b431a : SUCCESS in 1h 19m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dd1c600b082041abbd5045d3d8080ecd : SUCCESS in 25m 39s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2a1e9ce61ad64cdb86e50a54372a682b : SUCCESS in 1h 06m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3a449c2f37d241e6bb7afb949e4a89b9 : SUCCESS in 1h 06m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87d15f2d6c0e4eddabf4c5f03dbe8e39 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7d881c88bd84684a0e6594340c64871 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b237588a41e241c2bf3e5ff6dffa2e9c : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9e66e1760324e1c9f0f56d0df50247b : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1844fac92a49449eb8a356543c3d2453 : SUCCESS in 6m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e991a89ea554e74aea23660fa837582 : SUCCESS in 51m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/792b44ec16b14bbc9f0794aa598a068d : SUCCESS in 1h 02m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a0e8760589f145ffa5ffea4d141f8def : SUCCESS in 1h 13m 21s","accounts_in_message":[],"_revision_number":2},{"id":"f37374b6783ff22052a6ece8e764cdedb4752f8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 07:46:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e7bc04b3f369a7d8708ac177718b15805a042756","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 08:31:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e602eaa7c53d4295bf2bb87e594e6835 : SUCCESS in 1h 54m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/15ce73ddf6cd4558ad1217d8aa8505ff : SUCCESS in 1h 28m 01s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8aff956233714e969cc96c7b6809bd97 : SUCCESS in 2h 06m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a567ef15cc304af3b487e7c02844fb76 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26f107d5d90b46ddbe6da866ed600319 : SUCCESS in 6m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/401fc4ec5f814f958e8088978de4e589 : SUCCESS in 1h 44m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"097a96f0524917fab2e95b4f39eecc38f2b0fa39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 09:21:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8d16bd704a54492e9f9b08bca13ba06e : SUCCESS in 31m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/819af88395244876b5f2b977818ca95e : SUCCESS in 1h 19m 17s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f2e0f2c3e3d942fe858bb0fe2808c383 : SUCCESS in 1h 30m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1068f5b7ceb24f86a58088f6a2171f6e : SUCCESS in 32m 32s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8d1a672a47c94d5fab4b431d55fcff5c : SUCCESS in 1h 06m 08s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f164c394795b483684572b39a5268405 : SUCCESS in 1h 13m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa0dcc029ade40fab11efd58aed06182 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4f9d3c9175440af8c2b387ccd75d6dd : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/958b33a261ce496894b556154de2be74 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30b71af040ea4d7f82bdce40a7641e8d : SUCCESS in 5m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/36506cb7621a4cbb99db3adc7c5de139 : SUCCESS in 29m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/627d7464495542389bcb485f6093c163 : SUCCESS in 44m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a1a4245c9fdf40a08df89e72828d93e8 : SUCCESS in 1h 05m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a95de8c7f00d8a4d100cba0eafbc5a2fa30f8495","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 09:21:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f8d9083efcaddf6ecb63be01a5c980bdc75111e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 09:23:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/108bf3b0926c4be9af6539522ad2917b : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a1510870e86f9cde6537ac2cbbfb055337843c90","revisions":{"bcd833af17cbc9ec7c7092c6a09c3def410e6f15":{"kind":"REWORK","_number":1,"created":"2022-04-08 10:39:29.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/14/837114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/837114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/837114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/837114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/837114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/837114/1"}}},"commit":{"parents":[{"commit":"502fde1fb76ea4cabdedc87119bda43227d1ebec","subject":"drop infra-rename support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/502fde1fb76ea4cabdedc87119bda43227d1ebec"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-08 10:38:13.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-08 10:39:20.000000000","tz":120},"subject":"drop infra_image_prefix","message":"drop infra_image_prefix\n\nWe have one install_type only now so that infra stuff is not needed\nanymore.\n\nChange-Id: I60d50ba7699d46d001eaef365fb84b0ce8e9b262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcd833af17cbc9ec7c7092c6a09c3def410e6f15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bcd833af17cbc9ec7c7092c6a09c3def410e6f15"}]},"branch":"refs/heads/master"},"a1510870e86f9cde6537ac2cbbfb055337843c90":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-11 06:22:21.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/14/837114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/837114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/837114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/837114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/837114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/837114/2"}}},"commit":{"parents":[{"commit":"ad53e1fe22fb04623718635faced5d14a4d17abe","subject":"drop infra-rename support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad53e1fe22fb04623718635faced5d14a4d17abe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-08 10:38:13.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-04-11 06:22:21.000000000","tz":0},"subject":"drop infra_image_prefix","message":"drop infra_image_prefix\n\nWe have one install_type only now so that infra stuff is not needed\nanymore.\n\nChange-Id: I60d50ba7699d46d001eaef365fb84b0ce8e9b262\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1510870e86f9cde6537ac2cbbfb055337843c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1510870e86f9cde6537ac2cbbfb055337843c90"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
