)]}'
{"id":"openstack%2Fkolla~837141","triplet_id":"openstack%2Fkolla~master~Iea1cf9c539a73da625e335d6c1adb6cd71ec93d4","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-11 14:50:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea1cf9c539a73da625e335d6c1adb6cd71ec93d4","subject":"Drop qdrouterd image","status":"MERGED","created":"2022-04-08 14:48:25.000000000","updated":"2022-04-11 16:22:21.000000000","submitted":"2022-04-11 14:50:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837141","meta_rev_id":"a402f3a6ccfda83becdfd3b2bb0ca8020075cf4a","_number":837141,"virtual_id_number":837141,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 14:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-11 10:22:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-11 10:07:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-11 10:22:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-08 15:03:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-08 16:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-11 10:22:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"a3f3a99e109f0d8547b7508e9a478dd3950355c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 14:48:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5de87c0417881c4aa99e6f77054a22f69f41d765","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-08 15:03:14.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8562aedcd1784f9b1f97caecabe621e0576367fe","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 15:12:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53221b84ed43aed026504a61dd54b0a5d7da66ac","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-08 15:13:38.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"72dba6619ac09f407717e5b2e4448da5d1da9dc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 16:05:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/284dab3c9e7c4840a131e82550bf8ed8 : SUCCESS in 38m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f5ada695538448019ae6bd4b6dd52b22 : SUCCESS in 1h 01m 04s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f1a0d2f2b05d4958bba2df490f0bcfc3 : SUCCESS in 1h 13m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c49d69bd605542568c7e620d05fa8f36 : SUCCESS in 23m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8be46f43ca4741bfa5d1da7b78b15de6 : SUCCESS in 49m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6d74ec3ab1f34568a0afc6696646edb2 : SUCCESS in 59m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9de99947e4c48e19807cd3d436371fd : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f793a2dd45d24e8690369ea562077c4e : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e43a7175c2424d788546546b4fb63b8e : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc5883e0810246158cb4feebba110d48 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e9c031f86e142bc8360b8daa87f049f : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77aa9f9a10e241d88471e975c96b11fe : SUCCESS in 3m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/178589f555c046dfa2258371c0766b7c : SUCCESS in 47m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e8ca20a707649f3bf8c6ffd8620325d : SUCCESS in 48m 24s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8a931247017a4650afd7411d3e0f0fed : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":1},{"id":"319ad28260d3b29ecbae836bd19fefb92408334c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 17:20:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f69d1e8df83a4467ae096d0b98feec44 : SUCCESS in 2h 27m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bb9cd7676066471989417661641986d1 : SUCCESS in 1h 51m 33s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb6b7115c2bd4a298746a5f3ef437016 : SUCCESS in 2h 29m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0271e45a9cf6446898b5ae0f5cc0714b : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48f51ab9d9df4029b2b2de390eb2977f : SUCCESS in 6m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6fe99b85ccca46e29a593ada56c03ada : SUCCESS in 2h 09m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12d097b4d16f3de5e33077cf05737be2ad048f25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 10:05:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fa2b2daf0a9aba45bc4f33260045d9ed67e0c6f0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 10:07:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9de74f61dd5efa7769c2a323a3188b111432cea7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-11 10:22:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"56f8e2b707f4c61913918f0b22157be0d0f75889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 11:30:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9126ebaca8224886817b838f812c3512 : SUCCESS in 37m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0cff6904b13c4053b7e0e759ef9bf0df : SUCCESS in 1h 08m 31s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1fb4828f12994368988b13f9bd219772 : SUCCESS in 1h 19m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5b65ecc0f8d74bd8b64fb2f68e87ae7d : SUCCESS in 32m 03s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7a5a4e41757c4da7b082afbe70f7020d : SUCCESS in 1h 06m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4c4510e225f04bc9ab3d12d7b8cadcab : SUCCESS in 1h 08m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f4da27166db423a99127d176530afa8 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ff29096fa4440b78e94ae0d5eab79f2 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3023fc71964445d28f1a0d8968d83b39 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bb792f7802244cc986d7305e159f84f : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbfd4cc27e714eb68c5960a1a6785e68 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/958d6b3a7f82459c9e44bd9544fbe921 : SUCCESS in 3m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0506014a3d734c06bce0232a6fcb04fb : SUCCESS in 26m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0eb428a72e1f4bd9aadcd15e74c3ee93 : SUCCESS in 44m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d2dae5a8ef7e4d6785a678d708498299 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":2},{"id":"07a6a609de642eb806475255c2a21eedad0f4051","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 11:30:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"da510342e439e6080e5c19aa29978c25a3711b6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:50:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fc23f351b7144632b87d1a87c4064a37 : SUCCESS in 42m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/887b0674d915424fa0a45a75cf4562ad : SUCCESS in 1h 14m 59s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bcaeb52974b946a487b2899f7f9fa818 : SUCCESS in 1h 27m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0f0a439df1cb48f08797afd85d6619d0 : SUCCESS in 39m 55s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9a1e73293d224880aad825bce6b99c56 : SUCCESS in 1h 07m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7de0af6ee3a848b784106701fc037d6f : SUCCESS in 1h 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69dde7d24b924ce385a0b6538c0ea9bb : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3cf35875ab04a9995a66d4e8a9731b4 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/627519ad341b425a8647908c8a0241d0 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/579ea6a5b00c4c09a4e684566efdfc5c : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33e8054b427c479193d6856602e53895 : SUCCESS in 4m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4d8abfd80d7489eafef058953123aea : SUCCESS in 45m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f17cca515ec479697ccea421bc7338e : SUCCESS in 55m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9249876a8a034913a793663cf8eacb2c : SUCCESS in 1h 08m 19s","accounts_in_message":[],"_revision_number":2},{"id":"939ea1396dba0b0d4c6d66a14b48510d04a91e7b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:50:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a402f3a6ccfda83becdfd3b2bb0ca8020075cf4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 16:22:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6ff9e5be351f4a869f5947d0af1053b8 : TIMED_OUT in 6h 13m 56s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7a2ecde97c05474590255d1bb606af1e : SUCCESS in 2h 04m 48s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/33f8433b911345ae8c971b3097935442 : SUCCESS in 2h 15m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4246ee454fc40d2a58c3837157f5d00 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4342c791252424ca19ffc01fdabdeb1 : SUCCESS in 7m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/707c63931ce042f8b178b470aebb95cf : SUCCESS in 2h 10m 02s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f4319f8247057bf0c418e738dc2ebadd13b01989","revisions":{"3c14cd2b26ce0f03ec7fda4de109f78e0159a0b6":{"kind":"REWORK","_number":1,"created":"2022-04-08 14:48:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/41/837141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/837141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/837141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/837141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/837141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/837141/1"}}},"commit":{"parents":[{"commit":"d8a7119009f50631045565940921b785e71c0c21","subject":"Merge \"CI: remove binary jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8a7119009f50631045565940921b785e71c0c21"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-08 14:45:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-08 14:48:16.000000000","tz":120},"subject":"Drop qdrouterd image","message":"Drop qdrouterd image\n\nIt was marked deprecated in Yoga.\n\nChange-Id: Iea1cf9c539a73da625e335d6c1adb6cd71ec93d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c14cd2b26ce0f03ec7fda4de109f78e0159a0b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c14cd2b26ce0f03ec7fda4de109f78e0159a0b6"}]},"branch":"refs/heads/master"},"f4319f8247057bf0c418e738dc2ebadd13b01989":{"kind":"REWORK","_number":2,"created":"2022-04-11 10:05:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/41/837141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/837141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/837141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/837141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/837141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/837141/2"}}},"commit":{"parents":[{"commit":"657820362dd0b52f8462054e043fd892d14fbe76","subject":"Merge \"docker: drop binary parts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657820362dd0b52f8462054e043fd892d14fbe76"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-08 14:45:05.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:05:32.000000000","tz":120},"subject":"Drop qdrouterd image","message":"Drop qdrouterd image\n\nIt was marked deprecated in Yoga.\n\nChange-Id: Iea1cf9c539a73da625e335d6c1adb6cd71ec93d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4319f8247057bf0c418e738dc2ebadd13b01989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4319f8247057bf0c418e738dc2ebadd13b01989"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
