)]}'
{"id":"openstack%2Fkolla~837284","triplet_id":"openstack%2Fkolla~master~I0d32808c73ab3292a8f213424118dfe2313d465b","project":"openstack/kolla","branch":"master","topic":"zed-remove-support-for-binary-images","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-11 14:51:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0d32808c73ab3292a8f213424118dfe2313d465b","subject":"cyborg-agent: no need to check for install_type","status":"MERGED","created":"2022-04-11 10:29:08.000000000","updated":"2022-04-11 18:17:45.000000000","submitted":"2022-04-11 14:51:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837284-zed-remove-support-for-binary-images","meta_rev_id":"d7b97852207cce3640a19c623f0fb3324d3b6450","_number":837284,"virtual_id_number":837284,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 14:51:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2022-04-11 10:35:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-04-11 10:43:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2022-04-11 10:43:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 10:35:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-04-11 10:43:02.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-11 11:56:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44f2c0c74c569f37108a05b0a1ee1b9e47e802d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 10:29:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40909eab3b030694e60daaa371d568bbb182a191","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-11 10:35:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c4a827b72a6a916943b838e37c37b357c9f4ba2b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 10:43:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dffeed6b2c2798146077d4da9cefb2d1a79fb7a0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:11:49.000000000","message":"Topic set to zed-remove-support-for-binary-images","accounts_in_message":[],"_revision_number":1},{"id":"167cf88af7d3dc40549d70d44b89b7c8de14811f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 11:56:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f301828c20454aad80cdb18dd1c5615a : SUCCESS in 39m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/57893fbaa7fe4ff68f6facc33232c4ca : SUCCESS in 1h 11m 14s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f5d8f171775e4ad5a66567126e81f99f : SUCCESS in 1h 22m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/deb90fa392464ce684d8961bdcfb5cb4 : SUCCESS in 30m 21s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ebb4f4bb59c84ef2baedabd3c743075e : SUCCESS in 1h 04m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/79504035227e4515a94e32c19d78efa0 : SUCCESS in 1h 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82e5ba193aa449bc9f01102aef88a40d : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f36ce311cac4428a7a8e0938e19b861 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80bd902b4a9f49d0b0cb0157c4a6af5b : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/807ce1d0cbf741cf849ddfe5fb2db286 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f730abfb1094cec966e987a75b09c5b : SUCCESS in 5m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/29cd1e3f25b248f3a5b86c22a79d2396 : SUCCESS in 34m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd1a42b4b3704d3185b79065e2267a54 : SUCCESS in 49m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7dccd37191a84f06a428b862304bfbd8 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2a312d405e5ad7d9b041f64418640395a4111236","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 11:57:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7913d44adf8e1f7b5871ab20e8061a901b9071ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:51:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/45f13a6d44d74f6b9922bc17b90e3323 : SUCCESS in 1h 02m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/427f5d365ca449ae8a082728f98292c0 : SUCCESS in 1h 13m 57s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2c079c55f8094dc3844238279f1a9907 : SUCCESS in 1h 10m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/83a6fb5f57f94dc8a7f463033fadd73c : SUCCESS in 38m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1707629e927d478db27f75d0a1b6160a : SUCCESS in 1h 01m 03s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f9c86f3e31794a709cc0b9d404f94d2f : SUCCESS in 1h 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f83379bdcc240d69950d1c7f79ee2f1 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae70f80a954f49b1adfe6f081963b4e5 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d8d0364f4dc405dbe46fe668be436d6 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/598fed6fa65348c19ff43a61d2abe938 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb122ae4263d44cfa3c4fde38d32690d : SUCCESS in 35m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4046396da89d4f3c84fd9d3711682200 : SUCCESS in 55m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2d2aa42712d740579011abe14449bd62 : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0fcd41656633032c913d62ac6f90ee08cc2b60ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:51:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d7b97852207cce3640a19c623f0fb3324d3b6450","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 18:17:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/70d95a0a898e4806b7aa8f259041ba1f : SUCCESS in 2h 14m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c88e60209091433787d575c9a91c7e22 : SUCCESS in 1h 33m 20s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/62abef0d6ab64a6eaa837155085ef171 : SUCCESS in 1h 48m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61d389283213427abc6b377e295aefaa : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2aaeed2a0b74ecc824c920d9bcc68ae : SUCCESS in 6m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2a50ca9d2ec84e139ab3e17b9d97d34f : SUCCESS in 1h 26m 21s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae8eada65860e9315f4e4d1f69e97258f8b22fc5","revisions":{"ae8eada65860e9315f4e4d1f69e97258f8b22fc5":{"kind":"REWORK","_number":1,"created":"2022-04-11 10:29:08.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/84/837284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/837284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/837284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/837284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/837284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/837284/1"}}},"commit":{"parents":[{"commit":"657820362dd0b52f8462054e043fd892d14fbe76","subject":"Merge \"docker: drop binary parts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657820362dd0b52f8462054e043fd892d14fbe76"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:28:34.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:28:34.000000000","tz":120},"subject":"cyborg-agent: no need to check for install_type","message":"cyborg-agent: no need to check for install_type\n\nThere is only source.\n\nChange-Id: I0d32808c73ab3292a8f213424118dfe2313d465b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae8eada65860e9315f4e4d1f69e97258f8b22fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae8eada65860e9315f4e4d1f69e97258f8b22fc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
