)]}'
{"id":"openstack%2Fkolla~837287","triplet_id":"openstack%2Fkolla~master~I6890a7ec7baf507f8acd75b42bbd0fb17c2e4511","project":"openstack/kolla","branch":"master","topic":"zed-remove-support-for-binary-images","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-15 10:25:59.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-15 10:25:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6890a7ec7baf507f8acd75b42bbd0fb17c2e4511","subject":"drop install_type almost everywhere","status":"MERGED","created":"2022-04-11 10:36:27.000000000","updated":"2022-04-19 19:01:37.000000000","submitted":"2022-04-15 10:25:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"837287-zed-remove-support-for-binary-images","meta_rev_id":"7fb0ebc178b6d5b4226d609d0d1b5f49371d34a2","_number":837287,"virtual_id_number":837287,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-15 10:25:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-15 08:42:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-19 19:01:37.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-15 08:42:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 10:41:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-11 12:25:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-13 08:58:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-19 15:29:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"}],"messages":[{"id":"a536e9612ec0557e119408600b1b862f2dc08afb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 10:36:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38d40b2fff6186c3eec2034d811c9d4ad5514146","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 10:41:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6070d4080ab12618f779e74b466468df3e52e314","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 10:58:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84a3f8bb49d47d976bc19dabb52ff09c3b17784f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:02:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83e46c921b3502bbfb86efd2a2867b9d6bf60c1b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 11:02:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbd5594ce49a5b89ba88f6f5142dfc8043663346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:02:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80c72dfde3de3ff355b8afec2c7a1093ca1ad593","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 11:05:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a66190a5b09fec5825c4732873a7a6e853b95ba","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:08:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0f7960f819804b55567c4d376a7ede3405e69f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:09:35.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7a1a2ba405267019c05d95d10282280a75b3e81d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:09:56.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a61aa992804657b6bb6788ae2ed4a0598f669542","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:11:38.000000000","message":"Topic set to zed-remove-support-for-binary-images","accounts_in_message":[],"_revision_number":3},{"id":"7b22d871127658bedd9ea1ce4d7b15dcc76eb895","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 11:19:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fc1a1b3b1fa81abbde0e75c3032d4677585363e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:25:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0e81397fe5ae4bbe87c26b3085249803 : SUCCESS in 46m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2ea748db1f5e478a9b27432b5e9f2d27 : FAILURE in 39m 44s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6c7393a0b046446c936e2e6d56bddf23 : FAILURE in 1h 13m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/955af5d0059b4fd2ae075c9b97084f37 : SUCCESS in 33m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eaf63191bba74799ad28150f1d3fb4fc : FAILURE in 31m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/69badf6c5e6a4cd882b4e1eaa08ca723 : FAILURE in 52m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/612f078585044c5e9223bfcc971181e6 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ba7e42786a84a12a446ca3df2b0580a : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbffbb10216843bfb9b151536ce506c3 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d87c5183c60d4f7489b7c8e598923321 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13dada0c826047ae9792f5236c10e166 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28bc54b8f7864d56ba899b8f2aff31bc : SUCCESS in 5m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf7694051ea448e2a54e910c37b058c5 : SUCCESS in 32m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0c775be4c1a49d3971fa90fc65582bf : FAILURE in 22m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d2f0620a90ec4e649902e280a2a40ad3 : FAILURE in 54m 58s","accounts_in_message":[],"_revision_number":3},{"id":"13fbc48e230b24a9dce791563a2b6b5f7e9be952","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 12:28:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d5e46c3abc9704d1ed47d059c613c2339ef13246","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 22:03:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/07c24267489d4df88b154ffedb14f4cb : SUCCESS in 2h 35m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea4af9fed38848d9bc79d5883c41cd84 : SUCCESS in 1h 29m 31s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/077b8411fa634f659ce7337c65604cd6 : FAILURE in 1h 17m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ebb8d2281d64e82bd3e808939cffc92 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c136d231d424a76bbb9e70992720e03 : SUCCESS in 5m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8cef4aa892b4ec68fe6c5e7fe32e4c7 : SUCCESS in 1h 50m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"665b4485a5fe6d9d1f6e6968c7affb204b083ed3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-12 13:13:10.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"a507946593e34852772b1a2309be81d1936fc568","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-12 13:13:15.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"aca22e2f53ededa97411cdfe7464faed89315817","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 13:26:00.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/7b8543b5ce104d688e4681ba42916dff : FAILURE in 8m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/68de4ea19ae542dcab98a31785fa7db5 : FAILURE in 7m 48s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f8321baa2fdd4892be1c1009e1e401dd : FAILURE in 9m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a606e9b78cc4bdfaade845e6322c5c5 : FAILURE in 6m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebf29164380c4a7aba593566f3d02692 : FAILURE in 7m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/92745ea9ed194f848e5b86b6588b3835 : FAILURE in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4bb1d261d10b27ffda3ca28ec7d9f16d33079aeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-12 13:27:23.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"1538c026890147371621f7dc21d425e1c3ec5666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 14:37:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/b50247754a7740b6a49e5ce99fc9afb5 : SUCCESS in 40m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9c50000d3c264a48a2ee5a794aaff840 : FAILURE in 7m 59s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b6f5ff315ca94e1386cba1c4b3dc984a : FAILURE in 58m 01s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/804f24f9d6ff497fbb15042540927288 : SUCCESS in 33m 33s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a58f0548b95d43e5b0835d414f8d5b61 : FAILURE in 6m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/70605f5a405f48f6827618a9ed86ddf0 : FAILURE in 57m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deca9cc798bc496fbe1d29443c7edcec : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77fde07837e24219a1b7cfa2533fc244 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24b9fd5eba754f3db171a880a8e24a2d : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a00af7df3734b1e8a47e55d22c3260e : RETRY_LIMIT in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dce9f9afa2174781953effcbcf63f7dd : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06141022ae8f4b78a814740e763e9731 : SUCCESS in 5m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b106a09835074a76a7d8febc6a826540 : SUCCESS in 36m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a6770e7b51e4ed9bb11e86f367249fd : FAILURE in 5m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f3e4017963f141a89d232f9186c62463 : FAILURE in 47m 30s","accounts_in_message":[],"_revision_number":6},{"id":"a59fe4d5812e825a61d26a4fb1d27ee4ccdf340f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 15:39:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/79fd9a215dbe4f7f880a2d2d6fb5a75f : SUCCESS in 2h 09m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/30432e988bfa40c3929ae39f1f4977eb : SUCCESS in 1h 36m 04s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8bf4a2fc1a034cdbb52bf2abd410eeae : FAILURE in 10m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9fced45eba164fb08e9c544bbe88517e : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be8e364569dd476d9e5f5022c18ee713 : SUCCESS in 6m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1d8964d733f34272ac641a840d8d01b6 : SUCCESS in 1h 40m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2ac2b7bf0fb250078f68d53a6aff0f14104ae448","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-12 17:11:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d615348c89affa8d70e4b5ae850fb18204f2a1d6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-12 18:22:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"27e946997c1f81e70275b03a94b7915728fa182b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-12 18:26:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"96ae2ae5197182e79829a9f86fe5f9a59e5dd57a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-12 18:27:18.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7b619e620fec2cc54abdedf81b3cd5eb1b10c430","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-12 18:27:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8cacf6fe5d0b590bb6bb460b9ebe68cee56d9103","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-12 18:27:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d9d7656588f4c6968a81c0e1bac6b027a1c111d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:02:01.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/8510609d35cc43648942051fa9cf20dc : SUCCESS in 36m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d262e46d861049359f2173dd7340d721 : FAILURE in 6m 25s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/20f4bbc2095748b68fd98ab43bf41cd0 : SUCCESS in 1h 24m 34s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/18daad523a9d494c89c2594aca2e6e32 : SUCCESS in 28m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bc9fffc8cd3e422fb3cbe672d810e38a : FAILURE in 10m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4d803b6fb76d410c9bc415719f389008 : SUCCESS in 1h 04m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f359893db44b218e5b98bbb3f19893 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5548df70454449bda975de0f51acdd3c : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6604c1e3627d4ea79026975180229918 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cf7002108404befaf2d5c0e4b2e32ca : RETRY_LIMIT in 2m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e3b382caae941218b8a7fb770856fa9 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67a40a8414a047d48dc6f83f7a990ded : SUCCESS in 4m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ec0348530a51447e85ef1db2822c1edf : SUCCESS in 36m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e5588327f2c4e9e8761d3dde23fd833 : FAILURE in 5m 13s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fdcd9c3c76734e049aff8b14b98d4e13 : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":7},{"id":"11dc655f99a310dcd47b7cca2a8064feeb9415bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:27:19.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/b2ff4f44462d4e4f9c6d157a6807069c : SUCCESS in 2h 13m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7f64a624f9094734a341366708e3228b : SUCCESS in 2h 02m 02s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/88b41dbb224a423b81e03c3f10e69de6 : FAILURE in 9m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53f2d617191c48749ea1131a02142db8 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38587647d9f24a339c95dc3a096128de : SUCCESS in 8m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9168d559a50e448a9e767cbefbd79bf0 : SUCCESS in 1h 46m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"18dfc277d98044b96c1f652bd35789a6df97a7fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-12 20:23:10.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"77d7ecc1a5205223a7d4d2775f8ac2612b768184","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-12 20:30:58.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"48da2c5b1ee73d93129448b499fcfca543e6d5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 21:54:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/5be87662514b412aaadf8f476c727ca9 : SUCCESS in 58m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/626cf79503014b5a9aa61ece1f3d5cf7 : SUCCESS in 1h 10m 15s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/264611c95b634486af7ab8a01d519667 : SUCCESS in 1h 29m 07s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e3807222abef49b39177a8361ead1494 : SUCCESS in 35m 17s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8ade4d609a844713b47411029a2e40cf : SUCCESS in 1h 02m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a52fcca628524097bdee122bc8e191c6 : SUCCESS in 1h 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9db51a17764343ac74bfc423027d35 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd7d1c6e5ca942289c8176e06d63ac1c : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6365c4d4983b43ffb30175e06f416842 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/181fe624e8994f67ba3a890af107d0f5 : RETRY_LIMIT in 3m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c33f92aa2bb411da675cef986e79f12 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17f97eb3294c4cfea237888407ff932c : SUCCESS in 4m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bb8a1e83e3a944fea580b2cc6b5eabc7 : SUCCESS in 49m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8541545d386e40d5b0b94e0742ec7294 : SUCCESS in 39m 54s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d978f46acb76422f86d8a7882781a67f : SUCCESS in 1h 10m 19s","accounts_in_message":[],"_revision_number":8},{"id":"ae35da520e25e3c90e080a09d3f67d4c041f243e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 22:05:34.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/0ae9e9ebce5b490a8deff1ea0e974a9d : SUCCESS in 1h 36m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a6aa391e32164e3f8fe82868f6db9ca6 : SUCCESS in 1h 16m 55s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/63a13d49f7574f728ffe2c23cccbafc2 : SUCCESS in 1h 40m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1cb8bcf47b124c70be40dbcb4b9a0f98 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d20304a1a5804442af17555249e1de3d : SUCCESS in 8m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/51ed04dcf7dd4886b7046e88bc758b64 : FAILURE in 1h 20m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"40bc32ef215e524e5176e1ec74c8fea40e8aa5c4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-13 07:08:07.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"074dfc8cc7056828255b8636d50dec69115192c2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-13 08:58:27.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"22d194971e68d64abefc5173c2333eab78695eb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 08:59:05.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"e1e397b3616ccb3b2da518643853f79de4d09488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 10:21:58.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/7e0e19a2294c4be0aaf7ea1c659391d6 : SUCCESS in 44m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c4bb63ddc6464adc9c60422f8aca3fd6 : SUCCESS in 1h 10m 12s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2c3f7b4ea2f746fa9ce94b0f1b853e22 : SUCCESS in 1h 19m 21s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5d32cd367cac46e9ad1e4abaf9724c88 : SUCCESS in 37m 09s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d4b0e2e2ce934c368f0e5f0905e2f2db : SUCCESS in 59m 00s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/71713367bbf242a3a42688418920ae36 : SUCCESS in 1h 09m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a7a8c721a0c46fc99672e22ccd4fd86 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75e497818ef44dfa84a72c07dd465644 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/582e038d29d1468b822637136df958b0 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93e74f2b54a74b118ca83933e6b6e16a : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d994d8d6dfb942ea96f5719375ab5089 : SUCCESS in 3m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/152f50f8afed448f9ca48b7fd1f3e0e8 : FAILURE in 33m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/da7c769c587742bf9bc315fb82a01013 : SUCCESS in 49m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a4ce35d7376f4ee6acb8b0e8752b4471 : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":8},{"id":"e6a54b9c1d98fd3c5d8d11ca1686a6c55d39932e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-13 10:41:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e95fea92405910c7fd6f5058c99caecce1b44e57","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-13 10:47:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f20bdb161c14c93129e13bfa7f2cbf9125425a15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-14 19:08:36.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"842862d899a915d87e06ede5e1fed28b0eb6e03a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 20:34:20.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/d2b89114d3764be68cb7781006b77529 : SUCCESS in 56m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e3088b099dee44639f129a6fbfd4e0a6 : SUCCESS in 1h 13m 52s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c42c8f42597c4cb2851d87910ad3cfae : SUCCESS in 1h 22m 27s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0ded8e8ae5874e82a45027552b7c2461 : SUCCESS in 33m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2b50aad6d7b5492a9e3102c18e136651 : SUCCESS in 1h 01m 00s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a346a44786694311b5211adea6625a61 : SUCCESS in 1h 06m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8efcc8e7b15420ba364b716afe71e55 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54f490ec49134323905859d4d34d28d4 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b266312ac0e4ac99b8744a681fb6d21 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab728177ec7042a887c5ff056d7b7e56 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab7cf48b1e174d079088a4cd4d956a3b : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a2f38fae8e14e70b431abe9d75495c4 : SUCCESS in 4m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0cbeec07811e4b9a9a41fe69a13ea4fb : SUCCESS in 56m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0ba5c5a35794deeadc20d5cd2187655 : SUCCESS in 56m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c4004cbb57dc4270a31c511c778d9c86 : SUCCESS in 1h 22m 18s","accounts_in_message":[],"_revision_number":9},{"id":"637156e9cb9146b2ad7db79209e907735fe47a7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 21:20:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/4a4171ce6a854cc58b40d91e93ce9a85 : SUCCESS in 1h 44m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cfea3616f64d4806835c72ae5179e275 : SUCCESS in 1h 19m 27s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e4d0e1ad809d45f09c56c1ef1e5c8692 : SUCCESS in 2h 08m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f7f98f724644601bdff67332329edb2 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c526d8382c3a433a90a0fb6d6c7c2993 : SUCCESS in 6m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d52f586f777e49dfaa9e7a143cfdd0cb : SUCCESS in 1h 33m 51s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8d4513b8200b01a4edd815cf105c07e1f872f10c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-15 08:42:25.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"6210a94159c06a9aaf9cfc71b5df79ee2420a5c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 08:42:58.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"466837c6540734ccaa364a9e99bf972e89262229","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 10:25:58.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/5b6cec30a6524ff1996c49bf3b8400cb : SUCCESS in 40m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/06bdb6c675fe45d2a1ca0c5f4d0d9faf : SUCCESS in 1h 13m 31s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2e31e7ad58624794828e243c0855f4be : SUCCESS in 1h 25m 23s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5194a352452a48bb934dd2d6bfc951f4 : SUCCESS in 38m 32s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5080f25de17d43f2b9caa16fd6d494dc : SUCCESS in 51m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c1c9179a7fb14b16ba9351625b423646 : SUCCESS in 1h 11m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8946801cc48b42658195515a6ca13d11 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfaa03111881437e9c4c17279012dff9 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32a56adbe9f246aebfeecf157a9b4acc : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e1413678dfb49f589a2079e09609c2d : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/213f6e5449304836afcbff500d6945f2 : SUCCESS in 4m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8be50b1d9b0e445d89b25ef10467257a : SUCCESS in 35m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f6a33bc3d382495bb84d76db0439d310 : SUCCESS in 54m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6d68b3dea1f24d57b8614f0f805da00c : SUCCESS in 1h 35m 50s","accounts_in_message":[],"_revision_number":9},{"id":"cb48f8ca973e3414fad6acdbc5a4f283c63b5368","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 10:25:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"bfd22de94164fbc4ea57ed051e23b4fe9df0c740","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 10:28:17.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0645ef07cda74167963a74e2aa4f4eeb : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/134567d733bd48bbab628af5f9cf27ed : SUCCESS in 58s","accounts_in_message":[],"_revision_number":9},{"id":"e9d099202bc4cd27af2da598cfbc94e360b42a4e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-19 15:29:38.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4ac100ee35dfc181ee68e3574fc03eef7674b18a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 15:51:35.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"db5b9fcffbf06a52546ff62828882632fe818a92","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-19 19:01:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"1b697cc69bc37df86b4ba27924732d01a710096c","revisions":{"89ad71e9dca192d0be60f9d35397b2cfb412bed7":{"kind":"REWORK","_number":1,"created":"2022-04-11 10:36:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/837287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/837287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/837287/1"}}},"commit":{"parents":[{"commit":"c8b38fd784f8d5a65a27f1996cc9e954d94b1017","subject":"drop install_type from image names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8b38fd784f8d5a65a27f1996cc9e954d94b1017"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:32:19.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:35:27.000000000","tz":120},"subject":"drop install_type almost everywhere","message":"drop install_type almost everywhere\n\nThe only place where it stays is argument parsing where it gets ignored.\n\nChange-Id: I6890a7ec7baf507f8acd75b42bbd0fb17c2e4511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89ad71e9dca192d0be60f9d35397b2cfb412bed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89ad71e9dca192d0be60f9d35397b2cfb412bed7"}]},"branch":"refs/heads/master"},"cd4453e10dc1154614e005c9ab8900a78b52eec0":{"kind":"REWORK","_number":2,"created":"2022-04-11 11:02:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/837287/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/837287/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/837287/2"}}},"commit":{"parents":[{"commit":"c8b38fd784f8d5a65a27f1996cc9e954d94b1017","subject":"drop install_type from image names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8b38fd784f8d5a65a27f1996cc9e954d94b1017"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:32:19.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 11:02:22.000000000","tz":120},"subject":"drop install_type almost everywhere","message":"drop install_type almost everywhere\n\nThe only place where it stays is argument parsing where it gets ignored.\n\nChange-Id: I6890a7ec7baf507f8acd75b42bbd0fb17c2e4511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd4453e10dc1154614e005c9ab8900a78b52eec0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd4453e10dc1154614e005c9ab8900a78b52eec0"}]},"branch":"refs/heads/master"},"f58a834379d41144f8ee4b711fc68ada07dd6966":{"kind":"REWORK","_number":3,"created":"2022-04-11 11:09:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/837287/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/837287/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/837287/3"}}},"commit":{"parents":[{"commit":"c8b38fd784f8d5a65a27f1996cc9e954d94b1017","subject":"drop install_type from image names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8b38fd784f8d5a65a27f1996cc9e954d94b1017"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:32:19.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 11:09:34.000000000","tz":0},"subject":"drop install_type almost everywhere","message":"drop install_type almost everywhere\n\nThe only place where it stays is argument parsing where it gets ignored.\n\nChange-Id: I6890a7ec7baf507f8acd75b42bbd0fb17c2e4511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f58a834379d41144f8ee4b711fc68ada07dd6966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f58a834379d41144f8ee4b711fc68ada07dd6966"}]},"branch":"refs/heads/master"},"130bb85b7365edb4b120b0dbed7762b7aea50d63":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-04-12 13:13:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/837287/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/837287/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/837287/4"}}},"commit":{"parents":[{"commit":"c8b38fd784f8d5a65a27f1996cc9e954d94b1017","subject":"drop install_type from image names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8b38fd784f8d5a65a27f1996cc9e954d94b1017"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:32:19.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-12 13:13:10.000000000","tz":0},"subject":"drop install_type almost everywhere","message":"drop install_type almost everywhere\n\nThe only place where it stays is argument parsing where it gets ignored.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/837291\n\nChange-Id: I6890a7ec7baf507f8acd75b42bbd0fb17c2e4511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/130bb85b7365edb4b120b0dbed7762b7aea50d63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/130bb85b7365edb4b120b0dbed7762b7aea50d63"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1b6fa6fe0d36d2c041812eff83c2b8ee039cabbd":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-04-12 13:13:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/837287/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/837287/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/837287/5"}}},"commit":{"parents":[{"commit":"4b5cfa85eb7c70fd4d2536e48b6b4d279c0ed163","subject":"Merge \"drop list of images unbuildable for source type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b5cfa85eb7c70fd4d2536e48b6b4d279c0ed163"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:32:19.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-12 13:13:15.000000000","tz":0},"subject":"drop install_type almost everywhere","message":"drop install_type almost everywhere\n\nThe only place where it stays is argument parsing where it gets ignored.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/837291\n\nChange-Id: I6890a7ec7baf507f8acd75b42bbd0fb17c2e4511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b6fa6fe0d36d2c041812eff83c2b8ee039cabbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b6fa6fe0d36d2c041812eff83c2b8ee039cabbd"}]},"branch":"refs/heads/master","description":"Rebase"},"cb7cb25ee9eac179382b443c4928776ca1c17890":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-04-12 13:27:23.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/837287/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/837287/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/837287/6"}}},"commit":{"parents":[{"commit":"c8b38fd784f8d5a65a27f1996cc9e954d94b1017","subject":"drop install_type from image names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8b38fd784f8d5a65a27f1996cc9e954d94b1017"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:32:19.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-12 13:27:23.000000000","tz":0},"subject":"drop install_type almost everywhere","message":"drop install_type almost everywhere\n\nThe only place where it stays is argument parsing where it gets ignored.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/837291\n\nChange-Id: I6890a7ec7baf507f8acd75b42bbd0fb17c2e4511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb7cb25ee9eac179382b443c4928776ca1c17890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cb7cb25ee9eac179382b443c4928776ca1c17890"}]},"branch":"refs/heads/master","description":"Rebase"},"c88584453748b03a0912640b112b9dfdce13ad3e":{"kind":"REWORK","_number":7,"created":"2022-04-12 17:11:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/837287/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/837287/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/837287/7"}}},"commit":{"parents":[{"commit":"4b5cfa85eb7c70fd4d2536e48b6b4d279c0ed163","subject":"Merge \"drop list of images unbuildable for source type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b5cfa85eb7c70fd4d2536e48b6b4d279c0ed163"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:32:19.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-12 17:11:01.000000000","tz":120},"subject":"drop install_type almost everywhere","message":"drop install_type almost everywhere\n\nThe only place where it stays is argument parsing where it gets ignored.\n\nHardcoded to \u0027source\u0027 to make merging both kolla and kolla-ansible\nchanges easier.\n\nChange-Id: I6890a7ec7baf507f8acd75b42bbd0fb17c2e4511\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c88584453748b03a0912640b112b9dfdce13ad3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c88584453748b03a0912640b112b9dfdce13ad3e"}]},"branch":"refs/heads/master"},"8bc278ad22ccc9cdbb4bf690524d40094236f622":{"kind":"REWORK","_number":8,"created":"2022-04-12 20:23:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/87/837287/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/837287/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/837287/8"}}},"commit":{"parents":[{"commit":"4b5cfa85eb7c70fd4d2536e48b6b4d279c0ed163","subject":"Merge \"drop list of images unbuildable for source type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b5cfa85eb7c70fd4d2536e48b6b4d279c0ed163"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:32:19.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-12 20:22:46.000000000","tz":120},"subject":"drop install_type almost everywhere","message":"drop install_type almost everywhere\n\nThe only place where it stays is argument parsing where it gets ignored.\n\nHardcoded to \u0027source\u0027 to make merging both kolla and kolla-ansible\nchanges easier.\n\nChange-Id: I6890a7ec7baf507f8acd75b42bbd0fb17c2e4511\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/837620\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bc278ad22ccc9cdbb4bf690524d40094236f622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bc278ad22ccc9cdbb4bf690524d40094236f622"}]},"branch":"refs/heads/master"},"1b697cc69bc37df86b4ba27924732d01a710096c":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-04-14 19:08:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/837287/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/837287/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/837287/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/837287/9"}}},"commit":{"parents":[{"commit":"c4fda7baa3ffc36b555c32a34a00042b6035b917","subject":"Fix image builds with sources using a type\u003dgit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4fda7baa3ffc36b555c32a34a00042b6035b917"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:32:19.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-14 19:08:36.000000000","tz":0},"subject":"drop install_type almost everywhere","message":"drop install_type almost everywhere\n\nThe only place where it stays is argument parsing where it gets ignored.\n\nHardcoded to \u0027source\u0027 to make merging both kolla and kolla-ansible\nchanges easier.\n\nChange-Id: I6890a7ec7baf507f8acd75b42bbd0fb17c2e4511\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/837620\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b697cc69bc37df86b4ba27924732d01a710096c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b697cc69bc37df86b4ba27924732d01a710096c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
