)]}'
{"id":"openstack%2Fkolla~837294","triplet_id":"openstack%2Fkolla~master~I9cd225d3a7fd99fdeaf4926730c95d37f8f7eac7","project":"openstack/kolla","branch":"master","topic":"zed-remove-support-for-binary-images","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-12 10:19:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9cd225d3a7fd99fdeaf4926730c95d37f8f7eac7","subject":"drop list of images unbuildable for source type","status":"MERGED","created":"2022-04-11 10:57:30.000000000","updated":"2022-04-12 10:51:42.000000000","submitted":"2022-04-12 10:19:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837294-zed-remove-support-for-binary-images","meta_rev_id":"4c16223d62a35ba6604fa32c7d8c37d3da2eeede","_number":837294,"virtual_id_number":837294,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-12 10:19:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-12 08:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-11 11:01:16.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-12 08:38:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 11:01:16.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-11 12:15:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-12 08:38:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"af768e825329b1e36e0d32e104237ae738d27e80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 10:57:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4311d33e0a6586b15fe81e66ed5c8af7dd00e9ed","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 11:01:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d4fc8079d8398941d6063115bbf702b6a5b71cf1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:18:01.000000000","message":"Topic set to zed-remove-support-for-binary-images","accounts_in_message":[],"_revision_number":1},{"id":"585ee148ad7a1c4741fc18978dea539d56915d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:15:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6baa412ac493475eacdb7ac07252ce6f : SUCCESS in 35m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/64787bfb6117426cbc737c39c9e5585f : SUCCESS in 1h 00m 19s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2ad8d472ac4e47c1a75a13e1d9381401 : SUCCESS in 1h 13m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/25974cbac647490b91ffc36d60334864 : SUCCESS in 33m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4f0cbdf3ee1044a9aa4992d86ee320e7 : SUCCESS in 49m 13s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7f985f22cb30412b92cd65deecf4ae4d : SUCCESS in 1h 01m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2228d44a63b4087886fb36431bc47ac : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e4e71a56d544dd4b6c11ed06f0e33dd : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a0b82190c7e4bd9b6001d18492fb706 : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b0f9c466d0646fd83e59df29fbe5a59 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07274c7d44bd4bc49e9a2e9aaa1bed7a : SUCCESS in 7m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2881df15272e4f5db871005b8c399e31 : SUCCESS in 33m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1af9d406e9864773b572f73bad21277c : SUCCESS in 53m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/47015c71f3d94ed584535faca0a4ff80 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1d1637667823c1e6995a400cc2a02d518d29c4b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 20:33:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/670c17c2c60543508f90f71416f457a1 : SUCCESS in 2h 03m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fe1c386b3ebe431e8878ee6bc4bebfc6 : SUCCESS in 1h 48m 12s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f0b989d878b54d7dbd5265deb0bec115 : SUCCESS in 2h 13m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf18b30eefa249a0b4163481c04000ef : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95c44ae8d09747f3994c38e474449a3f : SUCCESS in 6m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/96237095bdcd440cbbbe0e3d7fc36370 : SUCCESS in 1h 41m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c54529c6f0b2241511ee2b7c58ee85f4a22e0ae8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-12 08:38:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c61678ef7d205d7568c1299d2b121371ec2b5ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 08:39:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fce5d0d8f5e0123359796a2c31d04e4162810e52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:19:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/2d54015d130441408d86dfb119fe79a2 : SUCCESS in 41m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ad4e918151064f4db1bde948b5af3337 : SUCCESS in 1h 20m 11s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f2e7672f791d4f32bea6ce2f1fe9219b : SUCCESS in 1h 34m 47s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ef153e1b47f04a778451fd46ef9c25a9 : SUCCESS in 43m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4f407e9ebba143dba023b4a88d825320 : SUCCESS in 1h 02m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/366a00e2a2b84f118dad4366e676099a : SUCCESS in 1h 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3baa45adb0af424fb86dc14efd6a09bc : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ae89b8be7d14d6bb6bdf6056c9effd2 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b1a856afe4a41e39bdbb20c43d45a5f : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a91a38d2f4e465f98dd2561bb6933b2 : SUCCESS in 7m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/eacf4e2f3b434c128b1429a6515b2c44 : SUCCESS in 35m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/924995aefe1e47ce9183a7739ee530a1 : SUCCESS in 54m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d3c1a41989f44e53bae13226b5101f03 : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":1},{"id":"46b2220dd2f21e293998ceb3c3a4d3fdb95babd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:19:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4c16223d62a35ba6604fa32c7d8c37d3da2eeede","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:51:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41949dbd90774f27a00b6f2a7fa56300 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7aceab5f6cf6a820991c7096a37eab1270b0f398","revisions":{"7aceab5f6cf6a820991c7096a37eab1270b0f398":{"kind":"REWORK","_number":1,"created":"2022-04-11 10:57:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/837294/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/837294/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/837294/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/837294/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/837294/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/837294/1"}}},"commit":{"parents":[{"commit":"657820362dd0b52f8462054e043fd892d14fbe76","subject":"Merge \"docker: drop binary parts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657820362dd0b52f8462054e043fd892d14fbe76"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:56:56.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 10:56:59.000000000","tz":120},"subject":"drop list of images unbuildable for source type","message":"drop list of images unbuildable for source type\n\nIf they are unbuildable then they should be in distro or distro+arch\nblocks now.\n\nChange-Id: I9cd225d3a7fd99fdeaf4926730c95d37f8f7eac7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aceab5f6cf6a820991c7096a37eab1270b0f398"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7aceab5f6cf6a820991c7096a37eab1270b0f398"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
