)]}'
{"id":"openstack%2Fkolla~837295","triplet_id":"openstack%2Fkolla~master~Ieba3260ffa11fa2d3982f379f95abed9f840028d","project":"openstack/kolla","branch":"master","topic":"zed-remove-support-for-binary-images","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-12 10:10:04.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-11 11:58:53.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"Ieba3260ffa11fa2d3982f379f95abed9f840028d","subject":"CI: drop \u0027source\u0027 from build job names","status":"MERGED","created":"2022-04-11 11:07:35.000000000","updated":"2022-04-12 10:13:15.000000000","submitted":"2022-04-12 10:10:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837295-zed-remove-support-for-binary-images","meta_rev_id":"313e976444d4f14a030cc0256731c9e3b51a7166","_number":837295,"virtual_id_number":837295,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-12 10:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-12 08:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-11 11:59:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-12 01:28:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-12 08:38:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 11:08:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-11 11:08:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-11 11:19:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-04-11 11:59:55.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-12 01:28:24.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-04-12 08:38:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"68e71c2b8f540899435c74635f588174b956f309","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:07:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e9af18a114ff92a7e45ca9ea19916f93ddbb471","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 11:08:04.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Job kolla-ansible-centos8s not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-centos8s\n          - kolla-ansible-centos8s\n          - kolla-ansible-centos8s-source-upgrade\n      check-arm64:\n        jobs:\n          - kolla-build-centos8s-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8s\n          - kolla-ansible-centos8s\n          - kolla-ansible-centos8s-source-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-centos8s-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8s-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos8s\n          - kolla-ansible-centos8s-bifrost:\n              files: ^docker\\/(base|bifrost|openstack-base)\\/.*\n          # Test rabbitmq \u0026 mariadb changes in multinode ceph jobs.\n          - kolla-ansible-centos8s-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-source-upgrade-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-centos8s-scenario-nfv:\n              files: ^docker\\/(base|barbican|heat|mistral|openstack-base|redis|tacker)\\/.*\n          - kolla-ansible-centos8s-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-centos8s-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-centos8s-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-centos8s-masakari:\n              files: ^docker/(base|masakari)/\n          - kolla-ansible-centos8s-octavia:\n              files: ^docker/(base|neutron|octavia|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8s-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8s-prometheus-efk:\n              files: ^docker/(base|elasticsearch|fluentd|grafana|kibana|prometheus)/\n          - kolla-ansible-centos8s-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"9ebdd29f804742ca51998ed87e28d61fa6a6678d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 11:08:16.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Job kolla-ansible-centos8s not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-centos8s\n          - kolla-ansible-centos8s\n          - kolla-ansible-centos8s-source-upgrade\n      check-arm64:\n        jobs:\n          - kolla-build-centos8s-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8s\n          - kolla-ansible-centos8s\n          - kolla-ansible-centos8s-source-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-centos8s-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8s-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos8s\n          - kolla-ansible-centos8s-bifrost:\n              files: ^docker\\/(base|bifrost|openstack-base)\\/.*\n          # Test rabbitmq \u0026 mariadb changes in multinode ceph jobs.\n          - kolla-ansible-centos8s-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-source-upgrade-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-centos8s-scenario-nfv:\n              files: ^docker\\/(base|barbican|heat|mistral|openstack-base|redis|tacker)\\/.*\n          - kolla-ansible-centos8s-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-centos8s-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-centos8s-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-centos8s-masakari:\n              files: ^docker/(base|masakari)/\n          - kolla-ansible-centos8s-octavia:\n              files: ^docker/(base|neutron|octavia|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8s-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8s-prometheus-efk:\n              files: ^docker/(base|elasticsearch|fluentd|grafana|kibana|prometheus)/\n          - kolla-ansible-centos8s-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":1},{"id":"9e63787cfaa80b5a2188e99b72f79f45c279a774","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:11:40.000000000","message":"Topic set to zed-remove-support-for-binary-images","accounts_in_message":[],"_revision_number":1},{"id":"249f35c0600eb559142efb15997175f5fa3e41f4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 11:19:04.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6bd3d3d150c5ef88c0ffa8d83b361ddcbdaa265f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:22:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"909cde9c0a509895e963746930698cb7caca8214","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:26:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"251382cb7b58ad589db936421bcf654ecb781453","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 11:26:47.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Job kolla-ansible-source-centos8s not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-centos8s\n          - kolla-ansible-source-centos8s\n          - kolla-ansible-centos8s-source-upgrade\n      check-arm64:\n        jobs:\n          - kolla-build-centos8s-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8s\n          - kolla-ansible-source-centos8s\n          - kolla-ansible-centos8s-source-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-centos8s-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8s-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos8s\n          - kolla-ansible-centos8s-bifrost:\n              files: ^docker\\/(base|bifrost|openstack-base)\\/.*\n          # Test rabbitmq \u0026 mariadb changes in multinode ceph jobs.\n          - kolla-ansible-centos8s-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-source-upgrade-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-centos8s-scenario-nfv:\n              files: ^docker\\/(base|barbican|heat|mistral|openstack-base|redis|tacker)\\/.*\n          - kolla-ansible-centos8s-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-centos8s-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-centos8s-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-centos8s-masakari:\n              files: ^docker/(base|masakari)/\n          - kolla-ansible-centos8s-octavia:\n              files: ^docker/(base|neutron|octavia|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8s-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8s-prometheus-efk:\n              files: ^docker/(base|elasticsearch|fluentd|grafana|kibana|prometheus)/\n          - kolla-ansible-centos8s-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":2},{"id":"dbb002ff1395d8466be0eb71efdec2c11d3d99eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 11:27:03.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  Job kolla-ansible-source-centos8s not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kolla-build-centos8s\n          - kolla-ansible-source-centos8s\n          - kolla-ansible-centos8s-source-upgrade\n      check-arm64:\n        jobs:\n          - kolla-build-centos8s-aarch64\n      gate:\n        queue: kolla\n        jobs:\n          - kolla-build-centos8s\n          - kolla-ansible-source-centos8s\n          - kolla-ansible-centos8s-source-upgrade\n      periodic:\n        jobs:\n          - kolla-publish-centos8s-quay\n      periodic-weekly:\n        jobs:\n          - kolla-publish-centos8s-dockerhub\n      experimental:\n        jobs:\n          - kolla-build-no-infra-wheels-centos8s\n          - kolla-ansible-centos8s-bifrost:\n              files: ^docker\\/(base|bifrost|openstack-base)\\/.*\n          # Test rabbitmq \u0026 mariadb changes in multinode ceph jobs.\n          - kolla-ansible-centos8s-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-source-upgrade-cephadm:\n              files: ^docker\\/(base|cinder|glance|mariadb|openstack-base|rabbitmq)\\/.*\n          - kolla-ansible-centos8s-zun:\n              files: ^docker\\/(base|cinder|etcd|iscsid|kuryr|openstack-base|zun)\\/.*\n          - kolla-ansible-centos8s-scenario-nfv:\n              files: ^docker\\/(base|barbican|heat|mistral|openstack-base|redis|tacker)\\/.*\n          - kolla-ansible-centos8s-ironic:\n              files: ^docker\\/(base|dnsmasq|ironic|ironic-inspector|iscsid|openstack-base)\\/.*\n          - kolla-ansible-centos8s-swift:\n              files: ^docker/(base|openstack-base|glance|swift)/\n          - kolla-ansible-centos8s-mariadb:\n              files: ^docker/(base|mariadb)/\n          - kolla-ansible-centos8s-masakari:\n              files: ^docker/(base|masakari)/\n          - kolla-ansible-centos8s-octavia:\n              files: ^docker/(base|neutron|octavia|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8s-ovn:\n              files: ^docker/(base|neutron|openstack-base|openvswitch|ovn)/\n          - kolla-ansible-centos8s-prometheus-efk:\n              files: ^docker/(base|elasticsearch|fluentd|grafana|kibana|prometheus)/\n          - kolla-ansible-centos8s-kvm:\n              files: ^docker/nova/\n\n  in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":2},{"id":"8cc649d88ee2493823ffdb5310f8d33ea2839a1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:29:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"72727db16e3344e759c86166f3a7191f9548853f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 11:58:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b5f98cdec50de36cb47a1fa8e28e07b929b4a514","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 11:59:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ce2d3ebbcfc881b545733308698c64c11a2d68fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:28:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/0fd7da9160f346f3bd8b134b2e123cc5 : SUCCESS in 53m 41s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/84debfad18c04aa3824a17e3eedc02e7 : SUCCESS in 36m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b6ec8be0d24fe98f342f990ebfd5e3 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d4df26809294e768b42b9529bd37dc4 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7df3c52398644b8492c27e53d121c1f9 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd35e0822b1b494ca53e04e84f7e72d1 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5235b2528ad540359eefc3c5e8f9f089 : SUCCESS in 6m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2f8348f0a848453c8a0c77de366bcce3 : SUCCESS in 29m 04s","accounts_in_message":[],"_revision_number":3},{"id":"0998b7fa498c0beb887ad3379bc488508417e8d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 23:09:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/0f09bd7040a74f6c93ddb5cc7ea22a7d : SUCCESS in 2h 27m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/18b90f0343924219a8e73c14b467b3b9 : SUCCESS in 1h 44m 27s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4799f843a84f4bae886035c1f02f3e62 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96924440194143fda0980c880f2468cc : SUCCESS in 6m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/55a0848f2a964943975142ea17228899 : SUCCESS in 1h 46m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8bce23f42a4470f5a39c50d32eeaf93dab195b72","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-04-12 01:28:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"14c31b9e2947d5f351b630970d0e69a23a6f557a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-12 08:38:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a4a9d01b0f04ab10659f874797316f19e91d07f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 08:39:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2087386dddd92bb246dc820e4794f08a43a58642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:10:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/71839f87e390486680488491262a1419 : SUCCESS in 36m 00s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5f10d4797add4736b271b6ad1570d478 : SUCCESS in 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/500b08cb1ec44b6b8e1e411f3b017451 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a151f088be74c9fb08b3ae2691ce683 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b939ac77e2b466590016ad54d7437f5 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bfb10df5b84474f8b1198ff5d5dbbfa : SUCCESS in 6m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9aec11739089419f99e7742ace5fe214 : SUCCESS in 31m 48s","accounts_in_message":[],"_revision_number":3},{"id":"c3f74271931681708ea65f666aac9afa1f25c9cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:10:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"313e976444d4f14a030cc0256731c9e3b51a7166","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:13:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b650f807ce164951bf6f232b6f43d80b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a0942b85496e9813c5a9754b16bca590527a55e8","revisions":{"26fb238198afffca66631303dd8605540089e2f0":{"kind":"REWORK","_number":1,"created":"2022-04-11 11:07:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/95/837295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/837295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/837295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/837295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/837295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/837295/1"}}},"commit":{"parents":[{"commit":"657820362dd0b52f8462054e043fd892d14fbe76","subject":"Merge \"docker: drop binary parts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657820362dd0b52f8462054e043fd892d14fbe76"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 11:06:19.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 11:06:19.000000000","tz":120},"subject":"CI: drop \u0027source\u0027 from job names","message":"CI: drop \u0027source\u0027 from job names\n\n*-source-upgrade kept old names as we need to add *-binary-upgrade ones\nto check previous_release-binary/* -\u003e current_release/* upgrade path\n\nChange-Id: Ieba3260ffa11fa2d3982f379f95abed9f840028d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26fb238198afffca66631303dd8605540089e2f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26fb238198afffca66631303dd8605540089e2f0"}]},"branch":"refs/heads/master"},"1703db1a7128f2b4357abd1193d8c1324e168621":{"kind":"REWORK","_number":2,"created":"2022-04-11 11:26:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/95/837295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/837295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/837295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/837295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/837295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/837295/2"}}},"commit":{"parents":[{"commit":"657820362dd0b52f8462054e043fd892d14fbe76","subject":"Merge \"docker: drop binary parts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657820362dd0b52f8462054e043fd892d14fbe76"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 11:06:19.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 11:25:53.000000000","tz":120},"subject":"CI: drop \u0027source\u0027 from build job names","message":"CI: drop \u0027source\u0027 from build job names\n\n*-source-upgrade kept old names as we need to add *-binary-upgrade ones\nto check previous_release-binary/* -\u003e current_release/* upgrade path\n\nk-a jobs will go in separate patch\n\nChange-Id: Ieba3260ffa11fa2d3982f379f95abed9f840028d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1703db1a7128f2b4357abd1193d8c1324e168621"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1703db1a7128f2b4357abd1193d8c1324e168621"}]},"branch":"refs/heads/master"},"a0942b85496e9813c5a9754b16bca590527a55e8":{"kind":"REWORK","_number":3,"created":"2022-04-11 11:29:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/95/837295/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/95/837295/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/837295/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/837295/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/95/837295/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/95/837295/3"}}},"commit":{"parents":[{"commit":"657820362dd0b52f8462054e043fd892d14fbe76","subject":"Merge \"docker: drop binary parts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657820362dd0b52f8462054e043fd892d14fbe76"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 11:29:09.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-11 11:29:09.000000000","tz":120},"subject":"CI: drop \u0027source\u0027 from build job names","message":"CI: drop \u0027source\u0027 from build job names\n\n*-source-upgrade kept old names as we need to add *-binary-upgrade ones\nto check previous_release-binary/* -\u003e current_release/* upgrade path\n\nk-a jobs will go in separate patch\n\nChange-Id: Ieba3260ffa11fa2d3982f379f95abed9f840028d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0942b85496e9813c5a9754b16bca590527a55e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0942b85496e9813c5a9754b16bca590527a55e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
