)]}'
{"id":"openstack%2Fkolla~837300","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Iecc4fa7e3204a4df7fd41e46e55b2895c43ddc43","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-04-20 10:33:35.000000000","reason":"Michal Nasiadka replied on the change"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-21 11:16:40.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-21 11:16:40.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-20 10:21:36.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"Iecc4fa7e3204a4df7fd41e46e55b2895c43ddc43","subject":"[release] centos: switch to CentOS cloud SIG packages","status":"MERGED","created":"2022-04-11 11:15:06.000000000","updated":"2022-04-21 11:18:01.000000000","submitted":"2022-04-21 11:16:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"837300","meta_rev_id":"faac255df6efe1985340ddd89e6ab6b40b76614a","_number":837300,"virtual_id_number":837300,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-21 11:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-20 10:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-20 10:32:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-15 10:57:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-04-20 10:32:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 11:17:36.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-04-11 11:17:56.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-04-14 07:04:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-04-14 08:20:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-15 10:13:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-20 10:33:35.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"97a272defaf54823c635f98bf25e71b84e6691c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-11 11:15:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bd1a2d54f0a540ab9c6e723017fd280bfe301c9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 11:17:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"373497ee3ab8eebf0977817e5e47761a2d901cd0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 11:17:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d15307852d6d9c22993f7aa2282c52d89d2d7b9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-11 11:38:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5cbc53a4748038c5b8272a2d6e1a01568eac30a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-11 11:38:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f903358011d73d9ac141ac1321c2bd56786a160d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-11 11:38:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d2fdd98b97b310a929afb1495801c57e9ec3268","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-14 07:04:48.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"f0062cb4e39aaa4c373ce4b192faf95c7b7e12e3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-14 07:05:00.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"ab83eebade0dd789dc83d3404e0617562e4c2100","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-14 07:07:15.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"69673c876900f590b4ab1085dfd685967a2baf26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 08:20:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b13e7fc42b944969aac61075afc8665a : FAILURE in 10m 08s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1472a99447b44d9faef5119cfada9d0e : FAILURE in 9m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3e498d719ebf4b659108c46953b041d6 : FAILURE in 16m 04s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/15f5fa79c3da4dc7a72ddacf1154c4f2 : FAILURE in 15m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6db86e48bc3443fb9ff02e20fdcce108 : FAILURE in 16m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/10835c57d027474e8da2ec8f754baea4 : SUCCESS in 30m 58s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/41e2c9dbf0264b958b3bd672bbdcab1b : SUCCESS in 17m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c3c990ef84a8424d825f8a3f8977f395 : SUCCESS in 58m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5596429134704ac98acd5be2b3bbfce9 : SUCCESS in 1h 08m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c55d9ed4aad247b681c094ba8b18fc2a : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94ffc5050b7442a494a6092a1703e5c9 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dee10670f794cd5a5efa31279e88b36 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de6c89c597846ca82fd788e756fa03c : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd969a951ec94c6498e9ef82936f22a6 : SUCCESS in 18m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1fda0cdbf29a45a3a771ab5a7e365573 : FAILURE in 37m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e988bc4aa190406ca484e842d1ad03d9 : SUCCESS in 51m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5cd3dc48b29947d9807adbb3362ee842 : SUCCESS in 44m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f5e6cce566534e0db89d52fbff742a49 : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":3},{"id":"d75dc87dc97ee195a2a24c29681cb27c4181e4db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 08:54:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/79a391057cea4c589918eeb16eee6a25 : FAILURE in 14m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/660e77acee7b400e8e54059440d7b9e4 : SUCCESS in 1h 21m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/557f1b0617ea4603a4e3d09954adc307 : SUCCESS in 1h 45m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7149ee81ef3491f8c0a72ef81a06691 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/10a9b88676484e8eb38a860f486c255e : SUCCESS in 8m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7974c0f9f13c4bc8a6a745f208e73ff5 : FAILURE in 1h 07m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"078b3c85efd086db7a2f4d118767b4eb95888599","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-15 06:49:44.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"9ddaac29bb43cd0b4f66a0b247d4e46b7cddd736","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 08:19:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ae8f6637c139465486ce0de40181f7bf : FAILURE in 14m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a2eb694126ff49c7abb54b077f26e6e4 : FAILURE in 13m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8199fbe68d944573b479950b49dfe4fb : FAILURE in 19m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6a94b566a69b45e382159503e69aea30 : FAILURE in 18m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3ed9ae5f879c4bf1a62f359bba9c35ff : FAILURE in 20m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c6cf55d6d7b643e0b348afe933fa1a2b : SUCCESS in 28m 04s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b26eb39519814e98868077b654cef281 : SUCCESS in 19m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cf5f3a5ea82c49ae819abcfe1b2f4f2f : SUCCESS in 1h 03m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ff793cef607f4467a2852e814661a04b : SUCCESS in 1h 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2705caa6fc745f8a524d92b9b4351c0 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/857447f03ca343e3ab773f1aa84e39e5 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/718849260d494c12872237aec26e8166 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e378196fa1a4d56b1ebd5d6d0e6967e : SUCCESS in 6m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6813b85dda3d4cfbb124b6c01add079f : SUCCESS in 20m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c297c3a07f942d39517d3a53fce628e : SUCCESS in 37m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/71c39c3b28d244a887a9c3f1bcba1343 : SUCCESS in 1h 22m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9168579b325d4905bee15f94cf8b3c19 : SUCCESS in 52m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b7c440c2c9f74db1a4dd62f903471540 : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":4},{"id":"e38f357fe5fb0ce256f4d3e772039df5b3c8aad0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 08:22:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d9021d5f3a64d1b9446e97799a41524 : FAILURE in 13m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f20894a4ce384658a1afe4be2136ab2e : SUCCESS in 58m 07s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1036cc7a58cf4fb19e283e7d0e0d8458 : SUCCESS in 1h 31m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cda9c055f82d4a6c90bde560e84556ac : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31f9c342573247b9a66b6b7985c64f99 : SUCCESS in 6m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cabf5b958ea242aca48f61f7955e178d : SUCCESS in 1h 06m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1c5f0356401ed2a2e1a31894db6bfb261d5ab878","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-15 08:54:32.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"0e08734d12127869e5c308a18d67807f560b55e7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-15 08:59:06.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"50e328ef817f2f96630314a805b96c59af01adc4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-15 09:01:42.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a37518b356bfa15b06f9879fcd84b67bb015eab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-15 09:04:57.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"b2e7dbe626b8f650e51810a476956775b82a79d7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-15 09:05:01.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6b9f524d4161a50d56fa64ead776a380808d985e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-15 09:24:19.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8dc16fc2c3c515c548a49773b80a096e3362fa30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 10:06:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3760b425266646e9b0f282019e07cac4 : FAILURE in 9m 39s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3eb274202de84e9ab96dbb18f1869788 : FAILURE in 9m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/32aa5d0f192f429a8777351d31911f08 : FAILURE in 15m 59s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d718ff82ebbc48c4983b5f67b88f0cdf : FAILURE in 17m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/29152173f49a4a1ca25602ec4ab0d503 : FAILURE in 14m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7f86f27f17f24b02948cb54fd0d150c7 : SUCCESS in 36m 01s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5a1efeb4a398497284d8bc37afbbbdb7 : SUCCESS in 15m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/62c17ace0473420abc1a204487fcb5fe : SUCCESS in 59m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8da61fe605544a6798cc4f4422c4ee29 : SUCCESS in 1h 00m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d5d922bdd2b45318e7bd77157c0748c : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4aa3569ba11c466e977bc67223fca5b9 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/471bcf98e66a415c98495a02ef00137b : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1838f9cf5649b6b64ef285002646d5 : SUCCESS in 6m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/058782f7d3d14f9a8946ce83614e821f : SUCCESS in 20m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/49db17eb941f4349af98079ef972ab9a : SUCCESS in 35m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fec8b3cb22e049e5a5d84c30053ba2e9 : SUCCESS in 54m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bb38a16cc0f74b938534a17dace5062b : SUCCESS in 50m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c190e09dcd724779bedaaec393375634 : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":6},{"id":"9c32ad3cc4a3485fce78a935c693dc6c1681d5c5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-15 10:13:33.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d467932d55ed1862d53dd68ba5b71154fc6ab5be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-15 10:54:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e412540921b0caa4518a31402ea427953eb762a3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-15 10:57:02.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"05872851390851e62ee02d6bf69e84ee9af8b4e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 12:28:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d229bf1a88eb4f4ebc1d49615dab78df : FAILURE in 26m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/23c6a0528ae84fc99caecb7a901e0de4 : SUCCESS in 39m 38s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1512d7dd57cc4ce9a09d7cab47ff096f : SUCCESS in 1h 17m 49s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/51f945c6025647c5849316c5f34fd659 : FAILURE in 29m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/032b928a82e844bc8bf7926ca5350332 : SUCCESS in 1h 27m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2ead3998e6824e1fbfea28304ea06090 : SUCCESS in 28m 26s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f7097f9be80a40f28a5850373ab72e2b : SUCCESS in 17m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/00296ebd9775470b9233adef2fdb6609 : SUCCESS in 1h 02m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ac27298707b842469d0258bea92b7d46 : SUCCESS in 1h 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1e96d1ef7d146c1999ea7f25ea728e2 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4293755b9e434664b8a884e604a597be : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9fe18517c2a94461b61bbb6647c4f486 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a657d61c23444e891e9667a9118c02 : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c783631b7354c39a7fb98b39c628c7c : SUCCESS in 23m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/899bd72744b049ad8d6ac6968b597877 : SUCCESS in 36m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f8b059203ca44c878aa0bd7d1f3de010 : SUCCESS in 57m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/810a6493d62e411abd70d9e93f0477ee : SUCCESS in 55m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2e79e194e33242a99e02a6cb5b053328 : SUCCESS in 1h 05m 20s","accounts_in_message":[],"_revision_number":7},{"id":"3535bde4e6e40bc4df136ea87b6174c6f7f9a4d9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-15 12:31:59.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6aea26d3e4f4e755177e2004ff82b3fa05bd876f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-15 12:50:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4d8cc871dd94885f32fd17b30a9bd2d9ac0fcdcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 13:05:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/aba09e6883a9450cb17d3aabb912200f : SUCCESS in 2h 04m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/01a1fe0465d04ae68005f2fc753ad755 : SUCCESS in 1h 31m 01s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b94b05265ffd478fb6f4d749222534aa : SUCCESS in 2h 09m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c44925dfebd84955ab0915f5cfef5f84 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/181ec644481341dcb918715225167730 : SUCCESS in 6m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b84d03cde34448558f576b4c1187b0e0 : SUCCESS in 1h 31m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"25ab8edd70585cfd9d0b56f8576df66c3fd916d3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-19 06:44:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6469caa372ac19c32815c669ce064edaca4a3b31","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 09:03:59.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"54c24c968d7961d49f5f502f59ecf23ab70d1d18","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 09:32:55.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"88ebfaa54bb380cec9ae020431097ba3ce7c1196","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-20 08:29:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0577972c150f838f155b431a760da32fb0627a7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 09:57:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/da680c2f60694a0880cf8d3b585333c8 : SUCCESS in 28m 31s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6c9abe7708fc407c80bea09bddb3c2d3 : SUCCESS in 31m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f016a4875ba046879cadcae0b2e0fba4 : SUCCESS in 1h 12m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b1fc04dbe33940c188fc5f0017726dbe : SUCCESS in 1h 06m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f115539e2f1b48c583ad1f9a8174e2e3 : SUCCESS in 1h 23m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/25803f578d284890adff26796bd22b6a : SUCCESS in 20m 56s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/db1787ee06df4e88820addbad402ce9b : SUCCESS in 17m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0a8989f492c04adfbc8dcfc420976dc4 : SUCCESS in 1h 04m 11s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e70e18db1edc435c91315707568dc049 : SUCCESS in 1h 18m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18cacb10891d4a6aab36ce1fa7cd7dfc : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a13174e5c86490ca74443a273f7d200 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24a6f4bd40104555b0cb422990057cf3 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2752d1bf831143bc95028d073c80086e : SUCCESS in 6m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fb5d89ba84504558b6cc7b67e1296c95 : FAILURE in 16m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a8f0d73a3aca4431abf8e167e8814974 : SUCCESS in 21m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3a09e8f87864ec2a20c909cc14d17d2 : SUCCESS in 49m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/609622f6d84944eaa4bb6796147274f3 : FAILURE in 18m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b4a6d65cee154191b6b8a6c290683fd0 : SUCCESS in 1h 02m 09s","accounts_in_message":[],"_revision_number":7},{"id":"e384ef73deb8c2952fd736bae87c1df0e746847c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-20 10:21:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"12e43cb712b94fb80a94ac418d4490ddb7ff2ab9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-20 10:21:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7ecebd83c48d7eb33e875f375133f2cb2b491bc5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-20 10:32:53.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"4e87de8cfdc29e164512d193cc7d57984e10539b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 10:33:33.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a8a35faa9f9da9dbba33f2384f35bc6a19a4ee36","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-20 10:33:35.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"a3b6fdf60763e74abaf76f9d52f4c25b5a522a27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 11:57:19.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4f36b179095f403cbacf74c231075d87 : FAILURE in 23m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0a6c23e813164b8fb650d75a20f71526 : FAILURE in 21m 35s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ca069297b66a4d14b33f717092d7d34b : FAILURE in 21m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2c905ac08c3640298a1a4b96101aba73 : SUCCESS in 20m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/41c0d109c1d946fd801f1dcec9c1baa2 : SUCCESS in 1h 06m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/daf557d4ff1a402fb9f4a2b522cea925 : SUCCESS in 1h 17m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b34520cd62c6485ab9ff84608637afb7 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8539a01749704844b3125a19f87a9fbf : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b60267539bf43f9809b3b88ed65f3a6 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf305f598a0e4ec58c66f7fb72435552 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ef231cc006a433383af7f5b7f661ab3 : SUCCESS in 21m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c3d4fbe85794080a9ee75e714488fa2 : SUCCESS in 55m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b3570706812b4600a0cacc50f12c1e52 : SUCCESS in 1h 04m 32s","accounts_in_message":[],"_revision_number":7},{"id":"7c3e1809369d509514d89359f4466e63db515461","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 13:06:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b24c90c78ac4bb1a03bfd4edfaa056c : SUCCESS in 2h 10m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0507d53a04254825b75bd0fa0a465784 : SUCCESS in 1h 42m 11s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f97167cd8734f4597e89f433d8edf6e : SUCCESS in 2h 07m 12s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c843163a00e14e19a90db1db671f1cb5 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/215aeb0fa5864137bf4c8f25849722f9 : SUCCESS in 7m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9775ba0eeab4ce59cece26d9c38cad1 : SUCCESS in 1h 32m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"222d0cc8112d16b294d04c1be1ad58a2506cd688","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-20 16:28:52.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a128895c1006befaa09ee26d8c06660da7d8fbe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:56:31.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3aa7c7c156b64e8ca9edccf830f54e09 : SUCCESS in 33m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/764a9f48f1a2434da5083bd711e6e592 : SUCCESS in 40m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2c3732df3d254cf3b782bb3687df0d4b : SUCCESS in 1h 13m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/addad5032db04fbcaad1f355ae86e8ae : SUCCESS in 1h 07m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9ea865d0fe7047a584770b09897a007a : SUCCESS in 1h 16m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e0fc75a3fe8d4e918f1015a6e530c621 : SUCCESS in 19m 15s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e86ea319db9f4f34a16dff954a526d1b : SUCCESS in 18m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/af71933d501b48d7b39d6504a724b675 : SUCCESS in 56m 32s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bd16fa96c5f749ddb2a3f44a95934b95 : SUCCESS in 1h 22m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efd3511aca8949cdbc2609aa35c7e000 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2f0ab0dd7db4c2e9bde9d44921f5be5 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3f47839bea6456593a67b0ba9e25749 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a72ffad58b548aa84a0976550cd8083 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9be55c2a98b14d629b9d265086ea6da2 : FAILURE in 17m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/57bb5e0f30f8405a8bab0c973132c190 : SUCCESS in 16m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/33f88c8d52224f0ea81c84eb86d99beb : SUCCESS in 46m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4650665fd45e42d2973d4c7a79657c13 : FAILURE in 17m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/72cba42b408342f6bbdb184c3922b357 : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":7},{"id":"0d0babe99614c24c049e27eb96a1a7813982be89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 17:57:03.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3c3817046fcd61643293906a904a94b8f8a53708","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 18:40:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/eade1714e33c4a8e9c2239b00810d720 : SUCCESS in 1h 58m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c83a60b08bf4da4bd6e5caee3ede46d : SUCCESS in 1h 51m 38s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc91a4e75fc54d94b1992e7d616f4e94 : SUCCESS in 2h 09m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/177c92c20fb74f42a84e2b63427a3152 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c9ce42ea92d460a8308674b43e75f08 : SUCCESS in 6m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6a46d0bd7d0e4fe2a578c908ac1f72ef : SUCCESS in 1h 53m 16s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2df02f4a610cd1957f8b5ef765a103a0cc0fca1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 23:35:20.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e732af6368be403ab427bb7781de24bc : SUCCESS in 24m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1b8533f4f07c405b9e0367e86e637060 : SUCCESS in 1h 01m 56s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e68679482ef047948e1e8635bc23d639 : SUCCESS in 1h 06m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e7f38567737843e89e05dbb88b6f7e63 : TIMED_OUT in 5h 33m 42s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/73ca5a0fe60e43238ed18d6dfd201150 : SUCCESS in 45m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/929e58b2bf2041d8ac705b7ecf82e757 : SUCCESS in 1h 02m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f54371d6be13413b98631f85eb2d2bd5 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c24cfa324f3c41eead7c471775087fb8 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3af137eff00e4a949a02138ae048bf05 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6eeae43af0f402d8f126dfa1ad8198d : SUCCESS in 5m 44s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/143a6fe16580480a83e66c9a2c7b77d0 : SUCCESS in 16m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad2804f5dbfc495190f98ca9e76dfa2a : SUCCESS in 57m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0516618f81fc4c7fac734290f2d6f1e5 : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":7},{"id":"94559c23aded0027ce805d0575a7129587b651ad","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 08:14:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f8c3af03a6a093e7a5a708bc46c1abec2b4a6c04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 09:49:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1c049a22105f4b7d822a89465514270c : SUCCESS in 26m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/668cd474de6242959e6a097f316c3796 : SUCCESS in 34m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/77cde73a47bd46b9a3ba2598f7953c55 : SUCCESS in 1h 16m 07s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6b74a058ddf14ebe83d20fae04d1ea74 : SUCCESS in 1h 18m 47s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/552515bd6ef04a0da4e6a95ee9f96623 : SUCCESS in 1h 32m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9814226449ec45569582de9972f69405 : SUCCESS in 23m 27s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e047349959e24e8ca452c219685ace83 : SUCCESS in 22m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/47c25d9a0e954b259e53c52d19a41a69 : SUCCESS in 1h 01m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/aabdfeb2263b43cca97e087fdef1129b : SUCCESS in 1h 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ca850d7b7784bdeaf8ba3f38d294c9f : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fde9b1b70e4c4e3581c8e4fe2436c674 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df4224cfbac2479c83b937c441fa58d3 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c2844d472045489c3769504ced235d : SUCCESS in 7m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8abcefa60ac24fcdb732789d05fc55e1 : SUCCESS in 19m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f295624d6a14ff786bd2cdf5529b527 : SUCCESS in 24m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb5bbb7a2c7e42eda4b6998bcc58ea3f : SUCCESS in 1h 00m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/97a37f1ecad64ca2ae1da33a21ab313d : SUCCESS in 57m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c46c12371ab54eee81e7f77cc2cc0b3a : SUCCESS in 1h 03m 55s","accounts_in_message":[],"_revision_number":7},{"id":"67b222a7bd47379567716f3e2bb681c126638be7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 09:50:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f180140583544015383eb61e6df43b023550e052","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 10:19:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/20ae18aa5c7948c1aa0c58a353807b59 : SUCCESS in 1h 38m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d0627acae0c04c61b40c6893c717417b : SUCCESS in 1h 27m 32s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/af2d07cf50ad41fb99857ff0a7254eb0 : SUCCESS in 2h 02m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c935c04e8ae7401a85c8147c5a37fb51 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5964220ce39141628d79a15a58d41fba : SUCCESS in 7m 16s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9216e9280ea0434a9c1ef1ee1f8922fa : SUCCESS in 1h 02m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"69e379b5279b214ca0556b296c67d8abf6d0dd45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 11:16:39.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e1ce68a664de4c2c9d84ecc11a2c9f72 : SUCCESS in 32m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5041c5051c784990a0edf26a8df0f5b0 : SUCCESS in 1h 08m 29s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f6a964547ef7423dac5cae021b4ef749 : SUCCESS in 1h 22m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/63d83d29e83348aaa5dc0682532f6ee4 : SUCCESS in 24m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/327b863e9edc47b7b08c57717ba45e25 : SUCCESS in 57m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c1e2b9b0b1874b4381823a50f10fc95d : SUCCESS in 1h 12m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36b594b9bde8466490a6979b5341410f : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10c90b04ab4a4cdfa98174f648420e3a : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66496d31af1942ed907a2e5f1bf3c399 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2132ab4a4c4ba08c5967f21ef77aed : SUCCESS in 5m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/629ea873fc2f406685fb1980cc2af047 : SUCCESS in 27m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7c68c6ddd1f4963a9773a7c5f774fb9 : SUCCESS in 56m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fb2946335f8040a1ad6a06ab765cc9b0 : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":7},{"id":"3532ea68fde360df1f06ed30808875ac7d9ba651","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 11:16:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"faac255df6efe1985340ddd89e6ab6b40b76614a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 11:18:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adf53399553c46e7a8eb54a2c9f06377 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9d22fc32fbe85e84376ac77180f026f1bc7ce2dc","revisions":{"2461bee0d028268c0d13e2efb71a7dc6e4aaf639":{"kind":"REWORK","_number":1,"created":"2022-04-11 11:15:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/837300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/837300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/837300/1"}}},"commit":{"parents":[{"commit":"65a706ff84c18a20d8d8c5147e66e4276382c132","subject":"build watcher images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65a706ff84c18a20d8d8c5147e66e4276382c132"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-11 11:14:23.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-11 11:14:23.000000000","tz":60},"subject":"[release] centos: switch to CentOS cloud SIG packages","message":"[release] centos: switch to CentOS cloud SIG packages\n\nChange-Id: Iecc4fa7e3204a4df7fd41e46e55b2895c43ddc43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2461bee0d028268c0d13e2efb71a7dc6e4aaf639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2461bee0d028268c0d13e2efb71a7dc6e4aaf639"}]},"branch":"refs/heads/stable/yoga"},"a11565fdd39e9b2609eab973dfa288f2da39d081":{"kind":"REWORK","_number":2,"created":"2022-04-11 11:38:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/837300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/837300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/837300/2"}}},"commit":{"parents":[{"commit":"65a706ff84c18a20d8d8c5147e66e4276382c132","subject":"build watcher images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65a706ff84c18a20d8d8c5147e66e4276382c132"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-11 11:14:23.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-11 11:38:29.000000000","tz":60},"subject":"[release] centos: switch to CentOS cloud SIG packages","message":"[release] centos: switch to CentOS cloud SIG packages\n\nChange-Id: Iecc4fa7e3204a4df7fd41e46e55b2895c43ddc43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a11565fdd39e9b2609eab973dfa288f2da39d081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a11565fdd39e9b2609eab973dfa288f2da39d081"}]},"branch":"refs/heads/stable/yoga"},"64e0da01dc761aae91e46702c3df998c73f94e03":{"kind":"REWORK","_number":3,"created":"2022-04-14 07:07:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/837300/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/837300/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/837300/3"}}},"commit":{"parents":[{"commit":"65a706ff84c18a20d8d8c5147e66e4276382c132","subject":"build watcher images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65a706ff84c18a20d8d8c5147e66e4276382c132"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-11 11:14:23.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-14 07:07:13.000000000","tz":0},"subject":"[release] centos: switch to CentOS cloud SIG packages","message":"[release] centos: switch to CentOS cloud SIG packages\n\nChange-Id: Iecc4fa7e3204a4df7fd41e46e55b2895c43ddc43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64e0da01dc761aae91e46702c3df998c73f94e03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64e0da01dc761aae91e46702c3df998c73f94e03"}]},"branch":"refs/heads/stable/yoga"},"2463e3edb046dcdea28e3c3662af6bccb2b4f567":{"kind":"REWORK","_number":4,"created":"2022-04-15 06:49:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/837300/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/837300/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/837300/4"}}},"commit":{"parents":[{"commit":"65a706ff84c18a20d8d8c5147e66e4276382c132","subject":"build watcher images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65a706ff84c18a20d8d8c5147e66e4276382c132"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-11 11:14:23.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-15 06:49:44.000000000","tz":0},"subject":"[release] centos: switch to CentOS cloud SIG packages","message":"[release] centos: switch to CentOS cloud SIG packages\n\nChange-Id: Iecc4fa7e3204a4df7fd41e46e55b2895c43ddc43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2463e3edb046dcdea28e3c3662af6bccb2b4f567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2463e3edb046dcdea28e3c3662af6bccb2b4f567"}]},"branch":"refs/heads/stable/yoga"},"1043badb598d96af44f3e31624feb5b9dbfe579a":{"kind":"REWORK","_number":5,"created":"2022-04-15 08:54:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/837300/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/837300/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/837300/5"}}},"commit":{"parents":[{"commit":"65a706ff84c18a20d8d8c5147e66e4276382c132","subject":"build watcher images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65a706ff84c18a20d8d8c5147e66e4276382c132"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-11 11:14:23.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-15 08:54:32.000000000","tz":0},"subject":"[release] centos: switch to CentOS cloud SIG packages","message":"[release] centos: switch to CentOS cloud SIG packages\n\nChange-Id: Iecc4fa7e3204a4df7fd41e46e55b2895c43ddc43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1043badb598d96af44f3e31624feb5b9dbfe579a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1043badb598d96af44f3e31624feb5b9dbfe579a"}]},"branch":"refs/heads/stable/yoga"},"039dec727611b910b12a96b79d66ee32e5180ecb":{"kind":"REWORK","_number":6,"created":"2022-04-15 09:04:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/837300/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/837300/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/837300/6"}}},"commit":{"parents":[{"commit":"65a706ff84c18a20d8d8c5147e66e4276382c132","subject":"build watcher images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65a706ff84c18a20d8d8c5147e66e4276382c132"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-11 11:14:23.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-15 09:04:57.000000000","tz":0},"subject":"[release] centos: switch to CentOS cloud SIG packages","message":"[release] centos: switch to CentOS cloud SIG packages\n\nChange-Id: Iecc4fa7e3204a4df7fd41e46e55b2895c43ddc43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/039dec727611b910b12a96b79d66ee32e5180ecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/039dec727611b910b12a96b79d66ee32e5180ecb"}]},"branch":"refs/heads/stable/yoga"},"9d22fc32fbe85e84376ac77180f026f1bc7ce2dc":{"kind":"REWORK","_number":7,"created":"2022-04-15 10:54:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/00/837300/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/837300/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/837300/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/837300/7"}}},"commit":{"parents":[{"commit":"65a706ff84c18a20d8d8c5147e66e4276382c132","subject":"build watcher images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65a706ff84c18a20d8d8c5147e66e4276382c132"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-11 11:14:23.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-15 10:54:10.000000000","tz":120},"subject":"[release] centos: switch to CentOS cloud SIG packages","message":"[release] centos: switch to CentOS cloud SIG packages\n\nChange-Id: Iecc4fa7e3204a4df7fd41e46e55b2895c43ddc43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d22fc32fbe85e84376ac77180f026f1bc7ce2dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d22fc32fbe85e84376ac77180f026f1bc7ce2dc"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
