)]}'
{"id":"openstack%2Fkolla~837383","triplet_id":"openstack%2Fkolla~master~I9868f73172efedf113c3cc1567098b3439e2a205","project":"openstack/kolla","branch":"master","topic":"hacluster-pcs","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-11 20:03:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9868f73172efedf113c3cc1567098b3439e2a205","subject":"Remove duplicate entry in UNBUILDABLE_IMAGES","status":"MERGED","created":"2022-04-11 17:12:06.000000000","updated":"2022-04-12 00:30:51.000000000","submitted":"2022-04-11 20:03:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837383-hacluster-pcs","meta_rev_id":"6db8c4645582e494093dacc609a2bca09fc059e6","_number":837383,"virtual_id_number":837383,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 20:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-11 17:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-11 17:13:22.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-11 17:18:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-11 17:13:22.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-04-11 17:18:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-11 18:32:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"689c60e61806c9ee25d7b8fcc2fa4f13f393ea3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-11 17:12:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f040f82a64446e03bd7d7143b7e550b40a44e7f9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-11 17:13:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e579ea7d5da5ae40ab4aaa10ee630d106639c32","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-11 17:18:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2a7fdae7ecbddcecd4bf28fbbee6ae32993bced7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 18:32:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fdc6a37992834c5b913bb13baf8822b2 : SUCCESS in 40m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6a2c6c35d1f04a05af8069c978f13ab1 : SUCCESS in 1h 16m 46s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4cbac9b4168b483686d45d684e297a33 : SUCCESS in 1h 07m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c46bcd4afce84e47af4dab1ef28eb17e : SUCCESS in 26m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/87c7c10cefe34d9daccafdf7a81af973 : SUCCESS in 1h 02m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cc8a73c99b44486191b10ca701b2a879 : SUCCESS in 1h 03m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed4fda8994c84cb1b94bbc33b8394c00 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33bfc0af7a614541b5992bcc7f9ce704 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4af83ec268b04b5fa98a52db75942503 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d291af43e02d4cb996483c506a30491a : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873dc2445ecf49d2b65a31fd2e8eb9f6 : SUCCESS in 6m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0febc5a513b54a76a3bece2726351d43 : SUCCESS in 44m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a9429ed7fee4632879ebf89a7d81e88 : SUCCESS in 43m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/636a9b0da5274e52a02e4547bba05c31 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9394e768768b101defb9e086d84a9e40a050615a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 18:33:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80fca940784db2e852e17911318ee3f7eaa48293","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 20:03:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a89c3bfee50d437f8a4a49a5fd3dd29a : SUCCESS in 43m 30s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6aa795aee51643a8b6a81fa3ac55c742 : SUCCESS in 1h 24m 53s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cd4efb7c81f7411fa54ca005ef25fab4 : SUCCESS in 1h 19m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/91f162ac9bd1478d8c8aabddab066da8 : SUCCESS in 31m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b510536a02a64d778ec0855d93821767 : SUCCESS in 1h 01m 38s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/173626dff4c04f55800796a26423f6cf : SUCCESS in 1h 09m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bc67d4c706b42bfa86292e7f2d1eba6 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c497eabecad4249a40ced6ee60f875a : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd3d5989f03446aea87e358c9785ab23 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ebeac3fed2e44b889cfaf95efbbe349 : SUCCESS in 6m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2936de31e3a344369a989219f271272f : SUCCESS in 35m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/133c71a370ea4cca93cd8feff072dd3d : SUCCESS in 56m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5bdc2ae047da43568ea22eba8e8abf54 : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c295e99a466198f7b10b4822ab9b5a985765fda5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 20:03:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b6d8e553daf31ce3fe60fa70fc317e0f12b03f21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 20:05:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0756422c54745dcb57b3d705ceb38bb : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1},{"id":"6db8c4645582e494093dacc609a2bca09fc059e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 00:30:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/433c115677964277bbdb12b0fcc77796 : SUCCESS in 2h 16m 36s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/489c64ee42fd4a1a8aed7c55dbebe081 : SUCCESS in 1h 45m 32s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/04334184cc8b454dba07bf297ac2f910 : SUCCESS in 2h 40m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d91a9e3c0974ee7ba0f126c4cf52ef3 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc4ee03edd2c4c4fa02e5c8afb7f4815 : SUCCESS in 9m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/06415c553479401d8de553cc5a6d5323 : SUCCESS in 2h 06m 24s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e2d6c31cb3075de4e4fdc0a7a47b910470233e9","revisions":{"6e2d6c31cb3075de4e4fdc0a7a47b910470233e9":{"kind":"REWORK","_number":1,"created":"2022-04-11 17:12:06.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/83/837383/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/837383/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/837383/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/837383/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/837383/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/837383/1"}}},"commit":{"parents":[{"commit":"19765c629734fe699871301be23fb308c14b2639","subject":"Merge \"for Zed cycle minimal Python version is 3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19765c629734fe699871301be23fb308c14b2639"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-11 17:11:05.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-11 17:11:05.000000000","tz":120},"subject":"Remove duplicate entry in UNBUILDABLE_IMAGES","message":"Remove duplicate entry in UNBUILDABLE_IMAGES\n\nhacluster-pcs is already in centos, no need to be in\ncentos+aarch64.\n\nTrivialFix\n\nChange-Id: I9868f73172efedf113c3cc1567098b3439e2a205\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e2d6c31cb3075de4e4fdc0a7a47b910470233e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e2d6c31cb3075de4e4fdc0a7a47b910470233e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
