)]}'
{"id":"openstack%2Fkolla~837806","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I2cbf1f539880d512aa223c3ef3a4b19ee18854ac","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-14 18:55:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2cbf1f539880d512aa223c3ef3a4b19ee18854ac","subject":"Fix image builds with sources using a type\u003dgit","status":"MERGED","created":"2022-04-14 15:44:33.000000000","updated":"2022-04-14 18:59:24.000000000","submitted":"2022-04-14 18:55:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837806","cherry_pick_of_change":837710,"cherry_pick_of_patch_set":4,"meta_rev_id":"7c2243b65bb1a7f867e456006c57a2c152fae7ed","_number":837806,"virtual_id_number":837806,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-14 18:55:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-04-14 16:25:59.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-14 15:44:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-04-14 16:25:59.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-14 15:44:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-04-14 16:25:59.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-04-14 17:34:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2d22ed250f83fba535d57d723e911cfb3d0452f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-14 15:44:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"50565d09b1f6738ce73f0192e4d2dbbdd46aba6b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-14 15:44:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"387c86b6c66c9fd789984c397ed0811e2749e740","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-14 16:25:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a6457210ecd00476e4799a83a7090aff005a77e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 17:34:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/caae2ceca4d245a2ac1cf20094e21a6c : SUCCESS in 44m 34s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/06f0a6ef2a3043fdb4d9df11d8cb391a : SUCCESS in 45m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e7bb0b8a41ae42448b4c753932a3791f : SUCCESS in 1h 34m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/110e9c0268744b0db77bc7731a3963cb : SUCCESS in 1h 29m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9d2f4c3d3b0348249623fda84e49efaa : SUCCESS in 1h 44m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/99b0e9db11dc496a96c8e277c36dda46 : SUCCESS in 36m 06s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/116a1f16b4f3413b864caaef9f92d6c0 : SUCCESS in 22m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7fb85cab5bf146f88ea44e1f83090e15 : SUCCESS in 42m 05s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5bd83d8e8a6d4e7fb978ac514a54b416 : SUCCESS in 1h 07m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c78b5ad6acaa49fbaa24915ecb52ab27 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fb3fa348a14470f9fa0d55d6351143d : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d1dae53b5304823baa359a3ec6bbec7 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cefcb08237494d359b0d52c3042c5e04 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1b924daaf4f4344a615a7f32dfca7bd : SUCCESS in 3m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e16c381eeb946b98ae37dda772ca31e : SUCCESS in 23m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe6f0ec961dc4d3899ff11e7f2955e63 : SUCCESS in 47m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d17d16c761294efca236572d74aafc04 : SUCCESS in 54m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/22358e9896b64e5c9506766d848ce7a5 : SUCCESS in 49m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/69ec92026e3e44db8afb1e9617c7e37d : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ba2291b09997130ed83c3d737c3cc8c7afac42e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 17:35:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4162aa80c66e73c6f6553ae7a4c599e3e9fd2bad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 18:37:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ec43d453b2f243a3b20124854872814a : SUCCESS in 2h 49m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/34a3e9d1ab934b0e923d8c58f5503dc4 : SUCCESS in 2h 19m 55s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6dc8b782f5fe4515b40a4f5eab702284 : SUCCESS in 2h 29m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0692a904a58e4297a71661e05d588c18 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b09204be60884e73aab33a5d8939c42f : SUCCESS in 7m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c178069c3cd648cdabe8e7999ddb00b6 : SUCCESS in 2h 10m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c7d2cb406f8110b5fea426ff8898dbbf67c0c8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 18:55:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fcc1505031e74bbd91ea7025b69c9825 : SUCCESS in 50m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e29c1409e7084fea8923d6ba6bed3dfa : SUCCESS in 1h 14m 26s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4100c88e5c67488f9666492e55c73fa2 : SUCCESS in 1h 19m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/62c0848172904dc3bc7563b3275d1960 : SUCCESS in 33m 06s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a8396fa21b30437c80fb1ac405781bcd : SUCCESS in 1h 05m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/13fb0e937f91478394d9548d432e8169 : SUCCESS in 1h 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a41c31d0edf04e299bc405a35be74b79 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01075c2629de433ca28f7bcc6db3ebca : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e74ead6240044619919c39a414d3d88 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8429479647e24cf8b822708d8df212de : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e7231b2b6804b61b10defc5a9661d9a : SUCCESS in 4m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4d334b720f34599909bcad95dd05811 : SUCCESS in 34m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c511400a3b544f779d38fb650fbf0ea8 : SUCCESS in 51m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a5f49eea646b440e8e676b5eb21769b1 : SUCCESS in 1h 08m 54s","accounts_in_message":[],"_revision_number":1},{"id":"61e609d5791247479b1dd393806597070eed144e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 18:55:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7c2243b65bb1a7f867e456006c57a2c152fae7ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 18:59:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa7d9119597549f6829ba931ec0827e8 : SUCCESS in 2m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/798d16667cb5494a93442a3f0723873b : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1822dcc0dc71680763d54cfeba09b5579a93a02e","revisions":{"1822dcc0dc71680763d54cfeba09b5579a93a02e":{"kind":"REWORK","_number":1,"created":"2022-04-14 15:44:33.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/06/837806/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/837806/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/837806/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/837806/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/837806/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/837806/1"}}},"commit":{"parents":[{"commit":"65a706ff84c18a20d8d8c5147e66e4276382c132","subject":"build watcher images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65a706ff84c18a20d8d8c5147e66e4276382c132"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-13 18:19:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-14 15:44:33.000000000","tz":0},"subject":"Fix image builds with sources using a type\u003dgit","message":"Fix image builds with sources using a type\u003dgit\n\nA recent change to git [1] introduced a new behaviour to work around a\nCVE [2] that disallows any git operations in directories not owned by\nthe current user.\n\nThis may seem unrelated to installation, but it plays havoc with PBR,\nwhich calls out to git to get to get revision history.  So if you are\n\"pip install\"-ing from a source tree you don\u0027t own, the PBR git calls\nin that tree now fail and the install blows up.\n\nWhen using type\u003dsource, kolla clones the repository, then creates a\ntarball from it, which is ADDed to the image. The ownership of the files\nin the tarball is preserved, which in this case will be the user running\nkolla-build. Since the Docker build runs as root, we hit the PBR issue.\n\nOur solution is to make sure that any tarball we generate from git\nsources have all files owned by root:root so that the root user is able\nto use git commands when building container images.\n\n[1] https://github.com/git/git/commit/8959555cee7ec045958f9b6dd62e541affb7e7d9\n[2] https://cve.mitre.org/cgi-bin/cvename.cgi?name\u003dCVE-2022-24765.\n\nCloses-Bug: #1969096\nRelated-Bug: #1968877\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\nChange-Id: I2cbf1f539880d512aa223c3ef3a4b19ee18854ac\n(cherry picked from commit c4fda7baa3ffc36b555c32a34a00042b6035b917)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1822dcc0dc71680763d54cfeba09b5579a93a02e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1822dcc0dc71680763d54cfeba09b5579a93a02e"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
