)]}'
{"id":"openstack%2Fkolla~837910","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I2cbf1f539880d512aa223c3ef3a4b19ee18854ac","project":"openstack/kolla","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-06-23 18:42:12.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2cbf1f539880d512aa223c3ef3a4b19ee18854ac","subject":"Fix image builds with sources using a type\u003dgit","status":"ABANDONED","created":"2022-04-14 15:45:11.000000000","updated":"2022-06-23 18:42:12.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":837809,"cherry_pick_of_patch_set":1,"meta_rev_id":"d33dc1fe87f8bb80afe9c615f30d30ece9091a5c","_number":837910,"virtual_id_number":837910,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-04-14 17:16:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":-2,"date":"2022-06-21 05:18:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-04-14 16:25:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-14 15:45:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-04-14 16:25:24.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-14 15:45:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-04-14 16:25:19.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-04-14 17:16:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-21 05:18:47.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"236edb4ec4018bc86bee13614e30825761b31d10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-14 15:45:11.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"56e10f605b335dc8e89656520b4a0f3b81b61373","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-14 15:45:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9778192d3c063c92e335060b91e2cd7ace8d10e8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-14 16:25:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9e9ffaed00638c4de0bf9eab691a3a050c87d669","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-14 16:25:24.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c29ea45255759d6e6fa6847d32c89af26f755b51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 17:16:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ef7c04b53e81458c8fae38ab4bff2193 : FAILURE in 32m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/59b83437d07145cc80a38b6e6bd5d7b9 : FAILURE in 30m 31s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d7689f9c715642328bd57f4c48260ecb : SUCCESS in 1h 05m 27s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d437317768e34980b6badef15934546b : SUCCESS in 1h 10m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bd7bcb68677d4f77a6ac0cc97f11da0c : SUCCESS in 1h 23m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c7f0766166d94890b5f4bb646b84de14 : FAILURE in 3m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/aa9100a2c7594ce283220d8c4da4feb3 : FAILURE in 3m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b6b8543c1c3a4851985e70d559878efd : FAILURE in 9m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/066f9876ed4647a098c0007c983927e5 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e1e2de1275445a481398d90024cf54e : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab3d75a3a1af461686ac9b98a7330694 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4686778c1d4d0fbdd9e797ad24b438 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d06ac00ac4904722bae39bc9d04bfb42 : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e50767e327440b495db0d6f2de53107 : FAILURE in 3m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce161ee9375a40998490bbe7b4cd93e3 : FAILURE in 4m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b97687ae92994c068fbbd67cbd0fbad5 : FAILURE in 8m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8dc499b6eaf4cbe8f2733c8eba666ad : FAILURE in 9m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d88cbacdca6a4f80a76c015b5b444eab : SUCCESS in 57m 28s","accounts_in_message":[],"_revision_number":1},{"id":"68b9d08a28e26f3bb1f24a618b31b84170b2ea34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 17:44:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/37e0525518264acc9104ee2eccda60d8 : FAILURE in 1h 07m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b6aed55c597429ea16ee1619caa72c6 : FAILURE in 8m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/56145e4a869b4795818dfc32d65a65c2 : FAILURE in 16m 46s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f08f0fc58603431db225e678466b30ed : FAILURE in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"208b7d55451629ee3946d80f01dcfabce1cb91a7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 09:12:43.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/kolla/+/836591","accounts_in_message":[],"_revision_number":1},{"id":"9d50b62b8c920a58669899273b21b1beee0b3a59","tag":"autogenerated:gerrit:restore","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2022-06-20 19:30:18.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"88a39028cb931dd3ffbb2126c26195a9e6ba4ff3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 20:41:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f556af924a3a4a42a5c6667dda82ee9b : SUCCESS in 1h 04m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/09bd989b0ee84ea3b0cd7b7512a2b422 : FAILURE in 53m 16s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6298bfb5e2244d00983adf12b2270b10 : FAILURE in 35m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bb13d2a537a14e58a08d818ff8c11424 : FAILURE in 47m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08038e40b8c5390bcf6f26dccff88c49aa2b48f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 20:48:11.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/298647667b8947b191c8af8458e2bbe9 : SUCCESS in 34m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/595bbae2914845dc8789d7bbf1bb2042 : SUCCESS in 35m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fe329c06bf6d4b16b940c8c52a546fbb : SUCCESS in 1h 00m 02s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c69167d13a46445aba442ab6d266a668 : SUCCESS in 58m 51s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/31e855baa73f479dace43c70fc532350 : SUCCESS in 1h 09m 06s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/25a8a75ce86b4a379b55be8c7ed070ae : SUCCESS in 20m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b19e7732aac8467ca599aa693d654cd8 : SUCCESS in 19m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3a9207ffd2134954aab25421cd2259fc : SUCCESS in 55m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70ee43b3704a4918b23d66661795e80d : FAILURE in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98d5fcab9b4c433c8b0d4eb06f894c2b : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51f6dc9de57c407d8cabb60092dc5caa : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2fbe6aa6104016a4c8bbed4d91da79 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45511e882dc74710890a0455800b269d : SUCCESS in 3m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/69906f083c3746be9624117da6e426a0 : SUCCESS in 32m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1989be99382b4a6f882ef50f33b38639 : SUCCESS in 33m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92bf775bbbd2492caeb7b18100d390a5 : SUCCESS in 52m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fab0a2cdaf9f4af393a7e40a24c628bb : SUCCESS in 48m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cea4b851571044f58c8d8b251e10ed37 : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fc854fa5eb156d88f7ec58d67e4ec3d90f25f84e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-21 05:18:47.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d33dc1fe87f8bb80afe9c615f30d30ece9091a5c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-06-23 18:42:12.000000000","message":"Abandoned\n\nyeah, we forgot to abandon","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90141fd98f75f5fa621e04016934f56c05876892","revisions":{"90141fd98f75f5fa621e04016934f56c05876892":{"kind":"REWORK","_number":1,"created":"2022-04-14 15:45:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/10/837910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/837910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/837910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/837910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/837910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/837910/1"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-13 18:19:38.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-14 15:45:11.000000000","tz":0},"subject":"Fix image builds with sources using a type\u003dgit","message":"Fix image builds with sources using a type\u003dgit\n\nA recent change to git [1] introduced a new behaviour to work around a\nCVE [2] that disallows any git operations in directories not owned by\nthe current user.\n\nThis may seem unrelated to installation, but it plays havoc with PBR,\nwhich calls out to git to get to get revision history.  So if you are\n\"pip install\"-ing from a source tree you don\u0027t own, the PBR git calls\nin that tree now fail and the install blows up.\n\nWhen using type\u003dsource, kolla clones the repository, then creates a\ntarball from it, which is ADDed to the image. The ownership of the files\nin the tarball is preserved, which in this case will be the user running\nkolla-build. Since the Docker build runs as root, we hit the PBR issue.\n\nOur solution is to make sure that any tarball we generate from git\nsources have all files owned by root:root so that the root user is able\nto use git commands when building container images.\n\n[1] https://github.com/git/git/commit/8959555cee7ec045958f9b6dd62e541affb7e7d9\n[2] https://cve.mitre.org/cgi-bin/cvename.cgi?name\u003dCVE-2022-24765.\n\nCloses-Bug: #1969096\nRelated-Bug: #1968877\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\nChange-Id: I2cbf1f539880d512aa223c3ef3a4b19ee18854ac\n(cherry picked from commit c4fda7baa3ffc36b555c32a34a00042b6035b917)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90141fd98f75f5fa621e04016934f56c05876892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90141fd98f75f5fa621e04016934f56c05876892"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
