)]}'
{"id":"openstack%2Fkolla~838357","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Ideaf66ce18c993802f4d20f490caa47cec0dde5e","project":"openstack/kolla","branch":"stable/yoga","topic":"quiet-kolla-stable/yoga","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-19 11:43:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ideaf66ce18c993802f4d20f490caa47cec0dde5e","subject":"Revert \"CI: add templated Dockerfiles to build logs\"","status":"MERGED","created":"2022-04-19 08:20:25.000000000","updated":"2022-04-19 11:46:30.000000000","submitted":"2022-04-19 11:43:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838357-quiet-kolla-stable/yoga","cherry_pick_of_change":838014,"cherry_pick_of_patch_set":2,"meta_rev_id":"e562f9020978d810a744664382455efeda82fba1","_number":838357,"virtual_id_number":838357,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-19 11:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-04-19 09:28:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-19 08:20:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2022-04-19 10:02:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 09:28:53.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-04-19 09:51:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d65161cf4fa944f5f11644a646da84c65a8071a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 08:20:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a7c0f5c1c56df3351f25ae87a36f6506e498ed14","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 08:20:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"88be95bb3e4ca0587513d13619123fba6664af6e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 09:28:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"89b15ac058a567bbca9477db53633101e067bcf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 09:51:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/493334f8ef3e4e83b76c3dec82195dca : SUCCESS in 36m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/714f8e7c566d4b59b041f9777b150dbb : SUCCESS in 32m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0fa84871e7734be3a4008d363dbf07d7 : SUCCESS in 1h 11m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fbd1bff6a758476d827b05123cad6f9e : SUCCESS in 1h 06m 00s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e793b74b53284f9987bedcaec3e24dc1 : SUCCESS in 1h 28m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5bd149ac703449f3a72bdf3b693375e7 : SUCCESS in 26m 14s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e42007d59f314dcc859c8ebbab3292c3 : SUCCESS in 20m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d366539f0acc476c8a7d3c26e4a5307b : SUCCESS in 1h 03m 30s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5e86a4f7000d46d8aa326002c354f16f : SUCCESS in 1h 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa3e22ba3564bf8af7b75d24c41e656 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dd4c48e3ceb476185cc50229647f423 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f471cf7d7bf34914a8a8dd5e1ac03c20 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e7f88a491a7414696415ece19af8ead : SUCCESS in 7m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7fa8e772c5104b14a342b10a5c5ac5a5 : SUCCESS in 19m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fca21d113ddc4d8dbad6334417c7b426 : SUCCESS in 28m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/15e5bb96d1b34222a84769d37e6942fb : SUCCESS in 48m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c4124fa348194fae9bca7cb0fd8f14dc : SUCCESS in 53m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fb902a35674e4208b260d394d3c59e7f : SUCCESS in 1h 00m 59s","accounts_in_message":[],"_revision_number":1},{"id":"67384cbe31c5ecbd97cd28a0dfcbae3898be360a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 10:02:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a5cf8c4e387897074ead7846314b8f8c40c855b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 10:03:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"487114c1a65aa616bea57acf97c253d66aec5057","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 10:36:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e44ef76f712f43dc9d24a55fbf293a61 : SUCCESS in 1h 41m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/124317f515db45d2a099e337328dc6a6 : SUCCESS in 1h 25m 15s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a67f91161fcd4a308ef985622823a2b3 : FAILURE in 1h 48m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66ccfb0525c34f27ae7c0b13dad434c2 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df2b2d35efe14b61b1bfbf5a766f0fdd : SUCCESS in 7m 21s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/71f243cbee11441882e36d914e2d3c9b : SUCCESS in 1h 04m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ba12b8a8fc67542fa2550454b2da973ec7c074dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 11:43:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4892617e043a4ae0b34346a16f95b7a2 : SUCCESS in 28m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/34e787cc6cd14b2aaf8aad240996573a : SUCCESS in 1h 35m 07s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/04cc0865fce843e8bbf30b7246a35deb : SUCCESS in 1h 27m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9a99b1781cb04e3d99e55c549897addb : SUCCESS in 27m 50s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/409de3e4256f4240a8a6dc99528eb17d : SUCCESS in 1h 01m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/390cccf811324269a7cead851e5a6084 : SUCCESS in 1h 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1200020376fd45e5af6ef04157ec9763 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd6f08b4b20f41f5a6cf92e3d509f951 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85cd6939199c4778af83f7ccf9da7261 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7615377f369a4f409cdb3c4b7c0d4c0b : SUCCESS in 7m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6ab77ee4b014c60b7ccd753e11d85b1 : SUCCESS in 21m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bfcaaf7397da44b39d7d146305dda4f2 : SUCCESS in 48m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/80468d9eb2604a3ba940c046d5ee272d : SUCCESS in 1h 03m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1643af84a2629920ecaf8e5f1e8bf8be7edef9c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 11:43:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e562f9020978d810a744664382455efeda82fba1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 11:46:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbdeac944b1c4052862a11d435e17d48 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bda93f42b8ded43c48aaeccc76c82eac989aa2c","revisions":{"2bda93f42b8ded43c48aaeccc76c82eac989aa2c":{"kind":"REWORK","_number":1,"created":"2022-04-19 08:20:25.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/838357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/838357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/838357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/838357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/838357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/838357/1"}}},"commit":{"parents":[{"commit":"1822dcc0dc71680763d54cfeba09b5579a93a02e","subject":"Fix image builds with sources using a type\u003dgit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1822dcc0dc71680763d54cfeba09b5579a93a02e"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-15 09:58:14.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-19 08:20:25.000000000","tz":0},"subject":"Revert \"CI: add templated Dockerfiles to build logs\"","message":"Revert \"CI: add templated Dockerfiles to build logs\"\n\nThis reverts commit 09c890b12463a4b2c8767231189e7016f337fbc3.\n\nReason for revert: That change caused all Kolla\u0027s work items to be\ntransferred. This includes tarballs, their extracted contents, cloned\ngit repositories, etc. to also be transferred to swift which takes\nextra time, pollutes logs and decreases the stability of jobs slightly\n(the mechanism is not designed for this kind of load).\nThe collection of templated Dockerfiles in CI has to be reimplemented\nif it is wanted.\n\nChange-Id: Ideaf66ce18c993802f4d20f490caa47cec0dde5e\n(cherry picked from commit f33ef03977536ecb29ea464b9e557a6caacf4345)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bda93f42b8ded43c48aaeccc76c82eac989aa2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2bda93f42b8ded43c48aaeccc76c82eac989aa2c"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
