)]}'
{"id":"openstack%2Fkolla~838362","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~If7d5c2807f0947a8bbbc1ceb8531c9b9c9287c1f","project":"openstack/kolla","branch":"stable/victoria","topic":"quiet-kolla-stable/yoga-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-19 16:12:27.000000000","reason":"Marcin Juszkiewicz replied on the change"}},"hashtags":[],"change_id":"If7d5c2807f0947a8bbbc1ceb8531c9b9c9287c1f","subject":"enable logging to file for quiet mode","status":"MERGED","created":"2022-04-19 11:38:50.000000000","updated":"2022-04-19 23:44:55.000000000","submitted":"2022-04-19 18:44:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838362-quiet-kolla-stable/yoga-stable/wallaby-stable/victoria","cherry_pick_of_change":838361,"cherry_pick_of_patch_set":1,"meta_rev_id":"fd61560afc4656d0d149d81a46b088e1c898b0bf","_number":838362,"virtual_id_number":838362,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-19 18:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-19 14:42:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-19 12:38:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-19 14:42:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 12:38:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-19 13:13:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-19 14:42:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2578beb6f7e69f8b7695eafbbf66d6c369ea52f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 11:38:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"5560dee586f6bfe8392925ac9f0ba13ba0bfd510","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 12:38:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"642919bfd64e72f2101d9cc8459e01cb2803759a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 13:13:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/711f3be314e14a248ae8ec293e4e3129 : FAILURE in 37m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/88e9c3f9354c4ebc9e81c50c172c0159 : FAILURE in 33m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0c67190f71e244828683de84bb4e702a : SUCCESS in 1h 11m 00s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9c3c7be0e24742ac97c7d68df008de63 : SUCCESS in 1h 18m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/01469f63e9a748059104dbc8a9775e7d : SUCCESS in 1h 21m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dfe951495a494510a5fcebac67bbfffe : FAILURE in 3m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c83d33ae685e408382f527ae1230aa79 : FAILURE in 4m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2d15bc1362534cf6baaf649ef0145c22 : FAILURE in 9m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9267442f2dde4116be9d26bf5ea6284b : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cd9d2b974c9434b979acd108a9dadb2 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/286c640f808547a0a6c9ae1c906b4c65 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6573c3953054caca7c2b461684b54fd : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10665f4446d848df836ac6cc26e14591 : SUCCESS in 4m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ce7118b7f4d48a888dffd3a56fac954 : FAILURE in 4m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e27a216032e4d5195b39d00a310dd16 : FAILURE in 3m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d5cf9d5f41047e996412707bd6371e7 : FAILURE in 8m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f324625f7b3249f19867cacf8d180fdd : FAILURE in 7m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/caa6c1bfdc0e4ac686b6ef2115f922f1 : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":1},{"id":"394740a5a5951b80fe1bed2f9f338930a0ea9415","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 13:13:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e25f81f2de0803698f1b87540b9e3223d6a92cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 14:29:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fa0fa613213d41dbbe1b398f812f4144 : SUCCESS in 29m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ae999117c01b41b098ffe549b94c22a2 : SUCCESS in 27m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ff17d92ed7414d88804e8bd42a57ada2 : SUCCESS in 1h 00m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/959311b1fb8441da9d6bba73f14beae5 : SUCCESS in 1h 07m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/39dfe8bb4b84451bb255d23ed2377aee : SUCCESS in 59m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9a15e720f5a64a0c8b2d97e9376e86ea : SUCCESS in 21m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5f283b26557345acab7f85536398ded9 : SUCCESS in 18m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/630482ef4aac49f3a477f28446aac9bf : SUCCESS in 1h 09m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/454b199a5ac24429b2dbfa47bf8bccb0 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fa4bd66f76b461f8c8008317572220a : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6970a250f460447f9016148d922829cb : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16f0293ff0224cfeb7be6228f65fcd16 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38fef50ba7cb43a19b97c2496193fba7 : SUCCESS in 4m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6b926427159a4e3ebf6c2e8d30b8380e : SUCCESS in 20m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/570b1b472f4d4b9aa8fff7982dd681ef : SUCCESS in 23m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3d5e4c6408546019ff20a6dee3583a3 : SUCCESS in 51m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b879a9635d0942edaf4371818e068066 : SUCCESS in 48m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e8f61aed19ae497fb343eb378ae4b2dd : SUCCESS in 1h 05m 44s","accounts_in_message":[],"_revision_number":1},{"id":"989e7c6481408bfec249013d930edaab55b0548f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-19 14:42:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"96ef2a6b3bc3b710a0d2ab3f78a957893cbe3b56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 14:43:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec3d0e702c7e75e730d5fbd92704c15642664344","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 15:54:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6ff9fb20023e485bbe7668a53299fb2e : SUCCESS in 28m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/21e6ce4497da4fee9eba3dfaa38a5ab0 : SUCCESS in 57m 37s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ddaf52b2b7bd40aaab499345e9427b78 : SUCCESS in 1h 05m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4d3237379a74770891d911a46d8badc : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa118cae981e4a02a93862beeb91e154 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcd914afd65b44608de1e88a04c44559 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3b7e65eda543d6ab6d9b3b2de7e337 : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75b00bd77529462588fb08fff30c842b : SUCCESS in 4m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c21e9775e70427bbabff04d92927a2f : FAILURE in 24m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de54a8bef5c64f82a4213f7b699a6ec6 : SUCCESS in 51m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/171d40b342bf4de29f76c0da44af8fe8 : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1127f5926322be0df8bbe053938338d6990efb44","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 16:12:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84496dbd7c1ece496b59d9b32a4aa764f50823ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 17:34:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d67b35742c8a4d478f28241995378c99 : SUCCESS in 39m 01s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/57023976f7364aa485073a803c6dad55 : SUCCESS in 38m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c5b243d1a73c422a8215132437bc0d66 : SUCCESS in 53m 34s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7da8dfaeac9a4417a5069401aa91b173 : SUCCESS in 1h 06m 48s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/961fcb8eb4a0491387e356fbc2eeefb6 : SUCCESS in 1h 13m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/892edb83eeec4c3993c7c4cdbdbce451 : SUCCESS in 32m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/83ddfc53bc9843adaa147adb589598b2 : SUCCESS in 16m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0e8311ee595c4208889bbe37087260e3 : SUCCESS in 45m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec303c1e601c42a09d626c48f3aa5f53 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a13c56091d84dbfbbb958593a0e9bfb : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f740467e1295403b860f244937b758f6 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/add60b575b7c4933a94d1e44dbbbc8ee : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90bdfcf3593c4da3887cf20b03b6a525 : SUCCESS in 4m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b871892fb2924f53baf776f0981dcf2a : SUCCESS in 20m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e61048a860874fd680dd2287b2eb12e3 : SUCCESS in 24m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/60af5e19748243bcb12d7e29eb756968 : SUCCESS in 56m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d316db625db94c62b5564c43c58cfc72 : SUCCESS in 52m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9c5b400786a0457a8b8493bd0cd1c773 : SUCCESS in 54m 42s","accounts_in_message":[],"_revision_number":1},{"id":"789a42e5e2c82ed0cac6aaafa76ba58b329bd71a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 17:35:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f46ed99b6487715ee31c7c1f4a06871697fd1447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 18:44:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a13b55ec0e1d4afa965c69813d47a4b4 : SUCCESS in 27m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c28d1d69e64648ab8d2e49a16075ef15 : SUCCESS in 59m 15s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2d518ba2ff9741f6ba79d44bd2c33e8e : SUCCESS in 1h 02m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd30e54597d4068a97aab628536247d : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a66c52577b14980b3aa471d9e3d3db6 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd3d6abc17fd4322ab66839cd27fc503 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbea26fc3b9d471e900372bbe1b76448 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/843b076300174c2f95fed044699b33dd : SUCCESS in 4m 01s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c6f7af7b1f464777af6aa33d578bcfef : SUCCESS in 18m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d495e4079ca4b46ab1cafd209f539b2 : SUCCESS in 45m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c4dba35a3791401cbf89707c86c56df1 : SUCCESS in 1h 04m 54s","accounts_in_message":[],"_revision_number":1},{"id":"90f296f8b412c6cd1427187dd5eab1f2f4ffb987","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 18:44:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c68a399baa7ebca879d3044db840fc2484d155d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 18:46:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b9313cc83da45d78bcd52484284592d : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a9b3b0422fc4ba79f0cdc3e1824644e : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1},{"id":"fd61560afc4656d0d149d81a46b088e1c898b0bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 23:44:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9786681bf39e4e46b2e16847e816f7ae : FAILURE in 2h 36m 25s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/32b5a137381c4bb7971a8797a11d3bb2 : FAILURE in 6m 38s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0599ab6b54d94ef599a3f04f81de86cd : FAILURE in 15m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e63d502c65c463a93284412fcb53ad5 : FAILURE in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a83002b3faaba9f5803eb045eb7613417aea9b8c","revisions":{"a83002b3faaba9f5803eb045eb7613417aea9b8c":{"kind":"REWORK","_number":1,"created":"2022-04-19 11:38:50.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/62/838362/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/838362/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/838362/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/838362/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/838362/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/838362/1"}}},"commit":{"parents":[{"commit":"3819c16a9d029c6bb18ebe1e80464e49d8581625","subject":"Merge \"Install glusterfs-client in Debuntu\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3819c16a9d029c6bb18ebe1e80464e49d8581625"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-14 13:32:47.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 11:38:50.000000000","tz":0},"subject":"enable logging to file for quiet mode","message":"enable logging to file for quiet mode\n\nWe have \u0027quiet\u0027 mode where only minimal data is printed to the console.\nBut there were no logs at all then.\n\nThis change generates log files (if \u0027logs-dir\u0027 argument is used) during\nquiet build.\n\nAlso enables \u0027quiet\u0027 mode for CI so Zuul will not have to parse 29MB\nJSON file each time.\n\nChange-Id: If7d5c2807f0947a8bbbc1ceb8531c9b9c9287c1f\n(cherry picked from commit 0cf5b1d4e6e871562c0ca0f48a21594f8d197471)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a83002b3faaba9f5803eb045eb7613417aea9b8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a83002b3faaba9f5803eb045eb7613417aea9b8c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
