)]}'
{"id":"openstack%2Fkolla~838368","triplet_id":"openstack%2Fkolla~stable%2Fxena~If7d5c2807f0947a8bbbc1ceb8531c9b9c9287c1f","project":"openstack/kolla","branch":"stable/xena","topic":"quiet-kolla-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-19 16:45:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7d5c2807f0947a8bbbc1ceb8531c9b9c9287c1f","subject":"enable logging to file for quiet mode","status":"MERGED","created":"2022-04-19 11:44:40.000000000","updated":"2022-04-20 01:37:56.000000000","submitted":"2022-04-19 16:45:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838368-quiet-kolla-stable/yoga-stable/xena","cherry_pick_of_change":838360,"cherry_pick_of_patch_set":1,"meta_rev_id":"04501f96e312bac6310cf0ffe453ebf3afb9980d","_number":838368,"virtual_id_number":838368,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-19 16:45:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-19 14:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-04-20 01:37:56.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},{"value":2,"date":"2022-04-19 12:40:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-19 14:42:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 12:40:48.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-19 13:22:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-19 14:42:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-20 01:37:56.000000000","updated_by":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"reviewer":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"state":"REVIEWER"}],"messages":[{"id":"7d043496dccefe65eb62356bd4a9c082a6384350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 11:44:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"8e71703115eac733101f0d9212fb9cee5426e928","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 12:40:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"62e796187fab96db2a0a408de253a9076a117f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 13:22:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/bd66e3fa8f8a40f7b418d8ae3bce1db3 : SUCCESS in 30m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/19b3f198b92d4338a1de0f52b73d34fc : SUCCESS in 52m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/00098769819942cb80d567200d6c0192 : SUCCESS in 1h 19m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fc3eb0bbff3a480183ba196237013846 : SUCCESS in 1h 10m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3876521507a94b8d8fc88db072d12dbe : SUCCESS in 1h 34m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/33153d9f023e492086ccb0134eb22a46 : SUCCESS in 29m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1bd315bb14f04670a7415180c77a7182 : SUCCESS in 20m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/be8a94c3ba4f4e46b0be739912d0296c : SUCCESS in 47m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eadf19e9e6b49669c59bc43a1194bf0 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0c692975d7a4c239d315bbaebd42357 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/129717592b804043a885d994c5540a63 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/561127d7c47141caaa4b6b59532df01f : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d587b3f145bc45d9a49b0a2bf047bbdd : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b85f2e5490d2403e96ab311f011770b7 : SUCCESS in 5m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2a4e2f3f87a48be960f9bb4a6556ac6 : SUCCESS in 23m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/98c60566992b4a22816f9eb47b114a93 : SUCCESS in 32m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36b833f8ca5042029e01e17c91ee0008 : SUCCESS in 55m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/df6c8877b7dc4033805cd3ba66914ca8 : SUCCESS in 55m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1ecf1be74406431b9b6b9cd0690c4c9f : SUCCESS in 1h 11m 33s","accounts_in_message":[],"_revision_number":1},{"id":"03cd9d79e531457e364589e3374be12cb9f33bc7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-19 14:42:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a8ffd8b4252a59fadcee3e67d93b1474b29b3f4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 14:43:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f7f25a963c77852673fcb071afb1c3d6463d701","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 16:45:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fd9fbea01af345b59398b16bfa858e95 : SUCCESS in 26m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9d0aab3889094c929adcea1e410b4c70 : SUCCESS in 1h 08m 39s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4c039bc196854f4bad43f208a280be13 : SUCCESS in 1h 32m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/854f1df0f7784c6dae83321bdbb88ca4 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01d7ab3c27c24a04a8653a1766d2c905 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8790d7e783764acebb9b1928f564d5d5 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a156681e754e3d836a826a0fc87c9a : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6c3a4f4e53641c7b8d8f733027625a1 : SUCCESS in 4m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b1e687293667439db38f1e4076fc3879 : SUCCESS in 20m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/75017a951e2847e1b9d37efc08f78477 : SUCCESS in 50m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/628e239b65f547eb9e47c5b6e60a89fa : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":1},{"id":"93726b4ba28a2044599488bcfb2b2c4a9c7fde98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 16:45:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0bec323612dbe3e2f1b79e63c72528551a8e619c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 16:47:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417c0ad84f6744f2bdd806b99233de68 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6e44740129c47f5a35d3c1c6fd82775 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f96ac57001436ef522e48030b988f986684a8849","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 01:07:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/42f827c1ad6e48dca64a1a30386b3a3f : SUCCESS in 2h 30m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8d7d22046d61480daa5e54aa446ac283 : SUCCESS in 1h 59m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9000217df7bc4fb9a36518bf9a706950 : SUCCESS in 2h 11m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f188409dcf840e1867961c4c5e4f256 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8b4881c3aac4dc5b8dc1fd02c4fb78b : SUCCESS in 7m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1581cb83887b476ea977289ec508cbd8 : SUCCESS in 2h 09m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04501f96e312bac6310cf0ffe453ebf3afb9980d","author":{"_account_id":33786,"name":"HanGuangyu","email":"hanguangyu2@gmail.com","username":"hanguangyu"},"date":"2022-04-20 01:37:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e3381b0e2ed1cfae1b7b4035c0fa339c2aad779","revisions":{"6e3381b0e2ed1cfae1b7b4035c0fa339c2aad779":{"kind":"REWORK","_number":1,"created":"2022-04-19 11:44:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/68/838368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/838368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/838368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/838368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/838368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/838368/1"}}},"commit":{"parents":[{"commit":"6b88dc0f09cab3b3a439b2ba6c5c38bf1f8b720d","subject":"Fix image builds with sources using a type\u003dgit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b88dc0f09cab3b3a439b2ba6c5c38bf1f8b720d"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-14 13:32:47.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 11:44:40.000000000","tz":0},"subject":"enable logging to file for quiet mode","message":"enable logging to file for quiet mode\n\nWe have \u0027quiet\u0027 mode where only minimal data is printed to the console.\nBut there were no logs at all then.\n\nThis change generates log files (if \u0027logs-dir\u0027 argument is used) during\nquiet build.\n\nAlso enables \u0027quiet\u0027 mode for CI so Zuul will not have to parse 29MB\nJSON file each time.\n\nChange-Id: If7d5c2807f0947a8bbbc1ceb8531c9b9c9287c1f\n(cherry picked from commit 0cf5b1d4e6e871562c0ca0f48a21594f8d197471)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e3381b0e2ed1cfae1b7b4035c0fa339c2aad779"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e3381b0e2ed1cfae1b7b4035c0fa339c2aad779"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
