)]}'
{"id":"openstack%2Fkolla~838491","triplet_id":"openstack%2Fkolla~master~I29c9d51f4b518a4e2d3157ab5d6cc434fb83f2ef","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-19 18:50:28.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-19 18:50:28.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-19 14:18:58.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I29c9d51f4b518a4e2d3157ab5d6cc434fb83f2ef","subject":"add a way to not have \u0027install_type\u0027 in image name (temporary)","status":"MERGED","created":"2022-04-19 11:23:30.000000000","updated":"2022-04-20 02:20:22.000000000","submitted":"2022-04-19 18:50:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"838491","meta_rev_id":"a926286585c2968442c12306b41efe12e1ec8fcf","_number":838491,"virtual_id_number":838491,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-19 18:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2022-04-19 14:29:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2022-04-19 17:38:16.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-19 11:27:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-04-19 12:42:37.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-19 13:11:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-19 15:27:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"}],"messages":[{"id":"c030c8e2e36890cd30826244b381125f071296b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 11:23:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13e0d2d0127bf2648d6382ae6627682b9e260ebb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 11:27:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c5ddd61189f1987057a2f4c85f9a455de4674eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 11:43:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d8805aa9bf795f9620697634f609b22bdab013a0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 11:43:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd34dcb5624e7ed6c75f297fbc0bce0fd7aa95e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 11:58:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"25dc9bb05b58b000100cf29213bb9135933d2247","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 12:42:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1b854c57f1ab79ad1ee460c15a20af8f6203c8b8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 12:44:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1315133be07480971138a5e28c3baf4244538046","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 13:11:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/16f58a0b1aeb4de781004cf813aace07 : SUCCESS in 25m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e0a678df88834d5584150aecb7549dd8 : SUCCESS in 1h 04m 33s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3cb01b5a640c4ce2b685f258f450b317 : SUCCESS in 1h 08m 33s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0c69023a07354be2aa6a1c06a647d9ef : SUCCESS in 18m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3134839937b04fc2899e4ca0cc169f97 : SUCCESS in 47m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/073000863b4943d4a816989c3839f1eb : SUCCESS in 54m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b55d94eb7b624415ab986567c80e23a4 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7340937ca0bb41fd8daa4f44f3462360 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdd77114a41f42d88641f253b65eb340 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c54e21fc0cd4d0bb2ca53623aae1968 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7afa39cf65a4caaaa29f0cb23615f42 : SUCCESS in 7m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/29f83068b0f14bd0bf5eeae3aa0f3ff5 : SUCCESS in 20m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aefffd76abca4ab690f451398c1f58a5 : SUCCESS in 51m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fde743c1578d4f8a9da288a81c973c85 : SUCCESS in 50m 43s","accounts_in_message":[],"_revision_number":3},{"id":"c839fed26528078cc2bcd58c1136fcfb933fac58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 13:49:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8d045fa6665e98da3eaacf993a7d4e013e07cc90","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 13:55:29.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3d43774223026197bb2128561292f9baf5782f8b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 14:05:56.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a96e239d6f98f9d6fe11063fd907eba96faee8d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 14:06:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fd394dc5e81a2087eb0d9bf3dc56edea7d6590b3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 14:18:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e1fa3f5508a53ddb0f29bcf6cfbb949ef4f2cb6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 14:23:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"82920e60d088d4a4e30f887d2ce12a24ddb0ae9e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 14:29:46.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"18ed7203f8263fe7e793c4ee81d7676e5099871a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-19 15:27:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"54ef123a9a04aa5ee966cd37707c96284b4dda07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 15:49:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/c9e4eb39396243178f07f8eecb14e9a8 : SUCCESS in 33m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6f38c68726d14ab7b21259e0e5b7bf8f : SUCCESS in 1h 17m 19s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/971523c05bbe424aa68238032ddc439d : SUCCESS in 1h 20m 30s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9be11cdb5c49487e8b3efd8e32657958 : SUCCESS in 22m 01s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/79bf071eedfa406fbab9375ac6b748c6 : SUCCESS in 1h 00m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ff0e661333694f90836e1f4f1f3ae9ff : SUCCESS in 1h 09m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05b69e0997da472eafb1a6359bca646b : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/126f0dc78a844f2cb0d4045fd3c61056 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/476ad94ad8fa4a6784f9db9cf3c1de2b : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc05edcab03444f3b78160364f4fd456 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b12915c4064ce490ef6e27c2473b1a : SUCCESS in 5m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/95eca7c67a97428fa8fe28031abe0f85 : SUCCESS in 27m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f9168a8b4deb4ae3857be8790c7f4667 : SUCCESS in 50m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/12cffd096e024ff48c6389b1710afd39 : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":6},{"id":"753cb21a7cd3a834e0c4c6be192e7502db6d7399","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-19 15:52:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"250cc079142e9c0c09cec7a087061751e459cc87","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-19 17:38:16.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a996167e083b13a3a1cd2b056b5efcead6da21eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 17:38:49.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0f86503b68521ff2021a5ddeccd2ee6f5a7a5769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 18:50:28.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/cd99bb48d81c4417b93e5e71a9d6e20c : SUCCESS in 23m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/78763bd368c3401e871ad42560aa60b7 : SUCCESS in 1h 04m 30s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7a6429c59a754bc39938fb7fcfc3b0ec : SUCCESS in 1h 01m 30s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6d5af49bfc4d4d7c9449673209300225 : SUCCESS in 19m 05s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1c75979775ed48179ca8f9c261c4fbca : SUCCESS in 51m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2f709138e1384f1bba606c6b9384700b : SUCCESS in 1h 00m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60d922fe3da4adab7d09466a8f4175c : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5d6b6a190414d619e37f1474806a4eb : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15bb40b5a6264280ba6ad0a2d3f8eea7 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1274d2708a9a449cb313fa8dd21e03ad : SUCCESS in 5m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c351f010d8044ab68ed0264c0442efef : SUCCESS in 18m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/07d5955cf0ee4af2ac526d5ece19f454 : SUCCESS in 41m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b112415a091e489e8bec7c7d0578ee89 : SUCCESS in 1h 06m 07s","accounts_in_message":[],"_revision_number":6},{"id":"727b745e135717f7bd736ef4a7ab9b9abc81c7e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 18:50:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"3f8472336a592391f45df05d8ee05ddddfb5f029","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-19 18:53:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5bf9e92244a4436a87cb59324dcc348 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":6},{"id":"eecfb263b8145bfb7d9c697a125e670dfe4d113e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-19 19:00:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a926286585c2968442c12306b41efe12e1ec8fcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 02:20:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/8c7bec54bc8146af93ef66b17f0f9824 : SUCCESS in 2h 19m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e2d35fd6da7e4aca8c7d4a8cda6bbe0d : SUCCESS in 2h 08m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/593478e090f046aa8605d8f6f0676c34 : SUCCESS in 2h 17m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f15b1aa32764598bd021613d5576026 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/554341775ebf45c6ab696dc648a92e8a : SUCCESS in 6m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2f3a694084d243a0b7bedc9441f5fdb9 : SUCCESS in 1h 45m 28s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e877556aafb2b571152567a8fc1f6ec640ffb2a4","revisions":{"07876d1d3aea1a37dd7c9c5908d6c9d91c748a00":{"kind":"REWORK","_number":1,"created":"2022-04-19 11:23:30.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/91/838491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/838491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/838491/1"}}},"commit":{"parents":[{"commit":"f33ef03977536ecb29ea464b9e557a6caacf4345","subject":"Revert \"CI: add templated Dockerfiles to build logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f33ef03977536ecb29ea464b9e557a6caacf4345"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 11:21:25.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 11:21:25.000000000","tz":120},"subject":"add a way to not have \u0027install_type\u0027 in image name","message":"add a way to not have \u0027install_type\u0027 in image name\n\n--use-short-names argument drops \u0027install_type\u0027 from image names:\n\nkolla/debian-openstack-base:14.1.0\n\nThis should allow to get rid of \u0027-source\u0027 in image names without\ndisrupting CI jobs.\n\nChange-Id: I29c9d51f4b518a4e2d3157ab5d6cc434fb83f2ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07876d1d3aea1a37dd7c9c5908d6c9d91c748a00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07876d1d3aea1a37dd7c9c5908d6c9d91c748a00"}]},"branch":"refs/heads/master"},"9e9ae28d367f52ea28f342766e159590d5ca5683":{"kind":"REWORK","_number":2,"created":"2022-04-19 11:43:02.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/91/838491/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/838491/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/838491/2"}}},"commit":{"parents":[{"commit":"f33ef03977536ecb29ea464b9e557a6caacf4345","subject":"Revert \"CI: add templated Dockerfiles to build logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f33ef03977536ecb29ea464b9e557a6caacf4345"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 11:21:25.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 11:43:01.000000000","tz":0},"subject":"add a way to not have \u0027install_type\u0027 in image name","message":"add a way to not have \u0027install_type\u0027 in image name\n\n--use-short-names argument drops \u0027install_type\u0027 from image names:\n\nkolla/debian-openstack-base:14.1.0\n\nThis should allow to get rid of \u0027-source\u0027 in image names without\ndisrupting CI jobs.\n\nChange-Id: I29c9d51f4b518a4e2d3157ab5d6cc434fb83f2ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e9ae28d367f52ea28f342766e159590d5ca5683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9e9ae28d367f52ea28f342766e159590d5ca5683"}]},"branch":"refs/heads/master"},"4dec8ebf310fa38a2bdcd427886a5e9339b8b8fc":{"kind":"REWORK","_number":3,"created":"2022-04-19 11:58:02.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/91/838491/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/838491/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/838491/3"}}},"commit":{"parents":[{"commit":"0cf5b1d4e6e871562c0ca0f48a21594f8d197471","subject":"enable logging to file for quiet mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cf5b1d4e6e871562c0ca0f48a21594f8d197471"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 11:21:25.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 11:57:54.000000000","tz":120},"subject":"add a way to not have \u0027install_type\u0027 in image name","message":"add a way to not have \u0027install_type\u0027 in image name\n\n--use-short-names argument drops \u0027install_type\u0027 from image names:\n\nkolla/debian-openstack-base:14.1.0\n\nThis should allow to get rid of \u0027-source\u0027 in image names without\ndisrupting CI jobs.\n\nChange-Id: I29c9d51f4b518a4e2d3157ab5d6cc434fb83f2ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dec8ebf310fa38a2bdcd427886a5e9339b8b8fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4dec8ebf310fa38a2bdcd427886a5e9339b8b8fc"}]},"branch":"refs/heads/master"},"e30162b80b71dc117abd87f3e91b6bfdaf9b248d":{"kind":"REWORK","_number":4,"created":"2022-04-19 13:49:40.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/91/838491/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/838491/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/838491/4"}}},"commit":{"parents":[{"commit":"0cf5b1d4e6e871562c0ca0f48a21594f8d197471","subject":"enable logging to file for quiet mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cf5b1d4e6e871562c0ca0f48a21594f8d197471"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 11:21:25.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 13:48:28.000000000","tz":120},"subject":"add a way to not have \u0027install_type\u0027 in image name (temporary)","message":"add a way to not have \u0027install_type\u0027 in image name (temporary)\n\n--use-short-names argument drops \u0027install_type\u0027 from image names:\n\nkolla/debian-openstack-base:14.1.0\n\nThis should allow to get rid of \u0027-source\u0027 in image names without\ndisrupting CI jobs.\n\nSwitch publish jobs to use short names so k-a tests will have ready to\nuse images for both long and short names.\n\nChange-Id: I29c9d51f4b518a4e2d3157ab5d6cc434fb83f2ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e30162b80b71dc117abd87f3e91b6bfdaf9b248d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e30162b80b71dc117abd87f3e91b6bfdaf9b248d"}]},"branch":"refs/heads/master"},"6c0abc2295b37b1a728832d938cb6bbf5afdb4fc":{"kind":"REWORK","_number":5,"created":"2022-04-19 14:06:08.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/91/838491/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/838491/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/838491/5"}}},"commit":{"parents":[{"commit":"0cf5b1d4e6e871562c0ca0f48a21594f8d197471","subject":"enable logging to file for quiet mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cf5b1d4e6e871562c0ca0f48a21594f8d197471"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 11:21:25.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 14:06:02.000000000","tz":120},"subject":"add a way to not have \u0027install_type\u0027 in image name (temporary)","message":"add a way to not have \u0027install_type\u0027 in image name (temporary)\n\n--use-short-names argument drops \u0027install_type\u0027 from image names:\n\nkolla/debian-openstack-base:14.1.0\n\nThis should allow to get rid of \u0027-source\u0027 in image names without\ndisrupting CI jobs.\n\nSwitch publish jobs to use short names so k-a tests will have ready to\nuse images for both long and short names.\n\nChange-Id: I29c9d51f4b518a4e2d3157ab5d6cc434fb83f2ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c0abc2295b37b1a728832d938cb6bbf5afdb4fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c0abc2295b37b1a728832d938cb6bbf5afdb4fc"}]},"branch":"refs/heads/master"},"e877556aafb2b571152567a8fc1f6ec640ffb2a4":{"kind":"REWORK","_number":6,"created":"2022-04-19 14:23:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/91/838491/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/838491/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/838491/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/838491/6"}}},"commit":{"parents":[{"commit":"0cf5b1d4e6e871562c0ca0f48a21594f8d197471","subject":"enable logging to file for quiet mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0cf5b1d4e6e871562c0ca0f48a21594f8d197471"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 11:21:25.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-19 14:23:12.000000000","tz":120},"subject":"add a way to not have \u0027install_type\u0027 in image name (temporary)","message":"add a way to not have \u0027install_type\u0027 in image name (temporary)\n\n--use-short-names argument drops \u0027install_type\u0027 from image names:\n\nkolla/debian-openstack-base:14.1.0\n\nThis should allow to get rid of \u0027-source\u0027 in image names without\ndisrupting CI jobs.\n\nSwitch publish jobs to use short names so k-a tests will have ready to\nuse images for both long and short names.\n\nChange-Id: I29c9d51f4b518a4e2d3157ab5d6cc434fb83f2ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e877556aafb2b571152567a8fc1f6ec640ffb2a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e877556aafb2b571152567a8fc1f6ec640ffb2a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
