)]}'
{"id":"openstack%2Fkolla~838918","triplet_id":"openstack%2Fkolla~master~Ic9e54568a2844991b12e04ebfb1444b5cf286dbb","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-04-22 10:12:43.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-22 10:12:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic9e54568a2844991b12e04ebfb1444b5cf286dbb","subject":"[release] Use yoga branch for neutron-vpnaas and monasca-api","status":"MERGED","created":"2022-04-21 15:38:32.000000000","updated":"2022-04-22 10:14:43.000000000","submitted":"2022-04-22 10:12:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838918","meta_rev_id":"62ac37a89c6da6fc62122fa0f2f00d58fc3399b9","_number":838918,"virtual_id_number":838918,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 10:12:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":1,"date":"2022-04-21 19:00:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":2,"date":"2022-04-22 08:51:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-04-22 10:14:43.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2022-04-22 08:51:52.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-21 15:41:17.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-21 17:03:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 19:00:51.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2022-04-22 08:51:52.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"f0b8a21b53b838a905871c25da179604668ddfcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-21 15:38:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f811550dba96b2c54a0a0173e0d7f8e5a30aac26","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 15:41:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76ce1d6d03ad870df118f4fe6755ca9d52beb915","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-21 15:43:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5c871145aa7dad2c22e44634a1c7a3de5c6145c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 15:47:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dffdaed82786d9bc6c1e4c490438529d3fff2447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 17:03:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/3c28f0d4bb9a45599c139e8f463f422f : SUCCESS in 33m 06s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c4e090b2f70a40f9ad6f613b4b0d4792 : FAILURE in 36m 53s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e1c93df1d18643d5a7357928a23f88af : FAILURE in 1h 16m 44s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/34132116e5dc4ac5bfbd334fdfd00158 : SUCCESS in 26m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b39157fb1b8a4be197179669d5556cee : FAILURE in 30m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5ef64bb955c045a99c2124634a2543e8 : FAILURE in 1h 08m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7678d4909a14211a3205b057ed0f992 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec5df9ddc3714c4abb0d92e5efe7ac87 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7412813de8b14359b31f7c5aa59c0c78 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3847a890528f4a8dbede9c2a1ed31355 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78eb5fa33f0441f88105a04b21bc89d7 : SUCCESS in 6m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4a5ed320f7834f2f9aec6b9bb82f4326 : SUCCESS in 26m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f8ebbd3cd3264fe78ee98ff4f89e56e9 : FAILURE in 20m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c9f99cf75717464c9d802ebf3267cc1d : FAILURE in 47m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ac467210efb75e34b066520f27d6a153c665b84a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 17:24:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/40044b5b23d743efba638b4de7d81ffe : SUCCESS in 1h 42m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6c88be0ad30d448c86a2c35db40f208f : SUCCESS in 1h 15m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a096a26ee404c84aa515906409daf0d : FAILURE in 1h 08m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/53a1b7dfcaa947ceb4479d90ad7ea0b5 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6cd0498afb1341e1afb21f5774877d60 : SUCCESS in 6m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/98ca3c1489b9462c923f3ffb2859fdbd : SUCCESS in 1h 40m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"134f693be943844bee86ae548d68dab67071ee58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 17:26:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"df4f7960595ff7e79ac646f46a58de465891ed5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 18:56:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/5da0e5f9d87646e78f054ed8fc9c961a : SUCCESS in 40m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3895c347623e48629c3008c777fd9011 : SUCCESS in 1h 04m 36s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f1a2fca3d2b24b549cc07c1aea2ae0eb : SUCCESS in 1h 27m 06s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/51d24747de2d4b25a5c9e6a394b14904 : SUCCESS in 19m 32s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b6409e4ae95d49ddacbf35b847559142 : SUCCESS in 1h 06m 10s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2009be17fc4e4c9dbcc74b2ae9616677 : SUCCESS in 1h 06m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83f13a490e944a8385a655a5a810ccd8 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/543dd7566ee246d79ad689dc912f49e1 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca760a9b2f8541afbc989acdc5b93142 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66081e900bde4735ba1223ca2f86bc03 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f8d474114664ada9e4a221ea45a57fe : SUCCESS in 9m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2e3d6ac6bcb144fbb617f1b6a433feb0 : SUCCESS in 18m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ded8d42282af43829198689964b2a387 : SUCCESS in 53m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c2e9fbfbf3c342ad9535d6ede10dc602 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":2},{"id":"e7c71cbadc6b8e95609a25e005a0c135c4c37eaf","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2022-04-21 19:00:51.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3f59689207ad1ffa8a96b62598e86561623d4329","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 22:44:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/e9e0993434764309a164ecba67dc2d68 : SUCCESS in 2h 08m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9f00c4cf9807492eaf52e2ee1761f459 : SUCCESS in 2h 13m 11s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/17a7aba9845841ffa9b6f98045447f2d : SUCCESS in 2h 28m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/09aeefd35c554d4d9698a5f26bca96dd : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1db7a5be8c914d86b8ecb5f0316e11f3 : SUCCESS in 6m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5f4218ddcf89446287cb8737596b006e : SUCCESS in 1h 38m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f25064926924633fec1d1752892b10b610150218","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-22 08:51:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"648bf34ad55de8a565d1a748cf632984a0f84cc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:52:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e5863e27e014021f56f4c483a6fa2030046870a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:12:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/30acf9c672184dda92fc0d12a67cc87c : SUCCESS in 30m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d6a08882b824483d8fae8c848cf6ba13 : SUCCESS in 1h 08m 24s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/662ac27b5dbc43478bd8607db2cac564 : SUCCESS in 1h 20m 03s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f1aa49e227cb4170b20ed55586287439 : SUCCESS in 22m 01s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2abd02b2c5264dbc93dd6bb8cb345eb8 : SUCCESS in 58m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8cdae4e69d334c5d9986c92db51482b3 : SUCCESS in 1h 09m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/200d08c0f9ac46bea2822012dfb100e2 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0531a5c06f5e443bb7e2a2982e6b544b : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/224b6b8d3b3d488f8a9e68bd9381804c : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64799f77422d420aa8b8521c68549d56 : SUCCESS in 7m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/88ca597548bb4c659fd5f3d1f9286fea : SUCCESS in 23m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca2a17016363432fb2b2449820934db9 : SUCCESS in 51m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7914968e89e44af4b77c04985363b96d : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":2},{"id":"de7807c003fcd00101c691f03600d9a949a0a9a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:12:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"938538ce953fd6f0534f20e9d8268195406463db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:14:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1c1ff035b14e0b9b25e8e4aa0e938a : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"addd99c8ea29033fe077b650aa41714b89aab34c","revisions":{"513854dc6e1312490ae65b2fb2799239c0c3cd2d":{"kind":"REWORK","_number":1,"created":"2022-04-21 15:38:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/838918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/838918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/838918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/838918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/838918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/838918/1"}}},"commit":{"parents":[{"commit":"e877556aafb2b571152567a8fc1f6ec640ffb2a4","subject":"add a way to not have \u0027install_type\u0027 in image name (temporary)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e877556aafb2b571152567a8fc1f6ec640ffb2a4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-21 15:38:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-21 15:38:21.000000000","tz":60},"subject":"[release] Use yoga branch for neutron-vpnaas and monasca-api","message":"[release] Use yoga branch for neutron-vpnaas and monasca-api\n\nThese were not available at the time we switched everything else. Venus\nremains unavailable.\n\nChange-Id: Ic9e54568a2844991b12e04ebfb1444b5cf286dbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/513854dc6e1312490ae65b2fb2799239c0c3cd2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/513854dc6e1312490ae65b2fb2799239c0c3cd2d"}]},"branch":"refs/heads/master"},"addd99c8ea29033fe077b650aa41714b89aab34c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-21 17:26:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/18/838918/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/838918/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/838918/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/838918/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/838918/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/838918/2"}}},"commit":{"parents":[{"commit":"08b80e34dae0236e6b106e7debcdb755508fe1d9","subject":"switch to short image names by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08b80e34dae0236e6b106e7debcdb755508fe1d9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-04-21 15:38:17.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-21 17:26:11.000000000","tz":0},"subject":"[release] Use yoga branch for neutron-vpnaas and monasca-api","message":"[release] Use yoga branch for neutron-vpnaas and monasca-api\n\nThese were not available at the time we switched everything else. Venus\nremains unavailable.\n\nChange-Id: Ic9e54568a2844991b12e04ebfb1444b5cf286dbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/addd99c8ea29033fe077b650aa41714b89aab34c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/addd99c8ea29033fe077b650aa41714b89aab34c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
