)]}'
{"id":"openstack%2Fkolla~838925","triplet_id":"openstack%2Fkolla~master~I5178c58fbd8c65efe825dc249c0f1368ef0fe8e0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-22 08:46:55.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-21 19:13:39.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I5178c58fbd8c65efe825dc249c0f1368ef0fe8e0","subject":"flatten images a bit","status":"MERGED","created":"2022-04-21 16:22:39.000000000","updated":"2022-04-22 08:48:29.000000000","submitted":"2022-04-22 08:46:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"838925","meta_rev_id":"f471326def1196d6322d86db08dde7f0bdfcf3a5","_number":838925,"virtual_id_number":838925,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 08:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":2,"date":"2022-04-21 18:59:45.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":2,"date":"2022-04-21 19:13:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2022-04-22 06:46:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-21 17:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-21 18:59:45.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2022-04-21 19:13:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"e915cf73a312e3287d7ae343c44a031eda09208a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-21 16:22:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c96626d64440c47c3b99aa26a0b0da42a00287f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 17:28:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/c57bcac0ecab4922895e14a2539f8a9e : SUCCESS in 29m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4aed7a39ac5c45dd8954fb5eab36b34a : FAILURE in 30m 05s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4cdf3a1b2b2f4a4eb84373fd804b6070 : FAILURE in 1h 00m 04s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9555a96234834e1a985797a12bc6c875 : SUCCESS in 24m 36s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/815b0eb7418a4f8e8cd5c68855009c95 : FAILURE in 30m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1548ad1508f84622a9677d0a39f2c34d : FAILURE in 55m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae2d2fc3d8644d9ca0633ab0e4e19c86 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55b251caa0ce4f6c9cba37601aa91689 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1f5363d5dea403488d9cb891854ceef : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7eb1269f772042c28363ad24dd2ad658 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69e40d6919a44866b05ca6397a318dfa : SUCCESS in 5m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6837945813cb4d95abf75a0bdb6784d6 : SUCCESS in 21m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64c5230912bf48d39f16103d6a0c1101 : FAILURE in 21m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1ec4e900f1794be0bafa966945adaa28 : FAILURE in 45m 13s","accounts_in_message":[],"_revision_number":1},{"id":"064d97afa5751b727df2a990bc0929e4b2343ef3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 18:53:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"20f5d10affc1f793b00bb6e9b6bd12739c92ac56","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2022-04-21 18:59:45.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba81dc2fb243a3bbfa6ea233f386bbcbbfe60610","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 19:13:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df7aa9cfc1830e7072020b58b21810d599c6217d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 19:13:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b342e0a8ffa79b0b44ea9b00e45c44e3f236c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 21:27:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/b5b7a60dc1bb41c0b8606bc86fb10c5e : SUCCESS in 33m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/84de0dc6d3754dc883a63d6b07682364 : SUCCESS in 1h 08m 26s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a59d5e1d1db94e7b9e987e8748bfe12e : SUCCESS in 1h 34m 25s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1b761b48d74642dc9f430687723e2635 : SUCCESS in 22m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a9b44b4b94a3477f8ee8d537e72041a8 : SUCCESS in 56m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/768a5887872f45ca91d46122d21c7cc3 : SUCCESS in 1h 09m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/526ec899f66140eab573b13886574a42 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/324f5f8e5d8c427da0a019b722951c2c : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a604367f81b749be927e9a4dcf1f4db5 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30c32e87751d4ae282ed8022a62bbfa1 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de0f51ac8f5c46288bce7cbb2cc7eaf7 : SUCCESS in 6m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/53beaf08fecc4ee299b934fdc4df08db : SUCCESS in 21m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/71dbc4be7b5645d4a0db51b2b1cb14f1 : SUCCESS in 49m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2099d8bb7d05475b82ad0a578b082206 : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":2},{"id":"51ffea8ab99001be64901918cf89e0418a0d1774","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 21:54:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s https://zuul.opendev.org/t/openstack/build/e4eb45d268094337b080111daeaea9a0 : SUCCESS in 52m 36s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/54228b6274924fdeaa01a735162f58b3 : SUCCESS in 1h 37m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/942e8766db9243ec8800c674668ffa75 : SUCCESS in 1h 59m 55s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/926b3effe3f64b62a149984ce85b1af5 : SUCCESS in 1h 51m 04s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c81b76b38bef4814bf0ca20fd07cdd64 : SUCCESS in 1h 44m 16s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/0c17e9814254403ab057053fc0c4a541 : SUCCESS in 1h 26m 50s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/503feadab99246aa9bb8981554605f11 : SUCCESS in 1h 34m 31s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/a97e38f6875b4e67b40367de9a948452 : SUCCESS in 37m 07s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/603493af395f47f5a63fa6a25ff34ac0 : FAILURE in 1h 05m 53s (non-voting)\n- kolla-ansible-centos8s-source-octavia https://zuul.opendev.org/t/openstack/build/aa2fbe67cbfd41cab575676c255dfe88 : SUCCESS in 1h 18m 42s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/559907b6efe84b6daa61e2e7ba17be8d : SUCCESS in 1h 19m 51s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/6093bd6a48414e9a9fceae12454c00f3 : SUCCESS in 43m 03s (non-voting)\n- kolla-ansible-centos8s-source-kvm https://zuul.opendev.org/t/openstack/build/c0440b9d5d2347a9a856e4b073229709 : SUCCESS in 1h 00m 16s (non-voting)\n- kolla-build-no-infra-wheels-debian https://zuul.opendev.org/t/openstack/build/6bbb1960535b47f38d8dd9b19c5d00df : SUCCESS in 21m 35s\n- kolla-ansible-debian-source-ironic https://zuul.opendev.org/t/openstack/build/cac0a80e33884a7eb319c58ed3bd480b : SUCCESS in 1h 07m 43s (non-voting)\n- kolla-build-no-infra-wheels-ubuntu https://zuul.opendev.org/t/openstack/build/25d0964cccbe43e99376151fa7f502f9 : SUCCESS in 22m 51s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/4098c66829f34adbacfd2ec1bfbbf7ee : SUCCESS in 1h 20m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/415a10eb343a43ce8cdfdc42656fdeb2 : SUCCESS in 1h 23m 09s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/15e374645f924efe92665e53f69ba4fa : SUCCESS in 1h 10m 05s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/abf2b7fb301d48b086f6434eaded0024 : SUCCESS in 1h 17m 25s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/9f4be172dcfd41b4a5dda436b016e188 : SUCCESS in 58m 01s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/86d17bdcfbd84a9e9c9dbb8c48df05f6 : SUCCESS in 25m 39s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/5c923401e4564e54990867e97f927850 : FAILURE in 41m 53s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/a6da1c6bc09643688d2906d47d5ec6cd : SUCCESS in 49m 45s (non-voting)\n- kolla-ansible-ubuntu-source-octavia https://zuul.opendev.org/t/openstack/build/d28f5d5b6eff4dfa9886da65f4957121 : SUCCESS in 50m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/fe8bf5a4ac924517af7cbc73c6908a89 : FAILURE in 47m 41s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/127e794538e44fa8b64d61bf403e3559 : SUCCESS in 28m 52s (non-voting)\n- kolla-ansible-ubuntu-source-kvm https://zuul.opendev.org/t/openstack/build/719fd7fe5cdf4c2e81b8b92cf9e9bf8b : SUCCESS in 43m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7c5d902dc7dced90ccb401496b8876b16cf3cac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 01:50:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/9ce401d305e148ab8ef5036731b84c53 : TIMED_OUT in 6h 01m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/498726cb6c5e4a83be55a711ad6ef2ef : SUCCESS in 1h 43m 38s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0c4f2c5dbe374f98ad606564035496d0 : SUCCESS in 2h 06m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f59499f1c8a4024a3c28ebd7385c486 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4cfade95fa294c0c8ae1189ee773f737 : SUCCESS in 8m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1f2b337658c24da0a9a0e77874d2c77e : SUCCESS in 2h 03m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6801534b9abe72c5564f5fd2b7192064dbdd776e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-22 06:46:39.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8552eb99ceac412a928ce86bae433c472c12c8c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 06:47:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e0427c14d90692414e00eeefa0573155068abb8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:46:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/84fcc8ae0bf545a68edd5a4989cd989b : SUCCESS in 31m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3751dd7b8d5847719e7f85a4f467e4ba : SUCCESS in 1h 09m 14s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c7a49fb6babf451c9b8f2050fad6e137 : SUCCESS in 1h 53m 32s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e2fae9b276b04c11a6508477b700b24d : SUCCESS in 23m 46s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7ad773f26b814012b17ec2bb460a3152 : SUCCESS in 1h 10m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e3a62e31d3d841ed893c9d6f85ca2cb4 : SUCCESS in 1h 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb15173627eb4559b0ecdb4951f51d14 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/412573a69763434397726e6a1c11d4f5 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e2c5a6d87334f6daaa27de388e8be6a : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4c46de29295458493a3786f957d5df1 : SUCCESS in 6m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f0746079db2e42d1a0c890aaf380dfd6 : SUCCESS in 31m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a69fc7773c8044fb9e4f1e4de2dbe1d2 : SUCCESS in 53m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7d0dec2199d7429692a40335464e03fb : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":2},{"id":"01976cbc68eb464237e305cfcc8cbb427f76108f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:46:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f471326def1196d6322d86db08dde7f0bdfcf3a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 08:48:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41389e8a9ca04f6c8d7efdfc9a75050c : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e21aeb5ae955671937055baf57836120af8d4312","revisions":{"545939ddaaf059e91a55648c8a59f6843a40a4da":{"kind":"REWORK","_number":1,"created":"2022-04-21 16:22:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/25/838925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/838925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/838925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/838925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/838925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/838925/1"}}},"commit":{"parents":[{"commit":"e877556aafb2b571152567a8fc1f6ec640ffb2a4","subject":"add a way to not have \u0027install_type\u0027 in image name (temporary)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e877556aafb2b571152567a8fc1f6ec640ffb2a4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-21 16:21:42.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-21 16:21:42.000000000","tz":120},"subject":"flatten images a bit","message":"flatten images a bit\n\nAs we have one type of images now some RUN calls could be merged so we\nwill have less layers in resulting images.\n\nChange-Id: I5178c58fbd8c65efe825dc249c0f1368ef0fe8e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/545939ddaaf059e91a55648c8a59f6843a40a4da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/545939ddaaf059e91a55648c8a59f6843a40a4da"}]},"branch":"refs/heads/master"},"e21aeb5ae955671937055baf57836120af8d4312":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-21 18:53:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/25/838925/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/838925/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/838925/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/838925/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/838925/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/838925/2"}}},"commit":{"parents":[{"commit":"08b80e34dae0236e6b106e7debcdb755508fe1d9","subject":"switch to short image names by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08b80e34dae0236e6b106e7debcdb755508fe1d9"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-04-21 16:21:42.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-04-21 18:53:14.000000000","tz":0},"subject":"flatten images a bit","message":"flatten images a bit\n\nAs we have one type of images now some RUN calls could be merged so we\nwill have less layers in resulting images.\n\nChange-Id: I5178c58fbd8c65efe825dc249c0f1368ef0fe8e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e21aeb5ae955671937055baf57836120af8d4312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e21aeb5ae955671937055baf57836120af8d4312"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
