)]}'
{"id":"openstack%2Fkolla~839032","triplet_id":"openstack%2Fkolla~master~Ia1797c32fc6a35f9f077c673abf4d8e16e51a760","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-04-22 12:28:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia1797c32fc6a35f9f077c673abf4d8e16e51a760","subject":"adjust permissions on _extend_start files (part II)","status":"MERGED","created":"2022-04-22 09:38:11.000000000","updated":"2022-04-22 12:29:59.000000000","submitted":"2022-04-22 12:28:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839032","meta_rev_id":"a02c13931ea8e4e594ac805e76cf4ddf86f2f623","_number":839032,"virtual_id_number":839032,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-22 12:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2022-04-22 09:49:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2022-04-22 09:42:11.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2022-04-22 09:49:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-22 09:42:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-04-22 09:49:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-22 10:57:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c85bc7e9db1fc4c17800035328989cc9bbb25034","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-22 09:38:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68cd72e0f5de9e191f473596609ce6b9da643b27","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-22 09:42:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08d95ef75b834cdacd097dec7f12051df79a373c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-22 09:49:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6264c9c343dc68f78732a23b023d082d0b6820cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:57:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/a939629a1a87482d8d3bea6bec11a716 : SUCCESS in 36m 30s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1eff03e3790d4f90a181eaa988860788 : SUCCESS in 1h 13m 45s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/11c2ce818eba47fab454b22043fa7030 : SUCCESS in 1h 12m 22s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a1e816c60dd04e4bbb9f7e3cd6c8c62c : SUCCESS in 18m 22s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1560b44db1564e9c99100a168b71afb8 : SUCCESS in 1h 00m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9592941c77594d22b80b46495b536218 : SUCCESS in 1h 06m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/020127e60da64d59b92dc67657bbfcb1 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baa8953bf26148a59180cce643491695 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03653c5e47da4a1daecc9a4134b50340 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c3497a498484c8e837859c4c8c61a4d : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9f1403640654c50a170a86a536fac10 : SUCCESS in 6m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/94835ee6824a4672bede54e46b68c83e : SUCCESS in 34m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f2404b93cf24b059f9fc363756cc7dc : SUCCESS in 47m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5b13e0a41f0f45119e7160a37fed43f8 : SUCCESS in 1h 03m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a061590e91353c32efef9ed6f4cda5e327b542a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 10:57:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7c9f4facb4be46c765528aa3d7d73b06a3cd78f4","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 11:49:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s https://zuul.opendev.org/t/openstack/build/d77850f01ce349bd98eae4b5bfd6f6d9 : SUCCESS in 30m 48s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/042fbe806b6d4e4590b46ff114b798dd : SUCCESS in 1h 43m 51s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/685e39ff4ad744859a35509170521b22 : FAILURE in 2h 04m 07s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/54a3857d49c5484c8d6a93b7acb20f99 : SUCCESS in 1h 46m 44s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5126e3993e164eb8b2af7a8dc2fd1003 : SUCCESS in 1h 48m 00s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/6744c6c2d6154e2a942f8629bd5b98b1 : SUCCESS in 1h 16m 44s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/f5edb9fc60da46469a6d68d0709bca46 : FAILURE in 1h 00m 21s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/73ff4b30bf7741e0a65c36eb1e6aae0e : FAILURE in 57m 45s (non-voting)\n- kolla-ansible-centos8s-source-octavia https://zuul.opendev.org/t/openstack/build/43b0c1af228646af90ecceb39c45ef6f : SUCCESS in 1h 08m 41s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/107c982445894d7e9e740db75e619be3 : FAILURE in 1h 07m 10s (non-voting)\n- kolla-ansible-centos8s-source-kvm https://zuul.opendev.org/t/openstack/build/9867582272354d26af74a6e2f494a0b5 : SUCCESS in 1h 02m 21s (non-voting)\n- kolla-build-no-infra-wheels-debian https://zuul.opendev.org/t/openstack/build/cb78c0a8290242d8a5573748edd7f2c9 : SUCCESS in 22m 25s\n- kolla-ansible-debian-source-ironic https://zuul.opendev.org/t/openstack/build/16d309da6b6f45f8830297ae0feb80d9 : SUCCESS in 1h 17m 08s (non-voting)\n- kolla-build-no-infra-wheels-ubuntu https://zuul.opendev.org/t/openstack/build/5a1ba8d86bb54a8aa4deffe664315a72 : SUCCESS in 22m 15s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/d4b217a56c24473682d6f908763dc8d3 : SUCCESS in 1h 03m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/71ed6e3a02344ad3aca5603fcf899cc4 : SUCCESS in 1h 24m 31s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/5b2285d8c4c04aeeb3f0de487af091da : SUCCESS in 1h 01m 49s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/43be445f9f2a4d13a348c10d2056da01 : SUCCESS in 1h 08m 07s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/8557101b5e7e41a6af7a15bdde8f4177 : SUCCESS in 1h 06m 37s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/9ea92725c1234af8bc806a4ce2b6b396 : FAILURE in 37m 56s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/a9473cb817f3486e86ce2db273250f8c : SUCCESS in 50m 57s (non-voting)\n- kolla-ansible-ubuntu-source-octavia https://zuul.opendev.org/t/openstack/build/6b5404212d6940efa703627205caf49d : SUCCESS in 57m 49s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/e301b38483ba46259dc74971cdb6176c : FAILURE in 48m 50s (non-voting)\n- kolla-ansible-ubuntu-source-kvm https://zuul.opendev.org/t/openstack/build/ebd748a41e1a41acae223afbeef3118b : SUCCESS in 41m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8d2d5f085787ce37f8df72c2edee72d225d5a6b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 12:00:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/3a6dca622caa4dfe9021af4f7b69ea8d : SUCCESS in 2h 15m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b753a98540464e758669fb285c5c8ae6 : SUCCESS in 1h 35m 18s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b628ee92a22741ef840a48d2cc2447b3 : SUCCESS in 2h 10m 18s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fbd731acf7743eba54f7b053908b95b : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9336bfae2ead4d45b6fcb141e1d5145f : SUCCESS in 5m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/16f5e44487fe43e9827fab6de28de24e : SUCCESS in 1h 44m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"153312908113a4ef299684c4b481df1bf69583dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 12:28:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/745e16c55f964523b571a59f1e4798a3 : SUCCESS in 31m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3803552d1cd641afb915239f2b7786c8 : SUCCESS in 1h 11m 14s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f4ba524e750f4a35b5a2bc697d69ba88 : SUCCESS in 1h 24m 40s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5b6616217f0d4f7092ed276f30368a01 : SUCCESS in 22m 53s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ec79a20f5a464f50a4386393d2765a09 : SUCCESS in 1h 06m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/33efab2aba394ef1b4e31316b243bdf6 : SUCCESS in 1h 13m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c306909b7394d46aea8266a368c941d : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6479e04076144ae3b4ad273693d7ed59 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7ee1f7e451944818564b2a6fbce19b8 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/575a0f7faad74038afff44ae0fc3379a : SUCCESS in 6m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6225fb2339a9413c9669b6d1f12e61d4 : SUCCESS in 22m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3548882c51114e17a1634c467fafa016 : SUCCESS in 52m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ec8d611c72114df89b4730156b791469 : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c4c80132f6e8bb9b58aace27c8ae86d3912bdaab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 12:28:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a02c13931ea8e4e594ac805e76cf4ddf86f2f623","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 12:29:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aa921997b204f76b00ce02d67185139 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"016c5ed30c0fe6c46004731d13ec0748daa6c70c","revisions":{"016c5ed30c0fe6c46004731d13ec0748daa6c70c":{"kind":"REWORK","_number":1,"created":"2022-04-22 09:38:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/32/839032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/839032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/839032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/839032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/839032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/839032/1"}}},"commit":{"parents":[{"commit":"7cb4f7d01c5e61ebd5f324822e41910564d5563f","subject":"Merge \"flatten images a bit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cb4f7d01c5e61ebd5f324822e41910564d5563f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-04-22 09:34:55.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2022-04-22 09:34:55.000000000","tz":120},"subject":"adjust permissions on _extend_start files (part II)","message":"adjust permissions on _extend_start files (part II)\n\nTim Shearer started it in 1d96a2bbe1b536b7d7f4cdbf55c6dabae6d058ae.\n\nSince all extend_start files are sourced rather than executed, the executable\nbits are now cleared throughout the project.\n\nChange-Id: Ia1797c32fc6a35f9f077c673abf4d8e16e51a760\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/016c5ed30c0fe6c46004731d13ec0748daa6c70c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/016c5ed30c0fe6c46004731d13ec0748daa6c70c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
