)]}'
{"id":"openstack%2Fkolla~839585","triplet_id":"openstack%2Fkolla~master~I11b4fbcaba165824252902d0270ce1dbbc3d6e9d","project":"openstack/kolla","branch":"master","topic":"jammy","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-07-27 15:43:40.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-07-27 15:43:40.000000000","reason":"Change was submitted"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-07-27 15:43:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I11b4fbcaba165824252902d0270ce1dbbc3d6e9d","subject":"Move to Ubuntu Jammy (22.04)","status":"MERGED","created":"2022-04-27 15:48:15.000000000","updated":"2022-07-27 15:45:46.000000000","submitted":"2022-07-27 15:43:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":45,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"839585-jammy","meta_rev_id":"9fe3437eed110260754a09f8a48e5f8f6fc3de8c","_number":839585,"virtual_id_number":839585,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-27 15:43:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-07-27 10:50:35.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-07-27 12:43:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-07-27 14:26:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 16:07:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-27 16:31:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 16:37:06.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-04-27 20:03:14.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2022-06-16 07:03:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-07-27 12:40:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-07-27 12:43:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"0a93a2ecb1ab7af235a954f16af96dfa45cc7157","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-27 15:48:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c5f55f0e9a9b8018e064aa60584e8e158040707","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-27 15:50:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9386e52f2950b4af2f047a16ba288d642516c757","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-27 15:58:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"200583588aaf32e4a34ae945cf284ad1233193bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:07:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52c502a475c0405d98d39bddbcfe34f8 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4c314333a2b47eb908a43cc16801062 : SUCCESS in 6m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e42eebbc1b6945408b72ab47567490aa : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ce0f68fbce678619c80c893acd87c4df3eb27267","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:31:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/981cba236b394b7083f84661dd527de6 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89e0589d431c4c4481c5d630a50dd7ed : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e20a0b650d5484cb27e0ff11762c2ce : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79b16c1dc7f44fa9938469a05686d6c2 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/717a00379e0d47afba94f7ca791e1d53 : SUCCESS in 6m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0297e3a4a56a488988dc1cc09109005a : FAILURE in 26m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc35bb842dbf4ce99a68144822117f69 : FAILURE in 21m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/31021206ab944833aa964a77c7190b58 : FAILURE in 23m 59s","accounts_in_message":[],"_revision_number":3},{"id":"e29ac047adf7bc074598d626f1ef6c29c40e81a2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-27 16:37:06.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"289c19e8f9e25d5a195597bc16efc1cb9911d0e4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-27 17:42:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"aa8a8457a0f36771a2bcf41386ab369b521d417e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 17:52:29.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fa6a2f23bb147549638874b29e91b87 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/743c7528b0744e4b9cfdd36138089a8b : SUCCESS in 7m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e1bb986f43b545b2bb0360290d41fdbc : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"38db0712c4a59a17bfb7134b6201c625b4c39a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 18:18:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f13d264cc54c4b798b2c0118a8814471 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0ce2f417b8246138d51f9f32a8a3b5b : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b66157a46064dba89c1ed8bf8fe0197 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b9ea52205cd4d9f835a98ed3bcf6b7a : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77f3e1688cfd490592ad81cf4d8ce1e1 : SUCCESS in 6m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ddef35e8a59d48f28a228fa7196e19b3 : FAILURE in 26m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2833cb1d75e74f86a6088ff999612af1 : FAILURE in 20m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/136554ffaf414843bee1661823b089de : FAILURE in 20m 48s","accounts_in_message":[],"_revision_number":4},{"id":"b36ac1ee53344f8364ac8091f535175e01ed9f8b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-04-27 20:03:14.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"50bf6636ec7817af1926ac4455e6292aa9386e55","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-28 07:10:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"853cebe9c9ad764a2d4dbfdf273f202d269970fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-04-28 07:13:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"405e7612eb8122d4b8e4a1379fccb07447cc0dc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 07:21:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52748fcfc9f24efe94ee93d7ae89f01a : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ac9c697b33b4b00b86840dbe137a8f5 : SUCCESS in 5m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8101f25fb7094437b49791d3facad444 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d5caaa85f58242cd219b4aa219459725414c96b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 07:47:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6702f8b03dd64293b685d01b9b32c47c : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89bc215f7269464a960489dab0892f65 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d5e9e82f0014ea7b5235faa5d625e11 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d99779b333342c59c3991cf32ea3ad9 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63a4955b25f43259aafd89ff0a4c0bb : SUCCESS in 5m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ac3615b416b54b3cb86ecdffa1fe00b2 : FAILURE in 26m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/242a9daca56247dcbe090020550d99f5 : FAILURE in 21m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f13767df31154ddd9c62d549cceb3e30 : FAILURE in 21m 22s","accounts_in_message":[],"_revision_number":6},{"id":"aa1d14772d9f0d627c03832db18bd20f6e0c6967","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-16 08:23:30.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3fcd3f241d6541807c0741bf87381fdca948e6c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 08:23:54.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 839585,6","accounts_in_message":[],"_revision_number":6},{"id":"d6fed49a84811484794895ab8b3366dde1341aef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 08:23:59.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 839585,6","accounts_in_message":[],"_revision_number":6},{"id":"a71a0ecc25d2de4dc7e0d55a00cf7d1519b8f893","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-16 08:34:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3aabf6fb40d545b0ee6b39a8dfe5b5aa6e29752c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 10:00:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/1094b3fd1dd045dba85cf0cfd2ac6c7a : FAILURE in 58m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f54865ba396f4ff7857345c8f12226bf : SUCCESS in 1h 14m 59s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4cc36957dfc84c3daa9e79615e6e046b : SUCCESS in 1h 23m 12s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cc44db9afd504b9faf634989b011d1cb : SUCCESS in 19m 22s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/473775db73714e2382f2b45c86b84f9e : SUCCESS in 53m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/161566abb3e14cadb2a60986cb034aae : SUCCESS in 1h 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a92a8b56c9a4981b14fea9e93e078e5 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbc4feeb5e7d4ae7ae8a9c43b9368fe1 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00407a975ac5435a87686921836d8f0a : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8a1d9e0c4be44db7a4adcefb7e4c8c96 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b9b02db06a34ead8f3bf51be523d40f : SUCCESS in 5m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/638eb4a1a2564ca2a2b574cf79914ce7 : FAILURE in 25m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0f1f66edcfc48479ea5d40add1d0a22 : FAILURE in 19m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ff71b00e6850423087fd3eda79b32232 : FAILURE in 21m 29s","accounts_in_message":[],"_revision_number":7},{"id":"cce91b79e054b58c28b54a5727f3e28282874400","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 10:22:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/96383612230146058546eff8db7c3c19 : SUCCESS in 1h 44m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b524a104b43f45a48212a105ea04227a : SUCCESS in 30m 46s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d267fb818fe42ef9551caae98495993 : FAILURE in 1h 01m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/340c26770061424cb529e7fda0a706c7 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb8bd9c7a6c3490ebaedae6ed84e450c : SUCCESS in 5m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2badc367dbe84664a2482bb5ec04c110 : FAILURE in 31m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dbdec4501bee6cdce330c59149b473bb1adfc78a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-25 06:38:38.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"2cc28cd2b3460908ad572d6bd52f358e54e54e7e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-25 06:39:00.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"8f83c0c0c178c9627a9cf3a704d74ce9f3d4b1e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 07:42:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/938fdd09f6a74cfc9ac81c7b4a094baf : SUCCESS in 22m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d595d4a6a88a4e60ae70dac5de934c4b : FAILURE in 27m 35s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0b0913ecf83349ab9dcd0c88b4a8ac4e : FAILURE in 1h 02m 37s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c5f12d1d18db4123a943af1f32e4f95a : SUCCESS in 19m 43s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8f5cdbc3f5054c10aa10f639471050fb : FAILURE in 34m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c725d65a02ea4d98886c3bc1687f702c : FAILURE in 53m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab96d43ab68d4318b5ef30fd7ae0a991 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d01e285e0cc1467f88ed8585ecc8bfc1 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e7c1608d687482390ab2db4b7b2dc54 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a68b87bd05b249a1992b5cb82d18a630 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fa6583ce5dd463d805cd2a169c23da8 : SUCCESS in 5m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/244861467c664d61b5e9ec92f265d007 : FAILURE in 23m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3a48c4c6b3446e99fdae894346b5f2f : FAILURE in 20m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/64597599b67f499884481378135af2da : FAILURE in 22m 09s","accounts_in_message":[],"_revision_number":9},{"id":"093252a8b51d1e4b2d6e91ac249b855acb824a53","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 08:16:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/35434683a030475bbdd6e53efd61f7e3 : SUCCESS in 1h 34m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d3af0671d7db491aa4d944a403b544cd : SUCCESS in 1h 19m 16s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d8384bd5e6b04886b0572b2eb595e431 : FAILURE in 1h 17m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4a1ff29c23348fea07809c18ace4923 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/984bb67b4879497b89ab7efe8c3f9b23 : SUCCESS in 6m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a643dd741b014a75bd0e61c54c2baa52 : FAILURE in 1h 04m 33s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"49befef555cf64900dbb575f8bffd026dcc9e38f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-25 09:36:24.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"72b500a175b24ce7968dc52f2d7cfed605756f06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 10:41:39.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/e6a4b12761b248bb9fc2b700d8993cb0 : SUCCESS in 26m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/80efbd8d672742f395dd379cd306fdfc : FAILURE in 35m 22s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e1df94aceb3d4732a2fb9845565ed51c : FAILURE in 1h 00m 38s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1827ae9ef9904c09973a2ada3829de41 : SUCCESS in 26m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/36a5caef7d904cd0a4969399b9538bbc : FAILURE in 37m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6ac0069bd10b41508d8a5e73805d43a9 : FAILURE in 52m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2427658063954b8c8f2fe751e6edb72c : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1942646ed43e4ba58560d2f63777459b : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3739125bfe4c435f84382147d16a1124 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a420eed1ef3a4ce0a04e978933ef5a09 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a069782485b747d4a6f6139f52b09299 : SUCCESS in 4m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5e4948f01748478690cb766d59719754 : FAILURE in 21m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e304c68280e463384df0a4cf66a9d9a : FAILURE in 19m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac2668a60f9f4f048253d1aad20fdb2d : FAILURE in 24m 51s","accounts_in_message":[],"_revision_number":10},{"id":"f154ed12d2baa96a22f3d72408ba639306879ee2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 11:07:38.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/ef02e47ee252469e935caba60f317fc1 : SUCCESS in 1h 28m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/be1b3147de5c49068606cdd8deb7207a : SUCCESS in 59m 06s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c9a8f7414ff4486f9780c69884faffea : FAILURE in 21m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47c3132755534eb786bcc6f3a663e68c : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8d475ae1fad47bc8807582a5026bbb8 : SUCCESS in 6m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1b2bd9d276bc4411b1d940ec6ff672d0 : FAILURE in 1h 01m 33s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d2c2fb20a8ffa82e576cec4478434742000d5bc0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-25 14:52:37.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"0ee8aae3ff4fb01a066b760a5debba72e11deb1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 16:00:13.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/cfa34014e1704ce88335a908d5622608 : SUCCESS in 27m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7459552f917a4c10bb173c05dacb3fd9 : FAILURE in 33m 16s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c415bb91d3e44a89bf6a6bf14a2d05bc : FAILURE in 1h 02m 20s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ec1dedae1520402f960b4cdd4945755c : SUCCESS in 22m 17s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ef2f66d9d92245c89c7b7030e89ba414 : FAILURE in 32m 10s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/56ee7e0b8f3445b99a4e8923927225e7 : FAILURE in 58m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2ebd71317ab448794143193b25fdb7b : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe19dd38cef94f5c910d75fa0c64f4c9 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef54ce923e3b4632bcfb94ea45af35ca : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8013fc0d02524b92a49619e2cee1cb03 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02823a0f45454d0fb518a3860bc4d1a8 : SUCCESS in 5m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ff5f3e845f0445888f33e2261a3feea0 : FAILURE in 20m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4986d61364545c69991d0f0be61d4ed : FAILURE in 19m 50s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c1e0399244244e4481d63b01c4a228df : FAILURE in 20m 17s","accounts_in_message":[],"_revision_number":11},{"id":"53b142b6c75630d102a3edf0a80a7b2056be9022","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 16:55:27.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/e0c79bf1c66a4f00a15a62d9435d78ac : SUCCESS in 1h 59m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f70dd05c7ee5477284fba0007ff3029a : SUCCESS in 1h 40m 29s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c36d436fcd6c4b2e9cb4b8b3ad1ec71b : FAILURE in 1h 12m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdb7e6f7acf34024ab1e4e5390b28a35 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b9ae6338083f41fb995a21963e034fcb : SUCCESS in 5m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/724f4f29e08048ca84afce6b44e4afe3 : FAILURE in 1h 17m 40s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"664bc6d42d1c0a6dee834041453d043655371abe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-30 14:38:19.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a89cf953571176650de5a6a2223c3ed641cebfd0","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-30 16:09:53.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"b75b6fff989ef7fd26ceb065d9a16b36f864817b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 16:10:13.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/cd72cc4bd57f436a9adba90b3867b5db : SUCCESS in 31m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9e816cfc561b4853bec15712cb2a8bbf : SUCCESS in 1h 15m 34s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/52c3a345959442cc8d73ec1d23cb28b1 : SUCCESS in 1h 28m 14s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9e3e24e9d77e4ab784e24f21c10bfc5b : SUCCESS in 26m 10s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a97fdb16536d4f1d995a3ed1b918609b : SUCCESS in 59m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6aec924b95994bfabc58fd03fc2e3581 : SUCCESS in 1h 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71a1c476db4e4faeab8e4eddaa8826df : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f3150a21b4842ff8eb5ff74331b7082 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad249eee11494f93b3db6d47f27be2d6 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04c738df4bcf4dc79b1c08136b2a892c : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5ffc4a5ef464cb5a0d79ad86dd19e0b : SUCCESS in 5m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2a57314d16f74a94b53e2bb7028272d9 : SUCCESS in 21m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e39e124704c4ad5809b7643d4f5a091 : FAILURE in 30m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ed5d457c0d7646efa66e88cb94966f40 : FAILURE in 1h 01m 19s","accounts_in_message":[],"_revision_number":12},{"id":"edb9fb37049ddaa3c962c61cfac2077b0852b364","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 16:24:33.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/c1601e7ece8f464cbdbcd108386c708d : SUCCESS in 1h 42m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/61fed7d01b914a0382ceda1ede20b28f : SUCCESS in 1h 12m 17s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/340447a1611b4d49aff4fdecf8674891 : FAILURE in 1h 42m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74f108091c5f4c3d9468f4c7ccc5c5df : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e61220e45e084acc814c0307850b95b5 : SUCCESS in 7m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1f280d53290f4692b772a6c0a27a1485 : FAILURE in 1h 09m 35s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"72079d125f7d7e7bf797849964795dc68c1a1423","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-31 06:51:34.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"8aea0700391654989173ce2669e320875f237305","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-31 06:51:45.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"2bdbdac86f5189217fa7a14fa771d807448dda5c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-31 07:14:52.000000000","message":"Patch Set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"f5f5a204ba3f3cdaaf7217e631376992e928d50c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 08:46:03.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/5f55f20157834afb89534cbeeca0f7ef : SUCCESS in 24m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e585df51af334270a829ca458a53e1fe : SUCCESS in 1h 13m 59s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b5d4baebade448b782e6e4bf003dcbeb : SUCCESS in 1h 23m 33s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cf24a417629b4ceba44583084f246de3 : SUCCESS in 23m 24s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/add2398d64904cbdbc2200fbe50267c7 : SUCCESS in 1h 04m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/59643d68904f4a16bccbf4bb6a31a72a : SUCCESS in 1h 24m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cea10d2fc8b4426791c06dc69c2c49b2 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a54b76e1e52f429ea7f75b12d55e71ef : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3965c05746a44a06884f40f4abad9ed4 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10f9ebe1c83545afaaa21a09c4b7d69e : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a4a5d98eb84e469decc23dc5ab75d8 : SUCCESS in 5m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a0bcb3eec39d4d3e926980144c65c5d5 : SUCCESS in 17m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/da53facde5a442b89f76d797744c2ed7 : FAILURE in 43m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cc6283f9a380429795119eda4d5229d6 : FAILURE in 1h 18m 08s","accounts_in_message":[],"_revision_number":14},{"id":"1489eebe3448eb300ecf028b66dfed1b4c809592","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 08:56:08.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/3e59773bb46c4c8d892928072eda32d1 : SUCCESS in 1h 30m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2ac46a5be925492f8ae2cfcb710ce32b : SUCCESS in 1h 02m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e7ee860a65c54ee58253d0f43a12fc49 : FAILURE in 1h 39m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/270131ebd4d64e0aa2bf39d85f6ce983 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6633b9632f704f43b87f87d1d206b326 : SUCCESS in 6m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9d6da794ea904df4aa7b10ffe72bccef : FAILURE in 1h 22m 04s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"cde54efcf38419dcc51730974209b62d28587d0b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-31 08:58:00.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"c27390c8cae107d3d73bd149d7346204f30f610e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-31 08:58:16.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"d3a89b9d943017bd4554898478ea78a0a8790121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 10:24:58.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/a36219883a7a472e9496f5ebea96840c : SUCCESS in 31m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6af3bf17743f4c16807ef74c70fc914c : SUCCESS in 1h 13m 03s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d1d980a9ce2248c4a95266501067a386 : SUCCESS in 1h 13m 07s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/269e9e84c41743179dedfb6946e3c22e : SUCCESS in 22m 46s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2c43218f225149c2a9a03442a02834d8 : SUCCESS in 47m 30s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e47cab720d8645589681c1b7c935aaf2 : SUCCESS in 1h 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ebe038a0d64ba58fd604e025372d29 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1243ddfe6f724cb1bbf64ff760e29e94 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9085ddbb2654338bc0a2507fc993541 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e6d2389a66f44dc8f936c80064dc102 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e90c2056117f42ed950034b48f5e717b : SUCCESS in 6m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8b6e8b2410c8466dbcd47cb5a640c387 : SUCCESS in 16m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/990ca076df3a4414bb325551f10d66b8 : FAILURE in 45m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4495746b01fd4da89a166ef4613df94a : FAILURE in 1h 03m 13s","accounts_in_message":[],"_revision_number":15},{"id":"e4cc57fb00596536bf208789f3460e6242cb5613","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 11:03:08.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/7e7620e57cee4801981edf3037e61399 : SUCCESS in 1h 44m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fa9db53309404046a0736d5494fd1666 : SUCCESS in 1h 31m 43s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b782d3b919b248a6bac9c4a8d05cb7e5 : FAILURE in 2h 01m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cdf173214c0465780243e09bf693aee : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74d3b7f6509046da9e37113e766763e9 : SUCCESS in 6m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/79ad32e5937a4d7e9c6d39616b4d0173 : FAILURE in 1h 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"46e581fce955c29d41c863e40545477b6733654b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-06 20:14:17.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"29c638e49156fd9f8ae278c1e76c5c05954405f7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-06 20:14:43.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"cbd6d54f8c2348ffef195decc98f463aa5a0d58e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 21:38:00.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/359dbf88bb2a4d039c09330ee9bd6947 : SUCCESS in 31m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e4caae7184384840807785c27cb1e04a : SUCCESS in 1h 11m 45s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cca6db9e885d4f40be5b2566570c5762 : SUCCESS in 1h 05m 46s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5c528e41670e4cb98bf76874557c1540 : SUCCESS in 24m 20s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4bde1d6183ba4750a8cf7c2ea5f6b65a : SUCCESS in 59m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/80bad5c5fb19408b81a997f7d768889b : SUCCESS in 1h 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22d6b6278f5f436b895290d6d6bbc995 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cba061db2ab4cc7a49566cc7430bbc1 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e544b4e94d14b1d8c145a8b4507f0d1 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c70206cdca4d4f7881e5bfee0189a9eb : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0282f46fd971485c97fcea47f99087c9 : SUCCESS in 4m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/863ab3f25ebb4ccea38e67af7f6dbe60 : SUCCESS in 19m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f04e4a79d6f1448091beb1ec20dac8f5 : FAILURE in 43m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/725cf2d1ff2e4dfea0006df38ea0b3e1 : FAILURE in 1h 16m 08s","accounts_in_message":[],"_revision_number":15},{"id":"387edc5e5a71c40cc451c36c735df4255d07151b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 21:56:43.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/12fcb9a04f6a49188fa78d6eed54b7b0 : SUCCESS in 1h 28m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/519043cd4672465e818e765609a5d5c7 : SUCCESS in 1h 38m 41s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2382dd8282984f30807c35e1ff829226 : SUCCESS in 1h 39m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e614fbf6d321428b87ca5612877adb4b : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a8f8e3fa3664117ab6d622b2552b69f : SUCCESS in 5m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ea5fc8d40494d36ae033c989b0a71ad : SUCCESS in 52m 25s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3df29652d49ebde8f185075cf0cc7396fa3f180f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-07 12:40:19.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"7055f5f6b3a3485465731c5a41701b3bf911e78d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 13:42:10.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s https://zuul.opendev.org/t/openstack/build/2fecfea3aead4643a8a8d3dc0cd4c672 : FAILURE in 17m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c159da4099b04c03947afaf2d709a11e : FAILURE in 10m 34s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2489b0e33a234d2897ecd45cf6366ea0 : FAILURE in 24m 32s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d7207be3a86b44a8a21a1f8011e0476f : SUCCESS in 18m 32s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f9d1cea0fca4468bc3e4d09cdaa906a : FAILURE in 35m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f502ae94a2da4199b4ad6af18f37a452 : FAILURE in 48m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37bace6122d543dab1c4d19ae17a4178 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c01f7753f074cfa941f026cbeeb9585 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf8ed9177a67495ebfddd81f0c7f98da : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cafd86b8ad648be8e34858bd44db136 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a517a68934a946279744935a52fe6b1c : SUCCESS in 6m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4239b345b2a7486a8ae6f24975717cd9 : SUCCESS in 29m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/33aa1e5327b545b0b20b29303dd487cb : FAILURE in 31m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e5ca9cef9ad3498eb659bd6b489e8058 : FAILURE in 58m 58s","accounts_in_message":[],"_revision_number":16},{"id":"bc61aba448d5e8690ecb72821e4fc1676e992603","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 14:25:29.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-aarch64 https://zuul.opendev.org/t/openstack/build/c07bbec255e5494fa869458a2872c020 : FAILURE in 1h 02m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/492762e02e0a457db0450f87fde9c759 : SUCCESS in 1h 18m 14s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b9aadcd9d88f419fb1ec376edba26116 : FAILURE in 1h 15m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/65b0463eb6ef4b26805106f5bb491115 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e53e6e65d71647e29d37ef11ad20b25f : SUCCESS in 6m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8b395cd1d56c4b819754c6a0c3a1ed2c : SUCCESS in 1h 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"37139fdf6481cc37f18788dcabaa9a219efad5a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-09 20:48:41.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"d85fc1f1b3260567fd8db498699fa563cacf4b25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 21:09:02.000000000","message":"Patch Set 17:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":17},{"id":"fd46509a55cea1f949cca07c7efae63a7823f647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 21:09:02.000000000","message":"Patch Set 17: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":17},{"id":"ae8e11908236b07df429275732d8163e683e92ae","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-15 05:34:56.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"928f897eb3cbbca1b4f45ff60e4d8c409ac0b9e0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-15 06:47:58.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"b6f6191b8a0e7f274ad503ac7202f6ab1e5eb63f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 06:48:37.000000000","message":"Patch Set 18: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 7, column 17","accounts_in_message":[],"_revision_number":18},{"id":"8be97d3a0ba7437c76c3a93b49dc5e857a3dfca6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 06:48:39.000000000","message":"Patch Set 18:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kolla on branch master.  The error was:\n\n  mapping values are not allowed in this context\n  in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 7, column 17","accounts_in_message":[],"_revision_number":18},{"id":"cd04d1564e9ab50762e7ac995861c08fffa77780","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-15 06:50:45.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"45de94872be0b7723bbe6e7c999fff41cb6086c0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-15 07:45:17.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"44f879b8c42c3e5cbc0541e10fd01fdc3eb9abfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 09:05:48.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1ef4f5fbf4a544c98c3e45b19f5e21f7 : SUCCESS in 22m 53s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/38bfa2addfe4424b925b32c5221f6802 : SUCCESS in 1h 04m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/568bcc4b406545039c7a0a0fb99d38ed : SUCCESS in 1h 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33cb7dee551942019b6be1547d52bf90 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2504222183345dfaebf5e77503aa4c2 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0357a16903e5481d8dfec418b58abdcc : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0e0c6380ce8464eb01fbc7e64685b79 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/529f0bc41aaa434aad98756ef6765b49 : SUCCESS in 5m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a2d1eb5791ee46bf9ec501963c07cd76 : SUCCESS in 22m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/741ba9fbbb394c11a73fb271d2591bb7 : FAILURE in 47m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a98a7e42dad540119439d5c01d9d25b6 : FAILURE in 1h 12m 25s","accounts_in_message":[],"_revision_number":20},{"id":"f10af78e0c10728f14c60140d2020ec2b28f2990","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 09:55:01.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6968e306bf9f495a8dbf1ab1333cfc5e : SUCCESS in 1h 26m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4e8ba82b328c40e2b28267ae229da542 : SUCCESS in 2h 06m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00d8d394635a41c2acaf11fb63936f2e : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bca44f078b24571b62fb6beca318e30 : SUCCESS in 4m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/676a3587a4804a5bb077c5fbd01871a6 : SUCCESS in 1h 27m 58s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e2c95d3c6425f01b044f9f18a77e320a9d003dd9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-15 10:26:10.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"da1c7031d3ce9e845fec4fe2e161670bdefccd96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 12:07:02.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ba56540f403c4a0e8b9813d7739669ee : SUCCESS in 40m 51s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf822bcd27674a51a4eefadd5a73cc36 : SUCCESS in 1h 32m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ca75444bf5f4b329b400e5ac22b7476 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48c06ab9c54c4d82a91f355f523ac51c : SUCCESS in 5m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f49c99ec14844ed580d3ad128337ab71 : SUCCESS in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"ac8b8c2b692202f8edc24bb16496065e0806044a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-15 12:08:27.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"266e79bf144ea4e537ae2b31aea85b8c0379a67b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-15 12:11:05.000000000","message":"Patch Set 23: Commit message was updated.","accounts_in_message":[],"_revision_number":23},{"id":"ef34c9818df8438f1ca263e3bd1283ce2bc828a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 13:32:24.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ecd9ba1db86f4028a1a00341e3d5aac0 : SUCCESS in 23m 06s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1a8eb26ecf6745aab7562a1439184ce1 : SUCCESS in 1h 05m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bcc205c47f734d139b9a41cb85d47069 : SUCCESS in 1h 16m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce5accffb6de47de9dc9069b2405badf : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebf5107bd8d5437c8eeca0ecfbfa9496 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b6d2831abe4416fb7f7affc7f2c5c25 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e84192813eca484fa762f47a0dc0378e : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb37b8dfb68a4195a1a35f7451b55ba5 : SUCCESS in 6m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6ef99a17c43547cb91c3debecacf5eb4 : SUCCESS in 21m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd5a051847e5427e897fb160cfae8652 : FAILURE in 27m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d355cbd3442147439fb720dfb58b757a : FAILURE in 1h 03m 36s","accounts_in_message":[],"_revision_number":23},{"id":"a3c638275cf5bfb66557da3acaaf75d76998e9f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 14:16:46.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b63dff5d561e4dfa858b11b2eefc498f : SUCCESS in 1h 13m 40s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb0098d0687945f5bbb69ced2260051c : SUCCESS in 2h 03m 01s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c7260ea70994999b2dfbba750b25d7d : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5357a5afc4645f9af9e04efb3559957 : SUCCESS in 5m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/38db1955fffb4a10a61b33874161498c : SUCCESS in 1h 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"8736eb50da8f792e08619f8d1fe0322878ed829b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-15 17:20:28.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"ba60a09e2ef803e058327f98eca4fcffb03ae4f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 18:51:33.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/95c81d89c6354b7dbe1aa80b385d8f97 : SUCCESS in 18m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b8d7fd4de7c14f8d91c284aa4e4ffd23 : SUCCESS in 47m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0b2654a092c24880b7a1b49fc1b428a4 : SUCCESS in 57m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a0401acd890405684057d489a27378c : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb29ce3458924b6b9176aff328258b61 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce18cb4b89c3454fa162343eb4b10d37 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e45c182e228a4fa18e8ca4fc7cd4638b : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e0fef1ca4848f3ae41533e05fee74f : SUCCESS in 6m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0d0091f9e790462bb0bba1e88077ae62 : SUCCESS in 25m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec78d5b6673e4c5ab9bb65c950106e1b : FAILURE in 22m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dc293c3ea57f4c1fa9912f5bf3a3b6a7 : FAILURE in 19m 38s","accounts_in_message":[],"_revision_number":24},{"id":"2017921722a003d005d9031297c5b11acee1851f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 19:19:24.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2a957b5cc4734fd2abb280e04c263447 : SUCCESS in 35m 08s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d13b2b91c6f64157a4430a209ebc2016 : SUCCESS in 1h 56m 25s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f30017dde90c4c10a758ac4f08acc9a6 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4ccd51b633e460bb9e0d3bac0793cf9 : SUCCESS in 4m 23s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/621a31c0afd845b19c61424777bcbbba : SUCCESS in 33m 42s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"61fd439d5e51cc0f164c01c33f28e0373a6f8e0d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-15 23:17:11.000000000","message":"Patch Set 25: Commit message was updated.","accounts_in_message":[],"_revision_number":25},{"id":"ca2db4bc2a76b84464234cf90c4a9f46b86527bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 00:39:50.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/23ebfbca1bed410fb8ca0fbcbcdbbd60 : SUCCESS in 23m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fbfd1d6ee518481fb24e344823afc98b : SUCCESS in 1h 03m 14s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/447320a7d84e4fe5bf77efe0a21d67bd : SUCCESS in 1h 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/070a0995adc6470fbc961486311c63c0 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d630d7f98f4436097c98592682794b8 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1240d38dacc6490984052963367cec0b : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2f568952b6c475b83bdbfaacb69c1b7 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d22a2159b13f40fcaa35394ec2041978 : SUCCESS in 5m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e274fd0c98184809b121e42ed6c1c9bd : SUCCESS in 19m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/411df2d89ef340ab911c42f86be05af0 : FAILURE in 20m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/781ef7f95f4144179393c1ea906ac342 : FAILURE in 22m 02s","accounts_in_message":[],"_revision_number":25},{"id":"f06761655f809eef182caa6b1bc748992b8d9a56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 01:31:07.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9abbd920320547fc9b0e4fd5735b91f1 : SUCCESS in 33m 03s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d26edfe949884affabf90901193dff9b : SUCCESS in 2h 12m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/069c4b8b04044bdf83f9a4c2d79ad2b1 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed4bb5d2fa8b4650a4b6e6d8135e4759 : SUCCESS in 4m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1408c2ea4032479a9e922d66d74f77ba : SUCCESS in 1h 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"318b37c6cce0548d6024b85e663d2b88fad9e83f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 07:03:24.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"f36f072ab2a86c453158ce6b8afaad9a5c96310a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:19:10.000000000","message":"Patch Set 25:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":25},{"id":"832bf933f947a443e8b74f427001c0d94b9576b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:19:11.000000000","message":"Patch Set 25:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":25},{"id":"35fd5c9ab804a5529ad426dab7d0bcb93947e5e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 08:20:28.000000000","message":"Patch Set 26: Commit message was updated.","accounts_in_message":[],"_revision_number":26},{"id":"0ebfb9e2854c61c0ccca85f76e22d3ce2a23ea5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 09:33:21.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/01fd49fe60db4c88a96e2c06554d09a4 : SUCCESS in 19m 56s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d09ba7d089544da3aa09f0308160804a : SUCCESS in 1h 02m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c3932ad3969748cd953e2cc649382fb0 : SUCCESS in 1h 08m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a314846627241c2aa658e5986c25d54 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b3b00540d474bf69a32805ecff9b002 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1ccdf3698d8497e9923476c74bd7d09 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9f2215a278240d982b05ccce25044ac : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6a5f8785034721b257550ef6fa524c : SUCCESS in 6m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4979b7b271cd44298a12a3fea4bb1146 : SUCCESS in 27m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4b65cdcb1794a8babd5199a71372f16 : FAILURE in 20m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6d9d366478a448e0847a7b9c359b3b46 : FAILURE in 23m 41s","accounts_in_message":[],"_revision_number":26},{"id":"7b58d33dc448888d7ac78da19f93a38c4b40d7fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 09:41:29.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6704c736a2c745a5b0ecdcbb0def67e5 : SUCCESS in 1h 03m 53s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bef61a69795a47809f6d968fecf3a20f : SUCCESS in 1h 11m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/afc33851cedd4f6fb39d52acb3d95dbb : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fcdcd4f0a2cb44f2b94fc5128b94f563 : SUCCESS in 6m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/38209b894349490199d4eafc0ae27a9d : SUCCESS in 56m 58s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"039b0b01967ced4435ec91bc33a65e1ee66f9838","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 11:40:55.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"6d826eda341e517b88fcb40149e98f097b36335b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 12:49:53.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5e416199614e4cd3875689354bd4d8a4 : SUCCESS in 22m 07s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/298a3e041dbb4eeaaa92adb2db6ffffe : SUCCESS in 49m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/73883d8fc2c245e690bd8d6870a8b843 : SUCCESS in 1h 00m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92cab2383fef45ab80f75afa2a5bfa0f : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62f638e9837f4016abe95da2e07a6821 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25e79d5ff78d4a68b12d349557c57e52 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8e0284cf5414f25bb84fe2e028eceba : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c07bb4905a474b827eb8aee3080f91 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/006405d9bae048bf9ae13a489390a206 : SUCCESS in 3m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/524fe7107e6d4a6088003d33174d4ab2 : SUCCESS in 19m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a23a93271a44abd81c000ea6d52d671 : FAILURE in 40m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c56b16397c1d4b2597cf2ef68a9a841e : FAILURE in 18m 09s","accounts_in_message":[],"_revision_number":27},{"id":"6d2da8dd14f3a28e69a505d325d5d1d6341e967a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 13:06:28.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4fbec45e92c34a15b0380c8dc77d3c84 : SUCCESS in 29m 55s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/12e87f0562694bc693ebce5490852b58 : SUCCESS in 1h 18m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da3e0f7933c94f02ad2543f17a200cfe : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75abfd1575c34891acd458c41aa1b2a9 : SUCCESS in 6m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c11920a034d344b186cb055268ddc857 : SUCCESS in 28m 19s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"a9571bbc6b78bc3fa40fa7bcf9a6bc5fe0310255","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 14:52:42.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"9c1b2cc170291c36dd62e72c136a8242b768b0a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 15:52:36.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/25141971989849ec8755994b90bba0ba : SUCCESS in 30m 20s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bbe84d72c5f643f4b2d2855ae261a4e6 : FAILURE in 48m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/567e9bf2ba6646d9aa4a909f70af6aa4 : SUCCESS in 57m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e715a390909e4adf93f8e1a1345f22f2 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d65aa8f173a40e9bb788c1a8d25b205 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b06534a07c834456b459830f3ae97d4f : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0cd5d5a5cbd4b73a94d300ebabbef86 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/647a469687f240848aca9f2cde351493 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da6bef5e3752413495e76827f60893b7 : SUCCESS in 4m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1b49fd5ca9e748778b26361e81ea220c : SUCCESS in 26m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/31a1085b97ea45a9afc420543e617217 : FAILURE in 42m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dbf0ba23cd29420a8b56a0ec6feff2da : FAILURE in 18m 59s","accounts_in_message":[],"_revision_number":27},{"id":"29a7468321f20db29eac6aee97059d6a15f7e6d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 21:22:13.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a9b1bf1f950f4cf79a30e25a7b753c2c : SUCCESS in 48m 06s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f4241ae33fc4f3889b8aca76b476351 : FAILURE in 1h 20m 51s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a88f57a52174815bef61ec33db60fb7 : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0444964681d14d4b897a5422a172574c : SUCCESS in 15m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ddd857b549874fe281b941bfa086c3a0 : TIMED_OUT in 6h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"48caa8e20063a176b7c0246027c3bfa6ebbac9d3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-17 14:22:55.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"b6a76e98d637dac58ab74be1770673990b701d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 15:47:55.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cd3eff0e8fdb490d8a6fd1c688c4bcfe : SUCCESS in 19m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/927a3d869180402c9f665e00f86826ff : SUCCESS in 1h 01m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7e6eccbb936d49599591ce4573c96498 : SUCCESS in 1h 13m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8e2bfc4601e4cc1ac79da617e9a4a0b : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44cdb065cd184d70a055190e0882cb74 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf1c7cb935c744698261684d555289c0 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e5b11fb07aa489da11f49dc3d9b1213 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d75e3bf4ff134143882bb8f9704f6195 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/262b12fc40e9416dbfba69cf687da534 : SUCCESS in 4m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/26fce7fc01de4f0eae0470676cae8b97 : FAILURE in 24m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1ab84b7ca4414b51b6771f922e56ad7e : SUCCESS in 52m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4fb9b70c5d854624a5b6b114f3b90ad7 : FAILURE in 21m 07s","accounts_in_message":[],"_revision_number":27},{"id":"31a09ae59d3262fc57dd86c3760f8f8da7e4585a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 16:23:55.000000000","message":"Patch Set 27:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2a44cfb56a0c486b940e67f9b8a9fb71 : FAILURE in 1h 41m 33s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/41da49cceaf94531865c9acfac901e34 : SUCCESS in 1h 54m 40s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e96da16b64544c4dac316b6e52819db5 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf446c401ddb48cdb2b23ed0077d9628 : SUCCESS in 7m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8b7f2fa2ff8948b584f8f1ac5ef28032 : SUCCESS in 1h 22m 47s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"279ceda77def19407c48ed6bf6dfc48c6cd0a13d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-20 07:39:03.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"283da5a98985a1f4cad381680c40257fd13cbd03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 08:55:51.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/166e6a26d5ec4e858076ef0f4e91103e : SUCCESS in 26m 50s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e221199e9d8d47b48b4afbde6bf46021 : SUCCESS in 1h 03m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/dd1a015d099e4b78bc6cc131d3d4c221 : SUCCESS in 1h 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea6547ebbb404f8eaf3cb97c32c023c4 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c091266946f4b47be54f6dbde3ffa99 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b818ddf1c5e4270b04781daa385ac24 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22bff2e3ea3e42aab686cdf56db6584b : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e064f1bb439462ca2bdee667ffb2681 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df5b87ad10674350bf14b7da22fed345 : SUCCESS in 4m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a7693b86e3254e10ac040dfd8909882c : FAILURE in 22m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b535183bd144a2e8e1d268a41e23e55 : SUCCESS in 57m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/90cc7b06b1384df5a2f14f9bfc919347 : FAILURE in 18m 41s","accounts_in_message":[],"_revision_number":27},{"id":"e174cebc72032c6f1a312a47dffaf4e6b82bf8ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 09:24:02.000000000","message":"Patch Set 27:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/daaa1d46d8854646b5a8b58bff79ba92 : FAILURE in 1h 26m 22s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6aaa9eebeba24e9d9568f9b25ba106bb : SUCCESS in 1h 38m 56s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d42a8f719a084cd0882bbe396b908908 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89091b14eaaf48fd92bb0f60a50329ef : SUCCESS in 4m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/24fb057867bf40aaade25cdfa868874b : SUCCESS in 1h 13m 43s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"8f6a4b177ff00ad6838849e22209e4b44f06a60e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-20 14:05:43.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"3f9b2c933d54e0eb0616dc92bf8bba7e876d9b3f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-20 14:06:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":28},{"id":"2656ab0f00c20e2402ca2c9055adb2c7df7d01e5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-20 14:06:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":28},{"id":"9e4304329b0c5b75fe61deeb6b8905251b712bb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 15:32:19.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f828ed2c52f74e928c67911652317b6b : SUCCESS in 23m 46s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fbf3c43570fc46f790533019b6c0fd8f : SUCCESS in 1h 09m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c2aa04c0b9aa40d6bde5e9a9d76e8ccb : SUCCESS in 1h 21m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa7b6866e0314ffe9179f1c415687c21 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73e8446467fa48539375d4f763c0f00d : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a22df131ed8b4a5aa7f55025ac51db13 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6468965cbcb843adab85260db228b6ee : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d01ecc4f16342aabf6c71bf65dec586 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b434a7d34a9b4cce916a27081986deab : SUCCESS in 3m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1e5b2730d44d478492f00c86032b3b06 : SUCCESS in 18m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/740c8ac8e41445a888f742ce02f16930 : SUCCESS in 57m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/964cfe863514499c931ae5baf6edc4fb : FAILURE in 18m 16s","accounts_in_message":[],"_revision_number":28},{"id":"bc4ae7d1cf0f83b50362eb034818367978013368","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 15:38:55.000000000","message":"Patch Set 28:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ee09cd96bee14f25b1f79eaecce3f9b2 : FAILURE in 29m 13s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/39eaae9c47e74fad9bf5f6fcfa1d34f6 : SUCCESS in 1h 23m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c430b65f36442debc88ec925b7ae475 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6790c8e16634c46aaecf22a78793026 : SUCCESS in 4m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6e99ceb0e9fd4ebf8b19e1f2d3d50cf7 : SUCCESS in 24m 52s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"3797cd1754efb687d8da0c7085088ff0950e6b40","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-20 19:43:16.000000000","message":"Patch Set 28: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"5a427b835bff193d750426b3b220fb23632eb88c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-20 21:16:46.000000000","message":"Patch Set 28:\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"9c9f5c1881b7339c69ee39eef9056365bd3fb5be","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-21 05:46:48.000000000","message":"Patch Set 28:\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"85ca96a630c8f6405afdaf0b5a572a22db090519","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-21 07:27:46.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"663e918cca321850bd9889529fff4d39a6929772","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-21 07:34:32.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"403a47db3ca3107edc7ff056bebf45b2fa89d7a5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-21 07:36:35.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"ab291beda57714755360831747d8d519131fdca6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-21 13:30:41.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"5333898570cdd85b9a4a2a11382c95ce66dffe22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 13:31:12.000000000","message":"Patch Set 29:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":29},{"id":"488e992814fb79cc556476f7a1c5c15de16e4205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 13:31:24.000000000","message":"Patch Set 29: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":29},{"id":"ffc987162477aaf37c1876deeba73518050afd32","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-21 14:54:48.000000000","message":"Patch Set 30: Commit message was updated.","accounts_in_message":[],"_revision_number":30},{"id":"e8c75ec16d73974d1a8a4d3e217d9037023e2b45","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-06-21 14:54:52.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":30},{"id":"b5387b3559dfa2b33a5030308db3fa551316156c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-21 14:59:02.000000000","message":"Patch Set 30: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"380753e1524e601401efa385145799aa23cf7dbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 16:22:00.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bdcf3010143f4f949e15acb13de5bcb1 : SUCCESS in 19m 45s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/afd743db9c7d4d9ea25fb3e67c51853a : SUCCESS in 44m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d3f6c085a8b6484588df57acb5948290 : SUCCESS in 1h 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d03af4ed9f434902a458e1a48bc663de : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da5b46fb1dfc4e018a65db55f163f225 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3125c44252764c6c9f0896df07ab890b : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea00cb4c0f534911b9389465fbc8b940 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62eb2a61d9d4eb5a6264fa81cdb889f : SUCCESS in 6m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d93bfa53931440fb402867d7f1b6037 : SUCCESS in 5m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e15d3af4a05745bea9ee73d11e5bc4e6 : SUCCESS in 28m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/08d91d684db941df911170bf54bb2693 : FAILURE in 42m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8dd8ecfb561144c382f6764fe5cac0e0 : FAILURE in 1h 06m 58s","accounts_in_message":[],"_revision_number":30},{"id":"1d4f1a1a9ab342db72955c483680bd3600322d05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 18:52:46.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/54de40246b644ab98991aa546653eb29 : SUCCESS in 3h 06m 57s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/62343cbd11e84b1fb120db8b489724f4 : FAILURE in 1h 49m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72e5809769f54453ada03d5a19651633 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90badb8ce189487fb3d8c8bd6ad441cb : SUCCESS in 9m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e293c4eaf1214d888fd13d45eb6ea1e4 : SUCCESS in 1h 22m 13s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"7ed98b28a90bbbb45c049aaee888b6aca8bec166","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-23 09:48:25.000000000","message":"Topic set to ubuntu-jammy","accounts_in_message":[],"_revision_number":30},{"id":"4b7662871e862ecbd6eb84e3b7648d9c2d37daa8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-23 10:00:49.000000000","message":"Topic ubuntu-jammy removed","accounts_in_message":[],"_revision_number":30},{"id":"0939f234c9b0cbe93a53cc0847880eb7be009260","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-06-23 10:00:52.000000000","message":"Topic set to jammy","accounts_in_message":[],"_revision_number":30},{"id":"9d18cb931e2745709a7f65980eac166503f37474","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-20 12:13:30.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"86026fa84468ce73e13bf7120e4c3aaeca7a32a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 13:28:27.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c8d8ee12d124ea6ab99d7cce470e881\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9ead70da952e45cb9fa5bffb0b44bb93 : SUCCESS in 25m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e8bfc29709a84a7ea3850880ca874b54 : SUCCESS in 56m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/75909116be20477b9c2dced37cd5256d : SUCCESS in 1h 09m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02146d094a8e466cba5571cb15a76a0f : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26e7c983c5f542859c31ecbda1f3cd2d : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1495dd2f42145578c5cf6a4bce3bf14 : SUCCESS in 7m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff7a1c7727c241dc92c4dedf83974e2b : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a8d8d2de894a1fa9ecda55f5799ba4 : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec71fede8849468c98c898ebea4f3fce : SUCCESS in 4m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/004c3ea6c9ec411aa6ab22f2401a65b3 : SUCCESS in 26m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d146bffd6a3e498bbb7f56c66c6e0ec6 : SUCCESS in 55m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e2e5ff73c02a4d4e8b9398ce12da3b06 : SUCCESS in 1h 02m 21s","accounts_in_message":[],"_revision_number":30},{"id":"0e8520f739ae04bd8c3ca07cdd65dbd380646128","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 14:12:04.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4be3490f5f7f431295bc1cd239d84546\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fa7b579c1505471fb943f5b124847223 : SUCCESS in 1h 06m 52s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b698d3ebd0f4a0dadd9e26f3a6f624d : SUCCESS in 1h 54m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b6c1783941b4fb290abee3c418fe386 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95810ea967504903b02eadb515f8a203 : SUCCESS in 6m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/45e4d9b059b74897abcfe48ae5da077e : SUCCESS in 1h 06m 40s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"2944acf7285d6157a158414ed593e54da58d5e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-22 09:13:44.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"8f9da14b174e1c48f3ca0b00ec526846b4505320","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-22 09:13:56.000000000","message":"Patch Set 31:\n\n(2 comments)","accounts_in_message":[],"_revision_number":31},{"id":"c53a5b175d6f9d2ddfe50f18f04bb7326d557b2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 10:21:57.000000000","message":"Patch Set 31: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e9cb13f426244578db8a5f50e387a42\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9b5fce81484a42eaa48d0aa7162ed20f : SUCCESS in 25m 50s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/573610aee6854715b7ca4869d0dfd359 : SUCCESS in 52m 50s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/63438ca3f763455198753eb70052a112 : SUCCESS in 1h 02m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e240c20965541198899262816dda45b : FAILURE in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d85e5e1f08fe49dbbf13c4a56f44be64 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80ad2967ef2e4c6b8e58b376b305fcff : SUCCESS in 7m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c383f5f4e0cf45258aaec117076fcd11 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ab9593fdff74125a121422c91c57bae : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af06e9202b014b7e8ed557e77ec97872 : SUCCESS in 6m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f6bbc719af6746cd9e3a755507994a07 : SUCCESS in 28m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bd7b93b86ebd44f883f13c13a8bfb4d5 : SUCCESS in 51m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1ddedc227030434dbb8d369866819ac4 : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":31},{"id":"070c58e8fd9f88f2608c4bf34faaeb9bc3abdcc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 10:33:52.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3197f47c8c043ba8010a37fc6f3f3ff\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6cc663109ce047b7a624d4ef837fc5f8 : SUCCESS in 25m 01s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c5d1d956a774551bf075d5925862829 : SUCCESS in 1h 13m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74c29eebc3784671a43b5d2e9ccd3eca : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b83497a7cd874bfa8416e119fbbb3e17 : SUCCESS in 3m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c5b73d68f5a44f6e9ed49a9a68062ea7 : SUCCESS in 1h 18m 01s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"e357c7dbeb8ccc9480df90e1bde8b8d4198afc7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-22 10:34:41.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"4a754eefa4dd597861c96b3eb9ef7e361a0b43f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 12:04:05.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d24b335f069490db1402012f7512d17\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1c4edfdbbf29432e9d2a6cbfdff85d29 : SUCCESS in 20m 34s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7432fa2e2fa34b20a2d90c153f5dd4fa : SUCCESS in 1h 08m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/901518e4d9dd4b369d316f089a300c76 : SUCCESS in 1h 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0041efeae9b40d29a0466ed3e54ebbb : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e733f28fe8f46cb990049c4dcad1887 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc9884c4b3b34d8592bbce4a6e6f3182 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b3b5e0dfcd014cf7a7ef3f2672070aba : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf766a9346124610880a1fc3d32b9d9e : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7b3eb606f7b413e98ad1822d2d13dd9 : SUCCESS in 5m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b68acea356f04c61993cc05aab73ae51 : SUCCESS in 21m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/31745a9920754e29a489895b3eea7894 : SUCCESS in 56m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dca19a9d171d47f898bfcbc56d7f4f90 : SUCCESS in 1h 07m 07s","accounts_in_message":[],"_revision_number":32},{"id":"d933d66cf3f6ab0c81a7f3f361376d268cb645fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 12:12:11.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48b7335dafef43edb6e9a27b8c308fca\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/afc3b7e1940449dc80a284266409d30e : SUCCESS in 1h 12m 57s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6adcdc2727484d0da061ca80af617b49 : SUCCESS in 1h 17m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/68bef4bda32a4555b0c8598f55cf7639 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7de787f8ca27424d8c4442cde5cd0421 : SUCCESS in 3m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5c84650173b64701940692a6ac251814 : SUCCESS in 1h 35m 08s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"0184c59c39335078252b79fd959b79da630e1e2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-24 22:22:11.000000000","message":"Patch Set 33: Patch Set 32 was rebased","accounts_in_message":[],"_revision_number":33},{"id":"c527db490f0bb9d24bd15c5555ae9a911a183f9a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-24 22:24:02.000000000","message":"Patch Set 33: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"9d9817ef7f7580953908ebbab63714d2ce2a3efd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-24 23:46:56.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/137d0c293de84af4b564b667120be72f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/83e192a420d1476fafe0c948edb7fd30 : SUCCESS in 21m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1443d831908848ce8035ad7db83c3243 : SUCCESS in 1h 07m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b5429825486b47b082398ec952608897 : SUCCESS in 1h 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4787dbdcbc448e9b688c87ae04c3ab3 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47c181893b0f45a1a4ec85b66efdfb95 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55e8cf2219f44cd989454352510f7307 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10f9374608994a429eeb65f0d7ee3a9d : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c11fe378ccb435784f6817ab95198da : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81077df4f21041bbbac98f76868d699e : SUCCESS in 5m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a857b08c4fdd4059ba91a84db40956bd : SUCCESS in 20m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cb6203ba91a54ca9a61f1789572f0b54 : SUCCESS in 56m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/88b6bf3a9b09461ea9d0c2dfb6841815 : SUCCESS in 53m 35s","accounts_in_message":[],"_revision_number":33},{"id":"3df18420cbfa397177696da8c397c618cd58a593","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 01:05:21.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/097f1902cece4af5b832e0f8f0e87f0e\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/94c34d2122ef4d5a894238b28e6e56c1 : SUCCESS in 2h 40m 47s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/deefe3d380334b0c9fa2a532a0ed9024 : SUCCESS in 1h 11m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4681fb4fefc40bca9e1f5173e2b6473 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/182c4d0cebca4ab7b345118373924b9d : SUCCESS in 3m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3c3470516c1b4a3ea315661e27d50c2d : FAILURE in 25m 25s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"9347d83667d23430eeac69e7bf54a03ed228b1e5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-25 10:12:43.000000000","message":"Patch Set 33:\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"c4b1287a14d83d552ca93c04b06594619ac390a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-26 05:59:33.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"5aa8f6a3d86b47baddab318dc74a9c850155a837","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-26 06:17:37.000000000","message":"Patch Set 34: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"de7d89a04ac1b62e723d6614d6f7f5d1fd0d0efb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-26 06:19:51.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"a7f66b0d31a464bf478083b73b83de1ec175256e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 07:25:48.000000000","message":"Patch Set 35: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b237a744b69a4e44a2db510c1749e609\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3e257d2dd3d3455884fa1fe7f4ba2c99 : SUCCESS in 20m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/346730be8c5e493c90c58f647f439e55 : SUCCESS in 50m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/be956ba93ff74fdaa472f345c827df4c : SUCCESS in 1h 01m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33260467ed4f4fdbac3b7c13e60e2ac9 : FAILURE in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54dae15376ad44288c97fccea72eb27b : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e267ec104164a0d97ac1a65fac1f34c : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/383dbcf51f0b40bea64b9acb287ca214 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e3f8aabcf39454aa41dd4b4bce380e4 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cb84c575b0e41a58ddbd68e61be460d : SUCCESS in 4m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40aa2eccf05c40c2bdc4184729c9787d : SUCCESS in 17m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3d60146d77fc432783b7622b23e2d982 : SUCCESS in 54m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8964b108067a4767a2de3b740dd5c2ce : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":35},{"id":"38ffdf467732ee73861438208abb28dcf0942617","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 08:09:54.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/566caa9bc6704554b72b6e01306c3f9f\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dbf2f6a0a3814545955c7a417a6eaafa : SUCCESS in 1h 09m 36s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a425e6cde9494ea5b284f57ff537c534 : SUCCESS in 1h 48m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c11ede828e5c484a8ccc681494db65dc : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eeb1419205bb409f88890d68b18437e5 : SUCCESS in 7m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ed8c4c90a4c74b0fa30f77b04c877ee5 : SUCCESS in 1h 06m 18s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"7df9d71bb3837b67bbcaa0455aab0dd9bddf6346","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-26 09:58:44.000000000","message":"Patch Set 36: Published edit on patch set 35.","accounts_in_message":[],"_revision_number":36},{"id":"280e3528d5b65cbdabaef670a3d295c6a34a2203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 11:13:22.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a1b5ea9de964d0e9bcccfd4314c185b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/db69436680a14b38883f9f77d5f3f35c : SUCCESS in 21m 03s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/29ac5ee746a6485da0a42917b1989a0c : SUCCESS in 45m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d662085de43e4c2dbd5cb804cee3f69d : SUCCESS in 1h 09m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee7e44badc114f8bb2965a6bf4eb9ddc : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86ed91cd7f1b4aecb4f7068fe3cf6a07 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8335673de40c49718045b54a5ca97521 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29a9cc0c1e9b4780b43b30d276e5a285 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59070963ecda4ee5af06c531d3d1a75b : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e796562cb21b4e87a956e243a2fc199c : SUCCESS in 4m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9fc787151fef420cae933408158d2ab7 : SUCCESS in 22m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/24cbf222df1246a4bc3e816487f9b7e9 : SUCCESS in 53m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/55b6c6ef26584be4a52bf92de84e6d11 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":36},{"id":"085f7a3db4b64b0d1a255d3f687e6f6134522a22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 11:16:38.000000000","message":"Patch Set 36:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c7cb3a3eaf44059843a7b3aea2aba43\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3efee5ea6e4244969137c8d0987278da : FAILURE in 1h 04m 50s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/40f5a5f8e6a94fbbb412ca3ad04c2e4c : FAILURE in 1h 04m 00s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7955fb5173ca4f02bf5bc657c846cd78 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a91d1e8681c432a891d25e2300d7961 : SUCCESS in 11m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5c3c264e20b84ad88e646d261dc1ac82 : FAILURE in 38m 15s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"2f22d831834cc23cfcfdefe8e6aba0f65f4da293","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-26 11:50:52.000000000","message":"Patch Set 36: Code-Review+1","accounts_in_message":[],"_revision_number":36},{"id":"106b200d36d46a6bdba74c7a2b77b2b291654c10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-27 09:27:22.000000000","message":"Uploaded patch set 37.","accounts_in_message":[],"_revision_number":37},{"id":"6f8705b4e1935b04485bff4b5deced5f557c5063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 10:41:25.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0bec613ebdb4472aa4f6e6ca31a75d5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/07e8639f1fab4f4a903e2fe82ecb3808 : SUCCESS in 21m 31s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7b65bc40246e4dcea08531468e79885b : SUCCESS in 1h 00m 56s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/81b77d0c95ab461e8eefe20c4294e118 : SUCCESS in 1h 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f935b3b87c7940b49f7ba34f491e0512 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/430000d108af4dc58816ee504e30c7a4 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94179abf9de34b658ab92ed2175c0f73 : SUCCESS in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed250784d69948bd920839679f5f5826 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4959f3805642bba1c542606b944f06 : SUCCESS in 6m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7d2458d80ce94ca792485c05e245fe42 : SUCCESS in 20m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/49dbe39d22314b31acbcf49285f11dbd : SUCCESS in 52m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/50d2c9b18d7f448ea0b68c663538feb5 : SUCCESS in 1h 09m 34s","accounts_in_message":[],"_revision_number":37},{"id":"a54521f76087eb74b5e1ab0d7e601b34f66a2c25","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-27 10:50:35.000000000","message":"Patch Set 37: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"c7106d2709d404926947d3729a3b2065849baae2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-07-27 10:50:49.000000000","message":"Patch Set 37:\n\n(3 comments)","accounts_in_message":[],"_revision_number":37},{"id":"298d4ffb9d12202a1c2f99ff752dc5371a816f9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 12:07:45.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c55ec1e5a4c649dca5075f91ea1d5556\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8ab490a253eb4fe7a11c7f35f8cea70b : SUCCESS in 2h 37m 46s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f5ce935b36f24b4da9b515fe8385eed9 : SUCCESS in 2h 12m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/40db2250c9dc4d3293fcae43abf7f62b : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1fbf2a9ff23d4e03b5203073f3256632 : SUCCESS in 3m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/16e88ed906c24a87bbd7e6f3c59ad4c3 : SUCCESS in 1h 56m 13s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"89c5b3174bd3b24130606ddd610ca8dd998f817e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-27 12:40:51.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"e91ea6c1ac70506591d8366c7037336f7169f94b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-27 12:43:51.000000000","message":"Patch Set 37: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":37},{"id":"91de06553ca10a86d01ca60e649aaa5837d14584","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 14:18:50.000000000","message":"Patch Set 37:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e675f013305b45469c12d8b833650e7c\n\n- kolla-build-no-infra-wheels-centos8s https://zuul.opendev.org/t/openstack/build/df50fd9d439b47ea96cc0384b8a454a1 : FAILURE in 20m 56s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/a64c03665bc64ea9bc88cbe7cc22bddc : FAILURE in 14m 35s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/56ae8f97496a4f6c83929908f0d2b2f4 : FAILURE in 14m 51s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/37aac91dfda945c39fdc0f0191413cd8 : FAILURE in 28m 18s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/cb65a873f5694bb2b2cf0161423693f8 : FAILURE in 16m 10s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1ee081a0a6764f1281ffd4b7911a2580 : FAILURE in 15m 57s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/8c2c001686714d5b8903d508e9146f62 : FAILURE in 13m 46s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/23cccdf749144700b3013512fdec64b5 : FAILURE in 15m 46s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/faa6edbeda584536a176f3216f5cd43c : FAILURE in 14m 25s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/41c0c34002c6440a8d52bdf51e2ce019 : FAILURE in 17m 27s (non-voting)\n- kolla-ansible-centos8s-source-octavia https://zuul.opendev.org/t/openstack/build/99fd4de2a2e24263990e606bbda7c366 : FAILURE in 14m 33s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/157f72f828fb4f9595c8c057e9820da4 : FAILURE in 15m 52s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/ea173e3b44e44b3f9e540988abd8d741 : FAILURE in 15m 58s (non-voting)\n- kolla-build-no-infra-wheels-debian https://zuul.opendev.org/t/openstack/build/81fd7c710825484b874a7cbcdfe15608 : SUCCESS in 31m 02s\n- kolla-ansible-debian-source-ironic https://zuul.opendev.org/t/openstack/build/74bb93c0bcd44fc1af26bc0fb8e2c30b : FAILURE in 1h 15m 19s (non-voting)\n- kolla-build-no-infra-wheels-ubuntu https://zuul.opendev.org/t/openstack/build/655d205dd9284171a5bee586256b615c : SUCCESS in 25m 17s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/17cf10ea11f24e038807f07b8deb36a3 : FAILURE in 21m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f1683e8fc66b4d768363b66e1db54921 : SUCCESS in 1h 37m 17s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/216b23e9f0d848078934410f50e72498 : FAILURE in 1h 08m 53s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/6f44077b757d4e649aa9aa6b8a89e0ef : FAILURE in 45m 40s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/c7bf7d66cce343ebae4dd17976825c1a : FAILURE in 21m 26s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/1724a39b9ef0463ba900a56b3841b4c8 : SUCCESS in 27m 24s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/3b17bf116b0949169cb68128890fadd1 : SUCCESS in 1h 03m 35s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/4f58f467aac247649b1e2e0561ecf26b : FAILURE in 35m 41s (non-voting)\n- kolla-ansible-ubuntu-source-octavia https://zuul.opendev.org/t/openstack/build/22ba2f35ca1d4384b780960d9deb877e : SUCCESS in 58m 04s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/03b5f23f0dc749c493569ca90ab625b5 : SUCCESS in 1h 10m 14s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/3324caa4375e44bc8124df1e0078ec12 : SUCCESS in 32m 36s (non-voting)\n- kolla-ansible-ubuntu-source-kvm https://zuul.opendev.org/t/openstack/build/ec02f540a184487297c71fbcdb2a807a : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"64801cdad45d1ea0ce7708aed2ccc79841d2a5d9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-27 14:26:37.000000000","message":"Patch Set 37: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"0284f3dfe2f3e9f91f503b29fa03f6733953ab33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 14:27:09.000000000","message":"Patch Set 37: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":37},{"id":"d4ea04b24ae9ebc25b3d584cd9c2f000c446a78c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 15:43:40.000000000","message":"Patch Set 37: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa65aa39181f4396be3adf4f4c9b8588\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/aa32b30378214ab281ced155c0ab2378 : SUCCESS in 20m 13s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/88009df67a6d40008130965ea2e3262c : SUCCESS in 1h 00m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3bdbb689c6434564b861cf3383db4cf8 : SUCCESS in 1h 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ca3622b12049dfb054050486b9ed92 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6770cdde2efa464794695a7ce7bc1a51 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b8b57863a5b41b79d90e827722b69ce : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/928bd65014874fb7a07b344d41956846 : SUCCESS in 6m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0d2cdd1e4b434408b6b95a9ae7c8cb79 : SUCCESS in 17m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf0abc77dac64a158d7cd01e72966194 : SUCCESS in 55m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2f0367c18fda4c49ab0bb4479a225561 : SUCCESS in 1h 10m 36s","accounts_in_message":[],"_revision_number":37},{"id":"25d7e9a4d4a13d3d5c9df55c5d9e19e49b5c2cfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 15:43:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":37},{"id":"9fe3437eed110260754a09f8a48e5f8f6fc3de8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 15:45:46.000000000","message":"Patch Set 37:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c23fa10178d74b65b6c499e9f9b602f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a9acc2f7d04b5aae8e53c09984d84f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":37}],"current_revision_number":37,"current_revision":"9f25aac18a31920c2f2f6630b35f3f734f03dcf4","revisions":{"62dff5c1e8f2a54d962dd8445d03e0ac3d5aa230":{"kind":"REWORK","_number":1,"created":"2022-04-27 15:48:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/1"}}},"commit":{"parents":[{"commit":"b02f643db03694074293858470a1db36246b4a0e","subject":"Merge \"adjust permissions on _extend_start files (part II)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b02f643db03694074293858470a1db36246b4a0e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:48:12.000000000","tz":0},"subject":"veryWIP: Move to Ubuntu Jammy (22.04)","message":"veryWIP: Move to Ubuntu Jammy (22.04)\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62dff5c1e8f2a54d962dd8445d03e0ac3d5aa230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62dff5c1e8f2a54d962dd8445d03e0ac3d5aa230"}]},"branch":"refs/heads/master"},"7bfbd57112d47195310496400462ebd3a9a31d7b":{"kind":"REWORK","_number":2,"created":"2022-04-27 15:50:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/2"}}},"commit":{"parents":[{"commit":"b02f643db03694074293858470a1db36246b4a0e","subject":"Merge \"adjust permissions on _extend_start files (part II)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b02f643db03694074293858470a1db36246b4a0e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:49:49.000000000","tz":0},"subject":"veryWIP: Move to Ubuntu Jammy (22.04)","message":"veryWIP: Move to Ubuntu Jammy (22.04)\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bfbd57112d47195310496400462ebd3a9a31d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bfbd57112d47195310496400462ebd3a9a31d7b"}]},"branch":"refs/heads/master"},"7bf62ee1bfddcdc559cfb8761f76d4065ff992b5":{"kind":"REWORK","_number":3,"created":"2022-04-27 15:58:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/3"}}},"commit":{"parents":[{"commit":"b02f643db03694074293858470a1db36246b4a0e","subject":"Merge \"adjust permissions on _extend_start files (part II)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b02f643db03694074293858470a1db36246b4a0e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:58:13.000000000","tz":0},"subject":"veryWIP: Move to Ubuntu Jammy (22.04)","message":"veryWIP: Move to Ubuntu Jammy (22.04)\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bf62ee1bfddcdc559cfb8761f76d4065ff992b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bf62ee1bfddcdc559cfb8761f76d4065ff992b5"}]},"branch":"refs/heads/master"},"c8948ba9797a5d7e874b2cfab7d9d420ba8c650e":{"kind":"REWORK","_number":4,"created":"2022-04-27 17:42:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/4"}}},"commit":{"parents":[{"commit":"b02f643db03694074293858470a1db36246b4a0e","subject":"Merge \"adjust permissions on _extend_start files (part II)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b02f643db03694074293858470a1db36246b4a0e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 17:42:26.000000000","tz":0},"subject":"veryWIP: Move to Ubuntu Jammy (22.04)","message":"veryWIP: Move to Ubuntu Jammy (22.04)\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8948ba9797a5d7e874b2cfab7d9d420ba8c650e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8948ba9797a5d7e874b2cfab7d9d420ba8c650e"}]},"branch":"refs/heads/master"},"6368c26fa090eae0b032c773d92df00132f415f8":{"kind":"REWORK","_number":5,"created":"2022-04-28 07:10:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/5"}}},"commit":{"parents":[{"commit":"b02f643db03694074293858470a1db36246b4a0e","subject":"Merge \"adjust permissions on _extend_start files (part II)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b02f643db03694074293858470a1db36246b4a0e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-28 07:10:10.000000000","tz":0},"subject":"veryWIP: Move to Ubuntu Jammy (22.04)","message":"veryWIP: Move to Ubuntu Jammy (22.04)\n\nTODO:\nfluentd - missing\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon\ntelegraf/collectd - collectd packages missing in Jammy\n\nCHANGES:\nprometheus-libvirt-exporter: go get in 1.18 doesn\u0027t install binary\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6368c26fa090eae0b032c773d92df00132f415f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6368c26fa090eae0b032c773d92df00132f415f8"}]},"branch":"refs/heads/master"},"dca2b12ca264c61026d64902e1ed12ffd5016d4a":{"kind":"REWORK","_number":6,"created":"2022-04-28 07:13:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/6"}}},"commit":{"parents":[{"commit":"b02f643db03694074293858470a1db36246b4a0e","subject":"Merge \"adjust permissions on _extend_start files (part II)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b02f643db03694074293858470a1db36246b4a0e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-28 07:12:41.000000000","tz":0},"subject":"veryWIP: Move to Ubuntu Jammy (22.04)","message":"veryWIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nTODO:\nfluentd - missing\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon\ntelegraf/collectd - collectd packages missing in Jammy\n\nCHANGES:\nprometheus-libvirt-exporter: go get in 1.18 doesn\u0027t install binary\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca2b12ca264c61026d64902e1ed12ffd5016d4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca2b12ca264c61026d64902e1ed12ffd5016d4a"}]},"branch":"refs/heads/master"},"e9e8120f32bb3b1ae30574f403426f8d5188b1a0":{"kind":"REWORK","_number":7,"created":"2022-05-16 08:34:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/7"}}},"commit":{"parents":[{"commit":"9edcdb0a7ab57764b6e8622e5a75d4a1ecce2515","subject":"Merge \"[publishing] use a new naming scheme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9edcdb0a7ab57764b6e8622e5a75d4a1ecce2515"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-05-16 08:34:19.000000000","tz":0},"subject":"veryWIP: Move to Ubuntu Jammy (22.04)","message":"veryWIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nTODO:\nfluentd - missing\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon\ntelegraf/collectd - collectd packages missing in Jammy\n\nCHANGES:\nprometheus-libvirt-exporter: go get in 1.18 doesn\u0027t install binary\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9e8120f32bb3b1ae30574f403426f8d5188b1a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9e8120f32bb3b1ae30574f403426f8d5188b1a0"}]},"branch":"refs/heads/master"},"a5a550b1cc37fa385cf8a5e5227110e03b4593c2":{"kind":"REWORK","_number":8,"created":"2022-05-25 06:38:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/8"}}},"commit":{"parents":[{"commit":"9edcdb0a7ab57764b6e8622e5a75d4a1ecce2515","subject":"Merge \"[publishing] use a new naming scheme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9edcdb0a7ab57764b6e8622e5a75d4a1ecce2515"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-05-25 06:38:38.000000000","tz":0},"subject":"veryWIP: Move to Ubuntu Jammy (22.04)","message":"veryWIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nTODO:\nfluentd - missing\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon\ntelegraf/collectd - collectd packages missing in Jammy\n\nCHANGES:\nprometheus-libvirt-exporter: go get in 1.18 doesn\u0027t install binary\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5a550b1cc37fa385cf8a5e5227110e03b4593c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5a550b1cc37fa385cf8a5e5227110e03b4593c2"}]},"branch":"refs/heads/master"},"7a1fe1473b9c70e3a884187d9332c2dc038f4fb6":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2022-05-25 06:39:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/9"}}},"commit":{"parents":[{"commit":"9edcdb0a7ab57764b6e8622e5a75d4a1ecce2515","subject":"Merge \"[publishing] use a new naming scheme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9edcdb0a7ab57764b6e8622e5a75d4a1ecce2515"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-05-25 06:39:00.000000000","tz":0},"subject":"veryWIP: Move to Ubuntu Jammy (22.04)","message":"veryWIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nTODO:\nfluentd - missing\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy\n\nCHANGES:\nprometheus-libvirt-exporter: go get in 1.18 doesn\u0027t install binary\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a1fe1473b9c70e3a884187d9332c2dc038f4fb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a1fe1473b9c70e3a884187d9332c2dc038f4fb6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7996f82649b0cf7ddef48173922918a70a497e95":{"kind":"REWORK","_number":10,"created":"2022-05-25 09:36:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/10"}}},"commit":{"parents":[{"commit":"9edcdb0a7ab57764b6e8622e5a75d4a1ecce2515","subject":"Merge \"[publishing] use a new naming scheme\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9edcdb0a7ab57764b6e8622e5a75d4a1ecce2515"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-05-25 09:36:24.000000000","tz":0},"subject":"veryWIP: Move to Ubuntu Jammy (22.04)","message":"veryWIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nTODO:\nfluentd - missing\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy\n\nCHANGES:\nprometheus-libvirt-exporter: go get in 1.18 doesn\u0027t install binary\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7996f82649b0cf7ddef48173922918a70a497e95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7996f82649b0cf7ddef48173922918a70a497e95"}]},"branch":"refs/heads/master"},"34462270fce8fcd7ca7464ede8bb2cced4b1fdce":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-05-25 14:52:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/11"}}},"commit":{"parents":[{"commit":"a51ebc9e7fc2163623cb3d9b7540c23ea984c859","subject":"Merge \"[CI] Move queue setting to project level\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a51ebc9e7fc2163623cb3d9b7540c23ea984c859"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-05-25 14:52:37.000000000","tz":0},"subject":"veryWIP: Move to Ubuntu Jammy (22.04)","message":"veryWIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nTODO:\nfluentd - missing\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy\n\nCHANGES:\nprometheus-libvirt-exporter: go get in 1.18 doesn\u0027t install binary\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34462270fce8fcd7ca7464ede8bb2cced4b1fdce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34462270fce8fcd7ca7464ede8bb2cced4b1fdce"}]},"branch":"refs/heads/master","description":"Rebase"},"2e0f48291094de9b3b8017fd6abdea4491de3a04":{"kind":"REWORK","_number":12,"created":"2022-05-30 14:38:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/12"}}},"commit":{"parents":[{"commit":"d424a63d6049d8d241ab1294aea035d4b187a111","subject":"Replace Certbot with Lego for Let\u0027s Encrypt container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d424a63d6049d8d241ab1294aea035d4b187a111"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-05-30 14:38:16.000000000","tz":120},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e0f48291094de9b3b8017fd6abdea4491de3a04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e0f48291094de9b3b8017fd6abdea4491de3a04"}]},"branch":"refs/heads/master"},"56fca2f5691e06dea0d4dae78c65edd2bd95395b":{"kind":"REWORK","_number":13,"created":"2022-05-31 06:51:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/13"}}},"commit":{"parents":[{"commit":"d424a63d6049d8d241ab1294aea035d4b187a111","subject":"Replace Certbot with Lego for Let\u0027s Encrypt container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d424a63d6049d8d241ab1294aea035d4b187a111"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-05-31 06:51:45.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56fca2f5691e06dea0d4dae78c65edd2bd95395b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56fca2f5691e06dea0d4dae78c65edd2bd95395b"}]},"branch":"refs/heads/master"},"7311c4f096f12dd6db2b43652c95cbbbda800f85":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2022-05-31 07:14:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/14"}}},"commit":{"parents":[{"commit":"d424a63d6049d8d241ab1294aea035d4b187a111","subject":"Replace Certbot with Lego for Let\u0027s Encrypt container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d424a63d6049d8d241ab1294aea035d4b187a111"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-05-31 07:14:52.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/844000\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7311c4f096f12dd6db2b43652c95cbbbda800f85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7311c4f096f12dd6db2b43652c95cbbbda800f85"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ef301405011519b6ce43b0885e9a5833d24a3818":{"kind":"REWORK","_number":15,"created":"2022-05-31 08:58:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/15"}}},"commit":{"parents":[{"commit":"d424a63d6049d8d241ab1294aea035d4b187a111","subject":"Replace Certbot with Lego for Let\u0027s Encrypt container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d424a63d6049d8d241ab1294aea035d4b187a111"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-05-31 08:58:16.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/844000\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef301405011519b6ce43b0885e9a5833d24a3818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef301405011519b6ce43b0885e9a5833d24a3818"}]},"branch":"refs/heads/master"},"26341de930c80fba2744cee991fa81109c158a29":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2022-06-07 12:40:19.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/839585/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/16"}}},"commit":{"parents":[{"commit":"42da715ff37f34f2d1728d8b7fca47af187ebf62","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42da715ff37f34f2d1728d8b7fca47af187ebf62"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-07 12:40:19.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/844000\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26341de930c80fba2744cee991fa81109c158a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26341de930c80fba2744cee991fa81109c158a29"}]},"branch":"refs/heads/master","description":"Rebase"},"813aa962dfa242068bb270498f4a5eff6c6a095e":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-06-09 20:48:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/17"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-06-09 20:48:41.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/844000\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/813aa962dfa242068bb270498f4a5eff6c6a095e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/813aa962dfa242068bb270498f4a5eff6c6a095e"}]},"branch":"refs/heads/master","description":"Rebase"},"412fc97f942564563074c834aecb794b310211d5":{"kind":"REWORK","_number":18,"created":"2022-06-15 06:47:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/18"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-06-15 06:47:58.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/844000\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/412fc97f942564563074c834aecb794b310211d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/412fc97f942564563074c834aecb794b310211d5"}]},"branch":"refs/heads/master"},"80e5b23789e4074aa457f68de0df1adbd90fab34":{"kind":"REWORK","_number":19,"created":"2022-06-15 06:50:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/19"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-06-15 06:50:45.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/844000\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80e5b23789e4074aa457f68de0df1adbd90fab34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80e5b23789e4074aa457f68de0df1adbd90fab34"}]},"branch":"refs/heads/master"},"dbfc1b022b0dea5b1ced26f3d34f1e834a744bf6":{"kind":"REWORK","_number":20,"created":"2022-06-15 07:45:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/20"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-06-15 07:45:17.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/844000\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbfc1b022b0dea5b1ced26f3d34f1e834a744bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbfc1b022b0dea5b1ced26f3d34f1e834a744bf6"}]},"branch":"refs/heads/master"},"7b6f11d274c1f4fcad20c61ea4cbf081e3e2417a":{"kind":"REWORK","_number":21,"created":"2022-06-15 10:26:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/21"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-06-15 10:26:10.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/844000\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f11d274c1f4fcad20c61ea4cbf081e3e2417a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b6f11d274c1f4fcad20c61ea4cbf081e3e2417a"}]},"branch":"refs/heads/master"},"745abb744dd6203d11378de5806d4e312aa96e1e":{"kind":"REWORK","_number":22,"created":"2022-06-15 12:08:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/22"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-06-15 12:08:26.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/844000\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/745abb744dd6203d11378de5806d4e312aa96e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/745abb744dd6203d11378de5806d4e312aa96e1e"}]},"branch":"refs/heads/master"},"3863ba3e8dafcb114b244be28fc5ea59b2b252fa":{"kind":"NO_CODE_CHANGE","_number":23,"created":"2022-06-15 12:11:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/23"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-06-15 12:11:05.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3863ba3e8dafcb114b244be28fc5ea59b2b252fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3863ba3e8dafcb114b244be28fc5ea59b2b252fa"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b6f65eb790edc838660f240e35469963f79cc03b":{"kind":"REWORK","_number":24,"created":"2022-06-15 17:20:28.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/839585/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/24"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-15 17:20:01.000000000","tz":120},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6f65eb790edc838660f240e35469963f79cc03b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6f65eb790edc838660f240e35469963f79cc03b"}]},"branch":"refs/heads/master"},"9f3d5dc16e209f25c2e208747eef0b783a18441c":{"kind":"NO_CODE_CHANGE","_number":25,"created":"2022-06-15 23:17:11.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/839585/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/25"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-15 23:17:11.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f3d5dc16e209f25c2e208747eef0b783a18441c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f3d5dc16e209f25c2e208747eef0b783a18441c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4e993fda769dac93dde3e5e59d5b63ae4794f244":{"kind":"NO_CODE_CHANGE","_number":26,"created":"2022-06-16 08:20:28.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/839585/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/26"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-16 08:20:28.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e993fda769dac93dde3e5e59d5b63ae4794f244"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e993fda769dac93dde3e5e59d5b63ae4794f244"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bfd9c86f175f73f35c2431a9f179664fb840bf50":{"kind":"REWORK","_number":27,"created":"2022-06-16 11:40:55.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/839585/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/27"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-16 11:40:42.000000000","tz":120},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfd9c86f175f73f35c2431a9f179664fb840bf50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfd9c86f175f73f35c2431a9f179664fb840bf50"}]},"branch":"refs/heads/master"},"0b68ced7ff2e0b329eab4c211528faa292b7bcc8":{"kind":"REWORK","_number":28,"created":"2022-06-20 14:05:43.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/839585/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/28"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-20 14:05:43.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b68ced7ff2e0b329eab4c211528faa292b7bcc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b68ced7ff2e0b329eab4c211528faa292b7bcc8"}]},"branch":"refs/heads/master"},"e500e3ad7fe6cbd6c979e0a67150a62c30bd3af4":{"kind":"REWORK","_number":29,"created":"2022-06-21 13:30:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/29"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-06-21 13:30:41.000000000","tz":0},"subject":"WIP: Move to Ubuntu Jammy (22.04)","message":"WIP: Move to Ubuntu Jammy (22.04)\n\nSome repos still use focal (influx, rabbitmq, proxysql, etc)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e500e3ad7fe6cbd6c979e0a67150a62c30bd3af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e500e3ad7fe6cbd6c979e0a67150a62c30bd3af4"}]},"branch":"refs/heads/master"},"47b631a5dbe63fb0099e827422078e9f8ebf4a2b":{"kind":"NO_CODE_CHANGE","_number":30,"created":"2022-06-21 14:54:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/30"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-06-21 14:54:47.000000000","tz":0},"subject":"Move to Ubuntu Jammy (22.04)","message":"Move to Ubuntu Jammy (22.04)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47b631a5dbe63fb0099e827422078e9f8ebf4a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/47b631a5dbe63fb0099e827422078e9f8ebf4a2b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"289a937f403ace77957b65161ae7b2e29e091cf9":{"kind":"REWORK","_number":31,"created":"2022-07-22 09:13:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/31"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-22 09:11:18.000000000","tz":0},"subject":"Move to Ubuntu Jammy (22.04)","message":"Move to Ubuntu Jammy (22.04)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\nmonasca - requires new librdkafka, no wheel for Python 3.10 for now\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/289a937f403ace77957b65161ae7b2e29e091cf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/289a937f403ace77957b65161ae7b2e29e091cf9"}]},"branch":"refs/heads/master"},"d14736f9884aa16cec35b5c67830b8a3526df8da":{"kind":"REWORK","_number":32,"created":"2022-07-22 10:34:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/32"}}},"commit":{"parents":[{"commit":"07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c","subject":"Switch OPENSTACK_RELEASE back to master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07a65c9061fe34bbbd39b5247df54ae2ef8a4c5c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-22 10:34:41.000000000","tz":0},"subject":"Move to Ubuntu Jammy (22.04)","message":"Move to Ubuntu Jammy (22.04)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\nmonasca - requires new librdkafka, no wheel for Python 3.10 for now\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d14736f9884aa16cec35b5c67830b8a3526df8da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d14736f9884aa16cec35b5c67830b8a3526df8da"}]},"branch":"refs/heads/master"},"b2294af108aa13301455587048097ee2ee137f0d":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2022-07-24 22:22:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/85/839585/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/33"}}},"commit":{"parents":[{"commit":"25ae26e77cb6a67447684c3de19cfb2dbaf0d7a8","subject":"Update ProxySQL to version 2.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25ae26e77cb6a67447684c3de19cfb2dbaf0d7a8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-07-24 22:22:11.000000000","tz":0},"subject":"Move to Ubuntu Jammy (22.04)","message":"Move to Ubuntu Jammy (22.04)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\nmonasca - requires new librdkafka, no wheel for Python 3.10 for now\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2294af108aa13301455587048097ee2ee137f0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2294af108aa13301455587048097ee2ee137f0d"}]},"branch":"refs/heads/master","description":"Rebase"},"f3bb091c7a48889bd57086b5e10375d7cf19fdc8":{"kind":"REWORK","_number":34,"created":"2022-07-26 05:59:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/34"}}},"commit":{"parents":[{"commit":"25ae26e77cb6a67447684c3de19cfb2dbaf0d7a8","subject":"Update ProxySQL to version 2.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25ae26e77cb6a67447684c3de19cfb2dbaf0d7a8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-26 05:59:32.000000000","tz":0},"subject":"Move to Ubuntu Jammy (22.04)","message":"Move to Ubuntu Jammy (22.04)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\nmonasca - requires new librdkafka, no wheel for Python 3.10 for now\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3bb091c7a48889bd57086b5e10375d7cf19fdc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3bb091c7a48889bd57086b5e10375d7cf19fdc8"}]},"branch":"refs/heads/master"},"81b53b01e8592b9b83c68321c7fad8781d3fa110":{"kind":"REWORK","_number":35,"created":"2022-07-26 06:19:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/35"}}},"commit":{"parents":[{"commit":"25ae26e77cb6a67447684c3de19cfb2dbaf0d7a8","subject":"Update ProxySQL to version 2.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25ae26e77cb6a67447684c3de19cfb2dbaf0d7a8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-26 06:19:51.000000000","tz":0},"subject":"Move to Ubuntu Jammy (22.04)","message":"Move to Ubuntu Jammy (22.04)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\nmonasca - requires new librdkafka, no wheel for Python 3.10 for now\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81b53b01e8592b9b83c68321c7fad8781d3fa110"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/81b53b01e8592b9b83c68321c7fad8781d3fa110"}]},"branch":"refs/heads/master"},"aff1951865c2be42bcc0d147f012e2845d3f2126":{"kind":"REWORK","_number":36,"created":"2022-07-26 09:58:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/85/839585/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/36"}}},"commit":{"parents":[{"commit":"25ae26e77cb6a67447684c3de19cfb2dbaf0d7a8","subject":"Update ProxySQL to version 2.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25ae26e77cb6a67447684c3de19cfb2dbaf0d7a8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-07-26 09:58:44.000000000","tz":0},"subject":"Move to Ubuntu Jammy (22.04)","message":"Move to Ubuntu Jammy (22.04)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\nmonasca - requires new librdkafka, no wheel for Python 3.10 for now\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aff1951865c2be42bcc0d147f012e2845d3f2126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aff1951865c2be42bcc0d147f012e2845d3f2126"}]},"branch":"refs/heads/master"},"9f25aac18a31920c2f2f6630b35f3f734f03dcf4":{"kind":"REWORK","_number":37,"created":"2022-07-27 09:27:22.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/85/839585/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/839585/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/839585/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/839585/37"}}},"commit":{"parents":[{"commit":"dc55bdb5eee7c41660c7ef7a5bd5064d95f003f4","subject":"Merge \"Install etcd binary from github\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc55bdb5eee7c41660c7ef7a5bd5064d95f003f4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-04-27 15:47:41.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-07-27 09:27:01.000000000","tz":120},"subject":"Move to Ubuntu Jammy (22.04)","message":"Move to Ubuntu Jammy (22.04)\n\nmariadb - downloads.mariadb.com doesn\u0027t have jammy version, should\nbe there soon - using Ubuntu provided version for now\n\ntelegraf/collectd - collectd packages missing in Jammy - marked\nas unbuildable\nmonasca - requires new librdkafka, no wheel for Python 3.10 for now\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/846070\nChange-Id: I11b4fbcaba165824252902d0270ce1dbbc3d6e9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f25aac18a31920c2f2f6630b35f3f734f03dcf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f25aac18a31920c2f2f6630b35f3f734f03dcf4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
