)]}'
{"id":"openstack%2Fkolla~840737","triplet_id":"openstack%2Fkolla~stable%2Fyoga~I09ea136c7354143f45f856d5e8d8ed1192879622","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-07 09:09:52.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"I09ea136c7354143f45f856d5e8d8ed1192879622","subject":"prometheus-libvirt-exporter: fix build with newer Go","status":"MERGED","created":"2022-05-06 13:25:19.000000000","updated":"2022-05-07 10:47:40.000000000","submitted":"2022-05-07 10:46:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840737","cherry_pick_of_change":840829,"cherry_pick_of_patch_set":1,"meta_rev_id":"dab1a39d2eb56a0034308b04f1c466ae1b18b819","_number":840737,"virtual_id_number":840737,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-07 10:46:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2022-05-06 16:08:38.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-05-07 09:09:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2022-05-07 09:09:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-06 13:25:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2022-05-06 14:45:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-06 16:08:38.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"c316ad92b27bdd7525263d5fee67c34a2ccf114d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-06 13:25:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7ba3d28120b16689c09559eb411decf177edd0b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 14:45:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f0b8b9850f814ea2be6ae3eea70a83ed : SUCCESS in 25m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e238e815ff204a42ae3eecbf17036f28 : SUCCESS in 32m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/52ce9f112a7f4a46bbba447a8b259bc1 : SUCCESS in 1h 16m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1fb96204278b440e83b62571f073250c : SUCCESS in 1h 04m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9e7411d4bb1b4c239b63ab36d68c5eaa : SUCCESS in 1h 17m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9cbba3c1369c4b0288f412a3a08b6cb7 : SUCCESS in 20m 50s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c31d7695af514cc5857078d4d1ac6cd3 : SUCCESS in 15m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d6bf647b00ac4013bc720e1e8434b9ad : SUCCESS in 56m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4ceb6ac3737544f887760b1dbc3ae138 : SUCCESS in 1h 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a51355682d874276942c22e44f98c981 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab2bc293bf744354b5064661c6e5224b : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/609da5deba5a4755b4e4ae365fb0a601 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d1021d52aa346c1a276202515964512 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2589e3ac014f47079c1b4eeac99185ef : SUCCESS in 18m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d2af719369c4f1cb6de200f084fc235 : SUCCESS in 27m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/85262947fc5f456caf81d4cd5c28b92b : SUCCESS in 52m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4de057a2f923449d92fc537d821944ba : SUCCESS in 48m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/09292ff69f0842679282683a4fda99c6 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":1},{"id":"96ccb6f1304500892eae836830a8b04adece7882","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 15:48:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d601e94a0e3e4d95a4c0098c8de5bf21 : SUCCESS in 1h 27m 04s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d1c3a54a59e498388fcfbb805c3d825 : SUCCESS in 1h 17m 41s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c352f99d286a43d795ee97d378b1a769 : SUCCESS in 2h 02m 03s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/33dddfb01d46463dab23dc48c34dd3bd : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d699e8841f1a4d23876999e99f22553d : SUCCESS in 5m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9995d0ab26774b8cac6e6676e01205af : SUCCESS in 1h 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1d44739e1418139e8b6e6706417ee33ef578b74","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-06 16:08:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"26f7c4ab5ada9f278edb6d240d84c4f3e6181c0a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-07 09:09:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89bbdfa87e667d3e1c53ab9e0d8e5eb644795c0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 09:10:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"668709c71f08c800650c67936c9580263508c422","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:46:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/287a5e7768554580a978cb2e3a7a4cef : SUCCESS in 30m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/37b8849e02f24dbca2fb1bda4844925f : SUCCESS in 1h 15m 54s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/cf7873f1bda0468fb135f8ab07e52698 : SUCCESS in 1h 32m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/485b6a1ece4141bdbc701ac77eb25dac : SUCCESS in 24m 34s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/197a9291d9694c069dcefec32f685803 : SUCCESS in 1h 05m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/86d04f5fd16a471a88c4714b43bf4cc6 : SUCCESS in 1h 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110e3e27a2654d4aa5cb3f64fca7dbe2 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4223a3cb551744e68206e6b7977cffbc : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a750fcf943047dcba76f0a8c9c6bb4a : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e39bc5ac8f342328ee136dba629fc60 : SUCCESS in 6m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/51e2b8e8aff34514aa679ef0df7516c6 : SUCCESS in 24m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/40085a635f11463290b9b04e5c71c4e1 : SUCCESS in 49m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2d50c9a8154b49a6bc560efcfaa62c7b : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":1},{"id":"eaa945dc83b3ec21dfd663671dbc0f8c4b900c62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:46:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dab1a39d2eb56a0034308b04f1c466ae1b18b819","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-07 10:47:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/260eeaac451d49d38f3d0e62349191da : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ca52d999b4eea98e813d78bd26d5e3aba97fc97","revisions":{"0ca52d999b4eea98e813d78bd26d5e3aba97fc97":{"kind":"REWORK","_number":1,"created":"2022-05-06 13:25:19.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/37/840737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/840737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/840737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/840737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/840737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/840737/1"}}},"commit":{"parents":[{"commit":"4343f56e423b9427aeda15609cf28b5717430fbe","subject":"Add multipath to cinder-volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4343f56e423b9427aeda15609cf28b5717430fbe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-05-06 08:25:22.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-06 13:25:19.000000000","tz":0},"subject":"prometheus-libvirt-exporter: fix build with newer Go","message":"prometheus-libvirt-exporter: fix build with newer Go\n\nI have no idea how Go works. This change make build happen.\n\nChange-Id: I09ea136c7354143f45f856d5e8d8ed1192879622\n(cherry picked from commit 9630032fd61c6f0623bd1ae7577756e7166427f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ca52d999b4eea98e813d78bd26d5e3aba97fc97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ca52d999b4eea98e813d78bd26d5e3aba97fc97"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
