)]}'
{"id":"openstack%2Fkolla~840814","triplet_id":"openstack%2Fkolla~stable%2Fyoga~Iaf5ce2b7276b275b21d892d0044a126ab901507f","project":"openstack/kolla","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"28356":{"account":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"last_update":"2022-05-06 10:47:33.000000000","reason":"Rafael Weingartner replied on the change"},"24072":{"account":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"last_update":"2022-05-09 10:14:01.000000000","reason":"Change was submitted"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-05-09 10:14:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaf5ce2b7276b275b21d892d0044a126ab901507f","subject":"cloudkitty: disable building for ubuntu/binary","status":"MERGED","created":"2022-05-06 05:47:51.000000000","updated":"2022-05-09 10:15:16.000000000","submitted":"2022-05-09 10:14:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"840814","meta_rev_id":"5847e5f98973b112015a526b2bad2e8e41ed3547","_number":840814,"virtual_id_number":840814,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-09 10:14:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"all":[{"value":1,"date":"2022-05-09 10:15:16.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":2,"date":"2022-05-09 08:45:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2022-05-09 10:15:16.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-06 14:07:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":1,"date":"2022-05-09 08:45:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-05-09 10:15:16.000000000","post_submit":true,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-06 06:00:03.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"state":"REVIEWER"},{"updated":"2022-05-06 07:24:29.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-05-06 07:28:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-06 14:07:19.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-05-09 08:45:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d57e0495567a17e2352db08c3557dc27a304289c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-06 05:47:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb00d71ab73e4096098524f2327a47a1fd753f7e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-06 05:50:08.000000000","message":"Patch Set 1: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e829068657be8a50d9891782338c5a98155ccec","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-06 06:00:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"239c5a0438a17b04c31df741fa8ae2817b1a879e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-06 06:01:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d60d0b7128be543f4f717767d3e78d26dc80f10c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2022-05-06 06:02:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bf90b1d0bb26d5b67f2540def1b11a84b7c00375","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-06 07:24:29.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf5095773b49a9e2f74168a864577473ddf1c083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 07:28:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7fa377939e54480fae89a91d7369ded4 : FAILURE in 23m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/874cae8d1a1b448bb05c52d9eb5a7651 : FAILURE in 31m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/047b94f98a164012903aedf34427ffbc : SUCCESS in 1h 18m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c795a7b06472447dbcc1a12f2b7e1cc3 : SUCCESS in 1h 04m 30s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ca47bb5b81c44fdc831ccf344a391587 : SUCCESS in 1h 17m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e895bd17c57641c7904593633a491724 : SUCCESS in 22m 23s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c203da1d9d2b497782b09afafa847092 : SUCCESS in 16m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/67829f2044294b1bb8906f29fff39a10 : SUCCESS in 1h 00m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2aba39b477c64804848e0ffc9003da4b : SUCCESS in 1h 20m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3f04c031114a9eba601c76e866a51e : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03b2a80d2bee4c9797210a92831ac1a2 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f2efcfce3594966b07f1ee142942c7e : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e05179bc6847d695658fe854f1c41e : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ed46aaeddda4981be870c6012fbce3d : SUCCESS in 16m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/59fd920524d645abb4d2210a060ef8d0 : SUCCESS in 22m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2249fe6d1a7a44f7960fea009e3b1de6 : SUCCESS in 50m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ef9899d9628a4fc6b61fa4021967d894 : SUCCESS in 46m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d623a9faaaf645aa8de20efc7d82aa10 : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":2},{"id":"0d52d4653711854e9cc7cd74e4ad50bdaa233b32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 07:52:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3304ded82cec4a79ba55be5b53e68d76 : FAILURE in 1h 38m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f92bf3559f64b029b8f0f95bb7872b9 : SUCCESS in 1h 00m 00s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ee33135d36a74c4a89d7e9d5337edeea : SUCCESS in 1h 47m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6177a84af35f416f824151a75f5cd5bd : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a73d03bf2f9b405d9bc08bea3f20d8ee : SUCCESS in 5m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/50cd636957814b17835c648ac749728c : SUCCESS in 1h 01m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"078f219e2def4cb63e0edafd22173379564f5af0","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-05-06 10:47:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ab88e6981d998d9566a604f7bf5df07fc9920a13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-06 13:40:11.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"142f3342d30443e11398763360ee7c1f743d781a","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-05-06 14:07:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5c7ead3361c3bcbb56c86b0e520e7c09a5a04b20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 15:04:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a589685c507742db97bbabf50212ae00 : SUCCESS in 29m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/21f016629ce840b4857835575dbd02e8 : SUCCESS in 34m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e4e188862076472f84f03b5c9226dcc7 : SUCCESS in 1h 15m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3811b813334146a4a7e6a1b012deec40 : SUCCESS in 1h 07m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a587be38b4594485b6ae680e97ae0f02 : SUCCESS in 1h 20m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ca2fbc5022944fac873e9076fd88d6a4 : SUCCESS in 21m 37s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b946a364069244278d69c3238579a52e : SUCCESS in 16m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/07b21c33d7754f66b3cbfcfaf1ebfd6c : SUCCESS in 1h 06m 30s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bb3a20d47c3149d186a7b3f033774e87 : SUCCESS in 1h 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56423913256d4fb28cac741e3e1e82a1 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da9f8188decc44f68e09f9f8f91b7f10 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96ce2ba5b06f4665ad3e45a627ae7b40 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e21323a34b494c31953ebcd6859aaf50 : SUCCESS in 6m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/23e877b9d2354f5a8d6607a0f2aa1302 : SUCCESS in 19m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b2ee8965e7348bb858d2778439445b9 : SUCCESS in 23m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/67e72d5a3bf845b3bdacef46752d2191 : SUCCESS in 54m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7cf4f2281483438d88aa96fcc80b76e5 : SUCCESS in 57m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d41f97a6812b463aaff7509dc63c84bb : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b097cae61f76fa15cbdf89201c11a1b18da678bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-06 19:57:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a016077b157648298bf3474a45537174 : SUCCESS in 2h 05m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/939e2f7f0ef0424ab26ca9d700e3e0ce : SUCCESS in 1h 35m 19s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/47bb1ee2f0524bb69c7ea71735649228 : SUCCESS in 2h 13m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc3899e2fba043ec808bad8c09f03233 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd0060b8b76147dc92c0a9d294efeb2d : SUCCESS in 5m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d675269d16c449786a469e77ee69b57 : SUCCESS in 1h 37m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"700574fb725ca7793c9e1bd2511d1edf28e032a5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-05-09 08:45:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f25aca5e5c7214391faf2fc356bbb40c4fb7cbe4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 08:45:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6b792badb63562099c08efb92fb3b56f9dc627ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 10:14:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/007381267df149d6b6cad47aba9e3bce : SUCCESS in 32m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a9089c434fd3437b82dd7adbe4b4f46d : SUCCESS in 54m 55s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3b6bd1efa4bc4d7c938e69edbbfbd231 : SUCCESS in 1h 27m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2672be49f7bd476ebdb0d016a444c084 : SUCCESS in 16m 18s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2ce67172b0214f179d7b6301623c2b61 : SUCCESS in 59m 09s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c2b9d13442ca43fa8a6e38e56b5b7299 : SUCCESS in 57m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c8027753cfe4a6d905ae2c334d07223 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93d185cb4b824412975441176b169815 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81a46f8af08648b89842043d61d96de3 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac6afd547f8440768393a35b57a167c3 : SUCCESS in 6m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/91681af5c8454a67866f8c22fefdf5ad : SUCCESS in 26m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b0d6da5e15b4bbca06ac5f4395becae : SUCCESS in 48m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/58de5fb627bf4f0e8252215fb3b6f87d : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":3},{"id":"a25d150354a0f3248ab5f46e0e61c091ffb876cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 10:14:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8e8e9e97353751dfe2dc502e8e8e18477340eee6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 10:15:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bafa68ca3e0f4be4b8ed174a20ddcbd1 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3d837170ea7196a08692a5f9f715e8b41da189a8","revisions":{"e31392a7e11e910bb90b20225ad4ffa776cdcfbc":{"kind":"REWORK","_number":1,"created":"2022-05-06 05:47:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/14/840814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/840814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/840814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/840814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/840814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/840814/1"}}},"commit":{"parents":[{"commit":"4343f56e423b9427aeda15609cf28b5717430fbe","subject":"Add multipath to cinder-volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4343f56e423b9427aeda15609cf28b5717430fbe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-05-06 05:45:56.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-05-06 05:47:20.000000000","tz":120},"subject":"cloudkitty-processor: disable building for ubuntu/binary","message":"cloudkitty-processor: disable building for ubuntu/binary\n\nThere is no package in UCA so we put old version.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2022-May/028419.html\n\nChange-Id: Iaf5ce2b7276b275b21d892d0044a126ab901507f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e31392a7e11e910bb90b20225ad4ffa776cdcfbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e31392a7e11e910bb90b20225ad4ffa776cdcfbc"}]},"branch":"refs/heads/stable/yoga"},"9105e18e47227de6421ac403d02bc1528fc4d6a6":{"kind":"REWORK","_number":2,"created":"2022-05-06 06:02:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/14/840814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/840814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/840814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/840814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/840814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/840814/2"}}},"commit":{"parents":[{"commit":"4343f56e423b9427aeda15609cf28b5717430fbe","subject":"Add multipath to cinder-volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4343f56e423b9427aeda15609cf28b5717430fbe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-05-06 05:45:56.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-05-06 06:02:15.000000000","tz":0},"subject":"cloudkitty: disable building for ubuntu/binary","message":"cloudkitty: disable building for ubuntu/binary\n\nThere are no packages in UCA.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2022-May/028419.html\n\nChange-Id: Iaf5ce2b7276b275b21d892d0044a126ab901507f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9105e18e47227de6421ac403d02bc1528fc4d6a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9105e18e47227de6421ac403d02bc1528fc4d6a6"}]},"branch":"refs/heads/stable/yoga"},"3d837170ea7196a08692a5f9f715e8b41da189a8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-06 13:40:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/14/840814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/840814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/840814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/840814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/840814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/840814/3"}}},"commit":{"parents":[{"commit":"0ca52d999b4eea98e813d78bd26d5e3aba97fc97","subject":"prometheus-libvirt-exporter: fix build with newer Go","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ca52d999b4eea98e813d78bd26d5e3aba97fc97"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz+kolla@linaro.org","date":"2022-05-06 05:45:56.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-06 13:40:11.000000000","tz":0},"subject":"cloudkitty: disable building for ubuntu/binary","message":"cloudkitty: disable building for ubuntu/binary\n\nThere are no packages in UCA.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2022-May/028419.html\n\nChange-Id: Iaf5ce2b7276b275b21d892d0044a126ab901507f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d837170ea7196a08692a5f9f715e8b41da189a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d837170ea7196a08692a5f9f715e8b41da189a8"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
